)]}'
{"id":"openstack%2Fswift-bench~734559","triplet_id":"openstack%2Fswift-bench~master~I3da40f2d2ef55e4665752450059dddb188414b90","project":"openstack/swift-bench","branch":"master","topic":"drop_mock","attention_set":{},"removed_from_attention_set":{"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2025-02-27 05:38:38.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-02-27 05:38:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3da40f2d2ef55e4665752450059dddb188414b90","subject":"drop mock from lower-constraints and requirements","status":"MERGED","created":"2020-06-09 10:14:14.000000000","updated":"2025-02-27 05:38:38.000000000","submitted":"2025-02-27 05:38:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"734559-drop_mock","meta_rev_id":"f36a9e9742c3530dffee26232a435d04f16f4bdb","_number":734559,"virtual_id_number":734559,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-27 05:38:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2025-02-27 05:20:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2025-02-27 05:20:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-09 10:40:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-27 05:20:48.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"dda204ab2f4a8c0a9ae16c3fe2313daf7bfafe93","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-09 10:14:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"273b3552450118d8227aff4e82b7fac741f59d8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 10:40:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/e229e2b3e6ef4fafad70b98d51032116 : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f4e7580ed88464999c5d96b701277e8 : SUCCESS in 5m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/715fe322c6484ad88dfa0bfd2cda1832 : POST_FAILURE in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/901327de619b4e10817e3e0d85b3f876 : SUCCESS in 5m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8c14f01b49b34965ac962c3cbe01747d : SUCCESS in 6m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/148c75b2a05f4fba86887e0a0b0535b5 : SUCCESS in 6m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"857171e777d167a16d58dd4a26b44aa8e573772b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-11 09:58:18.000000000","message":"Patch Set 1: Workflow-1\n\nStill support py27","accounts_in_message":[],"_revision_number":1},{"id":"61b5a7ea87069a67e922c0bfc9243ef324215b45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-27 04:49:45.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"10cd27950685eabe23525f350bc242037ff20883","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 04:55:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f6ceec3d1ef740d6bd17094d8c412e32\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/b3e0bd2b75624b858109df80d5644a96 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/872c2c12e65943eb84128e0d446d655b : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ec7b972ea9f34cf29a34e2e9901bbf37 : FAILURE in 2m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ec72bad55d8b49189011dbbfd44424f4 : FAILURE in 2m 58s","accounts_in_message":[],"_revision_number":2},{"id":"22c675ed4977a27429992d51a4b5c29a83aa7e6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-27 05:05:50.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"6c93929e431fc6cef7fe02546206386bdb754c61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 05:17:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e526f77c0d5495595805b4d26722a63\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/5dc2e7ebc4354e72b4d453a55f584e86 : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a95113da73b04c23b4f7635636ff68d7 : SUCCESS in 2m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f80b5a3282840ab8f3d80586754480c : SUCCESS in 2m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b43be08fad1444eb8b74661265f37496 : SUCCESS in 2m 59s","accounts_in_message":[],"_revision_number":3},{"id":"b3e4131529b42eea4c99ed93dcc99370359a16c3","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-02-27 05:20:48.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"80eb22fb80bd01fe9b2857b08f84a47bfe4f59d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 05:21:06.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"36e70a325eea3338d85c8050514ceb297d34c764","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 05:38:38.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b47d7b3270ed4bb8851d02c814f97488\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/03278ce52f8b4cb9bbeb5e5f0a71f3bc : SUCCESS in 1m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b328c187ffa41cfb09abec77e1526ab : SUCCESS in 5m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5803b81ace5644ec842d21f4adcf88f8 : SUCCESS in 2m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4a654aec4eb746a48341fbe83113d8f2 : SUCCESS in 2m 28s","accounts_in_message":[],"_revision_number":3},{"id":"f36a9e9742c3530dffee26232a435d04f16f4bdb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 05:38:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a463307d544371a37ca01312c6e7b1d63921fd68","revisions":{"028a96a977f2c71f3e24bef71d581d24f69db1d2":{"kind":"REWORK","_number":1,"created":"2020-06-09 10:14:14.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/59/734559/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift-bench","ref":"refs/changes/59/734559/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/59/734559/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/59/734559/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/59/734559/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift-bench refs/changes/59/734559/1"}}},"commit":{"parents":[{"commit":"31d474723c2118728cafee0517d7ad37da831682","subject":"Switch to Ussuri jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/31d474723c2118728cafee0517d7ad37da831682"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-09 10:14:09.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-09 10:14:09.000000000","tz":120},"subject":"drop mock from lower-constraints and requirements","message":"drop mock from lower-constraints and requirements\n\nThe mock third party library was needed for mock support in py2\nruntimes. Since we now only support py36 and later, we don\u0027t need it\nin lower-constraints and requirements.\n\nThese changes will help us to drop `mock` from openstack/requirements\n\nChange-Id: I3da40f2d2ef55e4665752450059dddb188414b90\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/028a96a977f2c71f3e24bef71d581d24f69db1d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/028a96a977f2c71f3e24bef71d581d24f69db1d2"}]},"branch":"refs/heads/master"},"0acaf5a32b4dd2f99ae289c3bc909775317cd44b":{"kind":"REWORK","_number":2,"created":"2025-02-27 04:49:45.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/59/734559/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift-bench","ref":"refs/changes/59/734559/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/59/734559/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/59/734559/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/59/734559/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift-bench refs/changes/59/734559/2"}}},"commit":{"parents":[{"commit":"13311e115959531fa137ac8b1bebfee179870b31","subject":"Merge \"Switch from optparse to argparse\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/13311e115959531fa137ac8b1bebfee179870b31"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-09 10:14:09.000000000","tz":120},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-27 04:49:07.000000000","tz":-480},"subject":"drop mock from lower-constraints and requirements","message":"drop mock from lower-constraints and requirements\n\nThe mock third party library was needed for mock support in py2\nruntimes. Since we now only support py39 and later, we don\u0027t need it\nin lower-constraints and requirements.\n\nThese changes will help us to drop `mock` from openstack/requirements\n\nChange-Id: I3da40f2d2ef55e4665752450059dddb188414b90\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/0acaf5a32b4dd2f99ae289c3bc909775317cd44b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/0acaf5a32b4dd2f99ae289c3bc909775317cd44b"}]},"branch":"refs/heads/master"},"a463307d544371a37ca01312c6e7b1d63921fd68":{"kind":"REWORK","_number":3,"created":"2025-02-27 05:05:50.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/59/734559/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift-bench","ref":"refs/changes/59/734559/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/59/734559/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/59/734559/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/59/734559/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift-bench refs/changes/59/734559/3"}}},"commit":{"parents":[{"commit":"13311e115959531fa137ac8b1bebfee179870b31","subject":"Merge \"Switch from optparse to argparse\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/13311e115959531fa137ac8b1bebfee179870b31"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-09 10:14:09.000000000","tz":120},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-27 05:05:42.000000000","tz":-480},"subject":"drop mock from lower-constraints and requirements","message":"drop mock from lower-constraints and requirements\n\nThe mock third party library was needed for mock support in py2\nruntimes. Since we now only support py39 and later, we don\u0027t need it\nin lower-constraints and requirements.\n\nThese changes will help us to drop `mock` from openstack/requirements\n\nChange-Id: I3da40f2d2ef55e4665752450059dddb188414b90\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/a463307d544371a37ca01312c6e7b1d63921fd68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/a463307d544371a37ca01312c6e7b1d63921fd68"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
