)]}'
{"id":"openstack%2Fswift-bench~741041","triplet_id":"openstack%2Fswift-bench~master~I46c196c389f9425c398506a4abe4dd0abed7ade9","project":"openstack/swift-bench","branch":"master","hashtags":[],"change_id":"I46c196c389f9425c398506a4abe4dd0abed7ade9","subject":"py3: Fix sigint handler","status":"MERGED","created":"2020-07-14 20:05:22.000000000","updated":"2020-07-18 03:30:22.000000000","submitted":"2020-07-18 00:03:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"741041-1595030628345-b2062de1","meta_rev_id":"2e6af208b32b47d519998ab8790e4814b117efe5","_number":741041,"virtual_id_number":741041,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-18 00:03:48.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-17 23:54:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-17 23:54:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-18 00:03:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"de51e28c053c7f1d7cbac0ab81bd1881642a5830","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-14 20:05:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3e374fe93c82d220867d91bf5c737316ab308d93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 20:10:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b8620824e9c4faabcba65819f76b635 : SUCCESS in 3m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7cd63ee48a594f08a7d2e33c10ead0a3 : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac4a4d1ef75943bf93377568042e545a : SUCCESS in 3m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ad5889b110794563853016e11d6c4414 : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b2cf44e1ae7466da83bda338016ca16 : SUCCESS in 4m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6cf634fb82fff4ba1a04fda49cd68d37c83bf20b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-17 23:54:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ef55aa8315b60cc96cefc8ff023b34fcab7d500f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 23:54:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"43e53c0265a80e8fcc4026cea26f2210e57692bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-18 00:03:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d02f3fec84245d2ab1275351ba4cb51 : SUCCESS in 3m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/20900b5c46e14cb38745331b8d5223a3 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e0dd9bce6840489bbc68a675b079f820 : SUCCESS in 3m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/071ea70ad2d8475caf65426b8047f33d : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":1},{"id":"7009723c81233488fe29cf6284d92cf9f29d985d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-18 00:03:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2e6af208b32b47d519998ab8790e4814b117efe5","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2020-07-18 03:30:22.000000000","message":"Patch Set 1:\n\nI see you didn\u0027t bother with \"from __future__ import print_function\", but tox-py27 passed. What I suspect is, that fragment is not tested beyond pep8.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4539128ffcbc7459b4a6a759b9654edf5312b4e3","revisions":{"4539128ffcbc7459b4a6a759b9654edf5312b4e3":{"kind":"REWORK","_number":1,"created":"2020-07-14 20:05:22.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/41/741041/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift-bench","ref":"refs/changes/41/741041/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/41/741041/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/41/741041/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/41/741041/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift-bench refs/changes/41/741041/1"}}},"commit":{"parents":[{"commit":"31d474723c2118728cafee0517d7ad37da831682","subject":"Switch to Ussuri jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/31d474723c2118728cafee0517d7ad37da831682"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-07-14 19:52:10.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-07-14 19:52:10.000000000","tz":-420},"subject":"py3: Fix sigint handler","message":"py3: Fix sigint handler\n\nChange-Id: I46c196c389f9425c398506a4abe4dd0abed7ade9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/4539128ffcbc7459b4a6a759b9654edf5312b4e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/4539128ffcbc7459b4a6a759b9654edf5312b4e3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
