)]}'
{"id":"openstack%2Fswift-bench~744606","triplet_id":"openstack%2Fswift-bench~master~I852a1eceb4b6b7db7affbd9e74a6e6648eaa2e66","project":"openstack/swift-bench","branch":"master","topic":"migrate-to-focal","hashtags":[],"change_id":"I852a1eceb4b6b7db7affbd9e74a6e6648eaa2e66","subject":"DNM: testing gate on ubuntu focal","status":"ABANDONED","created":"2020-08-04 03:32:13.000000000","updated":"2020-08-04 15:42:35.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e8443c247bb56228277bd7c6d9baa68850acf0e8","_number":744606,"virtual_id_number":744606,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-08-04 04:55:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-04 04:55:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d4178885370f33cbd9d906c21f4d63d0b1d6e293","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-04 03:32:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"282cefff524b8461fd7dc70656a73926058e8b9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 04:55:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2886b2b94e9d4a9faf18b7ac8d679301 : SUCCESS in 3m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5382b5356a74449898425c5a920f9709 : SUCCESS in 3m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9db471604b64aa89abcf492cebe86fe : SUCCESS in 3m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bf31d0dcac7845859770fc9a21b22520 : SUCCESS in 3m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d548070fea874245a939b3f28410fb6f : SUCCESS in 3m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e1fb171334d10ebb15c30d23ca4dbf4795957bd1","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-04 15:42:35.000000000","message":"Abandoned\n\nall working fine, no updates needed for Focal migration.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"771b4cbb83bea67808190f4d176f43120330d29d","revisions":{"771b4cbb83bea67808190f4d176f43120330d29d":{"kind":"REWORK","_number":1,"created":"2020-08-04 03:32:13.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/06/744606/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift-bench","ref":"refs/changes/06/744606/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/06/744606/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/06/744606/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/06/744606/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift-bench refs/changes/06/744606/1"}}},"commit":{"parents":[{"commit":"8845f72f40893606e2e24c141e6f0aea6062b59f","subject":"py2: Import print function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/8845f72f40893606e2e24c141e6f0aea6062b59f"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-04 02:24:37.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-04 03:32:12.000000000","tz":-300},"subject":"DNM: testing gate on ubuntu focal","message":"DNM: testing gate on ubuntu focal\n\nThis tests:\n- unit tests jobs\n- functional integration tests jobs\n- docs jobs\nStory: #2007865\nTask: #40221\n\nDepends-On: https://review.opendev.org/#/c/734700/\nChange-Id: I852a1eceb4b6b7db7affbd9e74a6e6648eaa2e66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/771b4cbb83bea67808190f4d176f43120330d29d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/771b4cbb83bea67808190f4d176f43120330d29d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
