)]}'
{"id":"openstack%2Fswift-bench~755151","triplet_id":"openstack%2Fswift-bench~master~I77a0450e6e682f568de2ecde45765b0fd8762e20","project":"openstack/swift-bench","branch":"master","topic":"bp/six-removal","hashtags":[],"change_id":"I77a0450e6e682f568de2ecde45765b0fd8762e20","subject":"Remove six and py27 tox","status":"ABANDONED","created":"2020-09-30 01:21:01.000000000","updated":"2025-02-27 04:46:09.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e9f199ee3aa0e2069d33bff860e5a0fee5026444","_number":755151,"virtual_id_number":755151,"owner":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":32475,"name":"Qitao","email":"15662728521@163.com","username":"qihongtao2"},{"date":"2020-10-08 00:44:42.000000000","_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-09-30 07:07:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":-1,"date":"2020-10-07 19:28:04.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32475,"name":"Qitao","email":"15662728521@163.com","username":"qihongtao2"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":32475,"name":"Qitao","email":"15662728521@163.com","username":"qihongtao2"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},{"_account_id":32475,"name":"Qitao","email":"15662728521@163.com","username":"qihongtao2"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-30 06:38:44.000000000","updated_by":{"_account_id":32475,"name":"Qitao","email":"15662728521@163.com","username":"qihongtao2"},"reviewer":{"_account_id":32475,"name":"Qitao","email":"15662728521@163.com","username":"qihongtao2"},"state":"REVIEWER"},{"updated":"2020-09-30 07:07:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-07 19:28:04.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"fc66bfca87e5dc8ff20e765a8164786ad5309dff","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-09-30 01:21:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cfa4c54da2f5d791ca1590e918a52dfed494e061","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 01:27:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/67e73fa1a3084afd818e54ab725f1fd4 : SUCCESS in 2m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c86eb419fe245e9b2fa1b13c6b06950 : SUCCESS in 3m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9ecdc13c7a6b446da5b9d9ad08d09351 : POST_FAILURE in 2m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/362b45c3847547a49efb04bc7f209bf6 : SUCCESS in 3m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f95a7f6ad954440a85246457b4936627 : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa199318183046b6934f2415ee87f1e1 : SUCCESS in 3m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8c5102f254e852bc0956219a2f89f8b689174fa1","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-09-30 06:06:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f2299942759488d33437f8773bb6fe35c42c998f","author":{"_account_id":32475,"name":"Qitao","email":"15662728521@163.com","username":"qihongtao2"},"date":"2020-09-30 06:38:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b62fbdded19f96f755086aa194f67469745bd4a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 06:42:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/434a883e1d234e829c4a1a2f134be64a : SUCCESS in 2m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cad846f01230430b93650bd0c256f064 : SUCCESS in 4m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/36e6ed2dd23d4a1a9438a47dd8461ffd : POST_FAILURE in 3m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/51efc6f240c74b5ca964163fd95d6efe : SUCCESS in 3m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c8d494442ad04207bab75279ab6cfaa1 : SUCCESS in 2m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/64b4e412d83d4373b44431018de407dd : SUCCESS in 3m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0c1ae2509f152871c5348822707e8c8f37fd21b8","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-09-30 06:43:04.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"68cbecc5f81b614cb6ffc12c89c6273872b4ca97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 07:07:04.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d4bc3bcf21964e7983428dbdf6894b0a : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f5f0b4ae04d4a3f9801f841c8e078c6 : SUCCESS in 2m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5a7d4ce2f425459abe5640349d7b1310 : POST_FAILURE in 3m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1e30d7edf874634a6e758b4766712ac : SUCCESS in 3m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d02fea7e41184a94826de7c388a80d61 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/821bd2340f434cb7aa6d0178dd4f061f : SUCCESS in 3m 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"59ce5e49e3149c9f8858de0de4f4d12fdef1fc87","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-07 19:28:04.000000000","message":"Patch Set 3: Code-Review-1\n\nThis repo still supports py27, so this is not appropriate at this time.","accounts_in_message":[],"_revision_number":3},{"id":"07a7ee69630b96385f50702d14a9c2293d1d4f59","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-07 19:28:09.000000000","message":"Topic set to bp/six-removal","accounts_in_message":[],"_revision_number":3},{"id":"54cda9e5f532d128b52544b4a452d22dceb84d89","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-10-08 00:44:42.000000000","message":"Patch Set 3:\n\n\u003e This repo still supports py27, so this is not appropriate at this\n \u003e time.\n\nOk， got it， thanks。","accounts_in_message":[],"_revision_number":3},{"id":"e9f199ee3aa0e2069d33bff860e5a0fee5026444","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-27 04:46:09.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"48497b278499f7d11bf11b91e166fd61c34e54bd","revisions":{"ad785381ce15677d5ed0fa6558d573a691aaed56":{"kind":"REWORK","_number":1,"created":"2020-09-30 01:21:01.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/51/755151/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift-bench","ref":"refs/changes/51/755151/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/51/755151/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/51/755151/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/51/755151/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift-bench refs/changes/51/755151/1"}}},"commit":{"parents":[{"commit":"8845f72f40893606e2e24c141e6f0aea6062b59f","subject":"py2: Import print function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/8845f72f40893606e2e24c141e6f0aea6062b59f"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-09-30 01:20:50.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-09-30 01:20:50.000000000","tz":480},"subject":"Remove six Replace the following items with Python 3 style code.","message":"Remove six\nReplace the following items with Python 3 style code.\n\n- six.moves\n- six.string_types\n\nChange-Id: I77a0450e6e682f568de2ecde45765b0fd8762e20\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/ad785381ce15677d5ed0fa6558d573a691aaed56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/ad785381ce15677d5ed0fa6558d573a691aaed56"}]},"branch":"refs/heads/master"},"8bb159de673eeefb7210b8343bb46b168542f169":{"kind":"REWORK","_number":2,"created":"2020-09-30 06:06:03.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/51/755151/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift-bench","ref":"refs/changes/51/755151/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/51/755151/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/51/755151/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/51/755151/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift-bench refs/changes/51/755151/2"}}},"commit":{"parents":[{"commit":"8845f72f40893606e2e24c141e6f0aea6062b59f","subject":"py2: Import print function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/8845f72f40893606e2e24c141e6f0aea6062b59f"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-09-30 01:20:50.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-09-30 06:05:18.000000000","tz":480},"subject":"Remove six and py27 tox Replace the following items with Python 3 style code.","message":"Remove six and py27 tox\nReplace the following items with Python 3 style code.\n\n- six.moves\n- six.string_types\n\nChange-Id: I77a0450e6e682f568de2ecde45765b0fd8762e20\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/8bb159de673eeefb7210b8343bb46b168542f169"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/8bb159de673eeefb7210b8343bb46b168542f169"}]},"branch":"refs/heads/master"},"48497b278499f7d11bf11b91e166fd61c34e54bd":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-09-30 06:43:04.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/51/755151/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift-bench","ref":"refs/changes/51/755151/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/51/755151/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/51/755151/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/51/755151/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift-bench refs/changes/51/755151/3"}}},"commit":{"parents":[{"commit":"8845f72f40893606e2e24c141e6f0aea6062b59f","subject":"py2: Import print function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/8845f72f40893606e2e24c141e6f0aea6062b59f"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-09-30 01:20:50.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-09-30 06:42:53.000000000","tz":480},"subject":"Remove six and py27 tox","message":"Remove six and py27 tox\n\nReplace the following items with Python 3 style code.\n\n- six.moves\n- six.string_types\n\nChange-Id: I77a0450e6e682f568de2ecde45765b0fd8762e20\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/48497b278499f7d11bf11b91e166fd61c34e54bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/48497b278499f7d11bf11b91e166fd61c34e54bd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
