)]}'
{"id":"openstack%2Fswift-bench~787082","triplet_id":"openstack%2Fswift-bench~master~I8225f85caa5db84d10eef91a7948306d98ba8962","project":"openstack/swift-bench","branch":"master","hashtags":[],"change_id":"I8225f85caa5db84d10eef91a7948306d98ba8962","subject":"Use py3 as the default runtime for tox","status":"MERGED","created":"2021-04-20 05:59:45.000000000","updated":"2021-07-01 01:30:08.000000000","submitted":"2021-07-01 01:30:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"787082","meta_rev_id":"c0832edf2d8d48ddc3ef4111d4a7f5df38a2e965","_number":787082,"virtual_id_number":787082,"owner":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-01 01:30:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-07-01 01:20:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-07-01 01:20:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-20 06:13:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-01 01:18:53.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"23f191f0de8c58eb7f5605fd9269effddd39873f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"date":"2021-04-20 05:59:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ab9b7695d9b0da24b1e5820a354672ec5c5bd904","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 06:13:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10b056c935954458a0dfbc547def44a0 : SUCCESS in 5m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/10084ec0e30147bdba16f7bb21cc1d2c : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a8926b344b242e5b2a77b98ae6324bd : SUCCESS in 3m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5422a59f2ffb443e8df46bcd6d17ffda : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dcff2649805d4b888455543e0f4e62fe : SUCCESS in 4m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b49f5224592cb91554c168890c0d72791aa853fb","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-07-01 01:18:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5323bd43e983df6c1910ff7688c508ca16b70846","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-07-01 01:20:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6f5ce87902ed939930ec42515df5ac51106912d0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-07-01 01:20:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3fc90ccad52c4ff9ddf389962d91735fcd398e19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 01:25:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44adaebafad54d58b53f9cfdf637b8ba : SUCCESS in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/84e3a488a0db4c32a64820b2fb2aa20d : SUCCESS in 3m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9874b9f270fc41a5b675d8cbe6c43884 : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee037e2196d24ea58856bc189fc532e2 : SUCCESS in 3m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"91f63bc1a552a6fb3c35433f0b11bbb9199416c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 01:25:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9e971202343aa90bd96526b3c5bcef56b0df1653","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 01:30:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1c42750be8c4983960331089778297e : SUCCESS in 3m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3946fbb0f9e540cc8a9ae6a2d40c35cf : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c98fc5a1dfd54c7391c816c7674c5263 : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":2},{"id":"c0832edf2d8d48ddc3ef4111d4a7f5df38a2e965","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 01:30:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9d708e6c902171615032b65c63f08432698a6a7b","revisions":{"335776031795b0f1cea9660796deceec0e0afda8":{"kind":"REWORK","_number":1,"created":"2021-04-20 05:59:45.000000000","uploader":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"ref":"refs/changes/82/787082/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift-bench","ref":"refs/changes/82/787082/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/82/787082/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/82/787082/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/82/787082/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift-bench refs/changes/82/787082/1"}}},"commit":{"parents":[{"commit":"8845f72f40893606e2e24c141e6f0aea6062b59f","subject":"py2: Import print function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/8845f72f40893606e2e24c141e6f0aea6062b59f"}]}],"author":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2021-04-20 05:59:11.000000000","tz":480},"committer":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2021-04-20 05:59:40.000000000","tz":480},"subject":"Use py3 as the default runtime for tox","message":"Use py3 as the default runtime for tox\n\nMoving on py3 as the default runtime for tox to avoid to update\nthis at each new cycle.\n\nChange-Id: I8225f85caa5db84d10eef91a7948306d98ba8962\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/335776031795b0f1cea9660796deceec0e0afda8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/335776031795b0f1cea9660796deceec0e0afda8"}]},"branch":"refs/heads/master"},"9d708e6c902171615032b65c63f08432698a6a7b":{"kind":"REWORK","_number":2,"created":"2021-07-01 01:20:01.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/82/787082/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift-bench","ref":"refs/changes/82/787082/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/82/787082/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/82/787082/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/82/787082/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift-bench refs/changes/82/787082/2"}}},"commit":{"parents":[{"commit":"f5096bf6c83a287905f79d0e005cfcb61ae5baaf","subject":"Merge \"Drop testing for py27\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/f5096bf6c83a287905f79d0e005cfcb61ae5baaf"}]}],"author":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2021-04-20 05:59:11.000000000","tz":480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-07-01 01:19:51.000000000","tz":-420},"subject":"Use py3 as the default runtime for tox","message":"Use py3 as the default runtime for tox\n\nMoving on py3 as the default runtime for tox to avoid to update\nthis at each new cycle.\n\nChange-Id: I8225f85caa5db84d10eef91a7948306d98ba8962\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/9d708e6c902171615032b65c63f08432698a6a7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/9d708e6c902171615032b65c63f08432698a6a7b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
