)]}'
{"id":"openstack%2Fswift-bench~798941","triplet_id":"openstack%2Fswift-bench~master~I0948bd22e257285c77fb61564684ab31080ecf54","project":"openstack/swift-bench","branch":"master","topic":"stestr","attention_set":{},"removed_from_attention_set":{"33370":{"account":{"_account_id":33370,"name":"Hitesh Kumar","display_name":"HeroicHitesh","email":"email.hiteshkumar@gmail.com","username":"heroichitesh"},"last_update":"2021-10-12 16:54:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0948bd22e257285c77fb61564684ab31080ecf54","subject":"Migrate from testr to stestr","status":"MERGED","created":"2021-06-30 18:13:18.000000000","updated":"2021-10-12 16:54:02.000000000","submitted":"2021-10-12 16:54:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"798941-stestr","meta_rev_id":"c358c16cc7302ba79e5b3ddedf8786a6ccff101d","_number":798941,"virtual_id_number":798941,"owner":{"_account_id":33370,"name":"Hitesh Kumar","display_name":"HeroicHitesh","email":"email.hiteshkumar@gmail.com","username":"heroichitesh"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-12 16:54:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-10-12 16:48:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-10-12 16:48:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-30 18:14:41.000000000","updated_by":{"_account_id":33370,"name":"Hitesh Kumar","display_name":"HeroicHitesh","email":"email.hiteshkumar@gmail.com","username":"heroichitesh"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2021-06-30 18:23:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-01 02:21:09.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2021-07-01 02:21:54.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"f7b28f19fd5d00ba53bb68e0df72a91463a43cb5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33370,"name":"Hitesh Kumar","display_name":"HeroicHitesh","email":"email.hiteshkumar@gmail.com","username":"heroichitesh"},"date":"2021-06-30 18:13:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"82f7bebe8d4511fc221e653a254336a19dc7da16","author":{"_account_id":33370,"name":"Hitesh Kumar","display_name":"HeroicHitesh","email":"email.hiteshkumar@gmail.com","username":"heroichitesh"},"date":"2021-06-30 18:14:41.000000000","message":"Patch Set 1:\n\nReview regarding migration from testr to stestr under OSC-5 of RHOSC\u002721","accounts_in_message":[],"_revision_number":1},{"id":"451da7a20cc716e55e8af35ce68c7d3b5f0bcf52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 18:23:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d2468ba7e38148bfb93bd826225d0906 : SUCCESS in 2m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8f8f1e11a2b46a1ad7639f5e57281a4 : SUCCESS in 3m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0bcefa5fe54642dab326428fe1044fe7 : FAILURE in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f6ed18673e94bddb2a54bdd7588d40e : SUCCESS in 3m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1afdaac4faa54b6e95847738aeea40bd : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a56403089c234703ad5c3bc664b11430 : SUCCESS in 3m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"970cc0d438ebc4d2a9216337a745afe15f907554","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-07-01 02:21:09.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nThanks for this! Sorry, been a while since there were any patches against this repo; I think that py27 job is toast, so I think we\u0027ll just (finally) drop it. If you don\u0027t mind, I\u0027ll push up a new revision.","accounts_in_message":[],"_revision_number":1},{"id":"fe1a219cf1ce5d11f70e43866b29e6e7821691cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-07-01 02:21:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"43a6193ad5e5546d14c89bbdab10ffed908da656","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 02:26:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/5c7840ef1d54413d8cc34d2824c1f42d : SUCCESS in 2m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8e2b75265f145ebb78d4d5433be20b8 : SUCCESS in 3m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04e9eef7239d49808e94980c9fa9dd58 : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/750bf73abbc94919914f4403bba4a119 : SUCCESS in 3m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cc9ce21c650c4c0899bc1bb996069c9a : SUCCESS in 3m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f41652801f5add320b2102c525f1174324aa18d2","author":{"_account_id":33370,"name":"Hitesh Kumar","display_name":"HeroicHitesh","email":"email.hiteshkumar@gmail.com","username":"heroichitesh"},"date":"2021-07-01 04:47:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2ef9802561d19b0a01ffa8eadc213ee7dee207d0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-10-12 16:48:04.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d295fa37fc1562e9ba6409b26d52fad2b071c7b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 16:48:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cf6aa94687069ba7d9f0320387e3f272e58fabf8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 16:54:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/ec663f8aa6424e80aa21a36b2c32f304 : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/792d2651840747359a98ed5485e2a919 : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d862eb379984b85a4eeffbcb9d15f9a : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/673ba87fe5c544a2be4092d88b4676df : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":2},{"id":"c358c16cc7302ba79e5b3ddedf8786a6ccff101d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 16:54:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7d3192cdee334c7b38a806a699d76611ee8567ce","revisions":{"a28de23ae6fb7f250caf1d9c5f21db1b6856f65a":{"kind":"REWORK","_number":1,"created":"2021-06-30 18:13:18.000000000","uploader":{"_account_id":33370,"name":"Hitesh Kumar","display_name":"HeroicHitesh","email":"email.hiteshkumar@gmail.com","username":"heroichitesh"},"ref":"refs/changes/41/798941/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift-bench","ref":"refs/changes/41/798941/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/41/798941/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/41/798941/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/41/798941/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift-bench refs/changes/41/798941/1"}}},"commit":{"parents":[{"commit":"8845f72f40893606e2e24c141e6f0aea6062b59f","subject":"py2: Import print function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/8845f72f40893606e2e24c141e6f0aea6062b59f"}]}],"author":{"name":"HeroicHitesh","email":"email.hiteshkumar@gmail.com","date":"2021-06-30 18:11:25.000000000","tz":330},"committer":{"name":"HeroicHitesh","email":"email.hiteshkumar@gmail.com","date":"2021-06-30 18:13:12.000000000","tz":330},"subject":"Migrate from testr to stestr","message":"Migrate from testr to stestr\n\n* Replace .testr.conf by .stestr.conf for migration and update\n  .gitignore, tox.ini and test-requirements.txt files accordingly\n* Use py3 as the default runtime for tox\n\nSigned-off-by: HeroicHitesh \u003cemail.hiteshkumar@gmail.com\u003e\nChange-Id: I0948bd22e257285c77fb61564684ab31080ecf54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/a28de23ae6fb7f250caf1d9c5f21db1b6856f65a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/a28de23ae6fb7f250caf1d9c5f21db1b6856f65a"}]},"branch":"refs/heads/master"},"7d3192cdee334c7b38a806a699d76611ee8567ce":{"kind":"REWORK","_number":2,"created":"2021-07-01 02:21:54.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/41/798941/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift-bench","ref":"refs/changes/41/798941/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/41/798941/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/41/798941/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/41/798941/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift-bench refs/changes/41/798941/2"}}},"commit":{"parents":[{"commit":"50e1e58fe8002ac05cc228c21139b894415e25d5","subject":"Merge \"Use py3 as the default runtime for tox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/50e1e58fe8002ac05cc228c21139b894415e25d5"}]}],"author":{"name":"HeroicHitesh","email":"email.hiteshkumar@gmail.com","date":"2021-06-30 18:11:25.000000000","tz":330},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-07-01 02:21:39.000000000","tz":-420},"subject":"Migrate from testr to stestr","message":"Migrate from testr to stestr\n\nReplace .testr.conf by .stestr.conf for migration and update\n.gitignore, tox.ini and test-requirements.txt files accordingly\n\nSigned-off-by: HeroicHitesh \u003cemail.hiteshkumar@gmail.com\u003e\nChange-Id: I0948bd22e257285c77fb61564684ab31080ecf54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/7d3192cdee334c7b38a806a699d76611ee8567ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/7d3192cdee334c7b38a806a699d76611ee8567ce"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
