)]}'
{"id":"openstack%2Fswift-bench~874341","triplet_id":"openstack%2Fswift-bench~master~Ie56dc0cdcc56577570e13e48732d7d72c63820e4","project":"openstack/swift-bench","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-02-27 04:17:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie56dc0cdcc56577570e13e48732d7d72c63820e4","subject":"Switch from optparse to argparse","status":"MERGED","created":"2023-02-17 23:46:49.000000000","updated":"2025-02-27 04:17:06.000000000","submitted":"2025-02-27 04:17:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874341","meta_rev_id":"44fc45bddd54ea061d7a4e6177d618578243d791","_number":874341,"virtual_id_number":874341,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-27 04:17:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2025-02-27 04:05:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2025-02-27 04:05:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-17 23:50:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-27 04:05:10.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"ad5b1684a916b086fef95596451093bb674b21dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-02-17 23:46:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d9f36af287ee396eaa311edf52eeaa425c1666fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 23:50:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9f504341f4441fd9aa742312a4a7dc0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a3c7712b95d4b9f923fdbe7698f1894 : SUCCESS in 2m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/00785535ae744f2cb47d01c308942f90 : SUCCESS in 2m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ec8f468778784ca4a1b3dbceb467ff47 : SUCCESS in 2m 38s","accounts_in_message":[],"_revision_number":1},{"id":"14a0d6ee627943c30bc577b019f6f302f83383e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-07-10 22:42:47.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"2368c2eb0e8aa87d2681fd473791d85f5a1c8863","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-10 22:51:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80c7cc32462c42fe91f8da0046db7db9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68d430d753fb47b7bdf0d05c1e14443f : SUCCESS in 2m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/54194ea1a01b4054a8acdde62010a9e6 : SUCCESS in 2m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/05488c7e55f14a6e86db1573e67e3e88 : SUCCESS in 3m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c7d6d4882bbd4a5785fda34c051909d7 : SUCCESS in 2m 40s","accounts_in_message":[],"_revision_number":2},{"id":"341f33edd3888b5b94bc539e3c8ebcdda9c2ca63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-27 17:16:09.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"f21accd8b775b2458937d8be90e69d071a4477c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 17:27:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3e926f2312f4a318375ab194175a29d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c195a4f844b7478bb6c9c4efd0ee5991 : SUCCESS in 2m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d67ac909d2e94470ac3d78b8b3a2a8c0 : SUCCESS in 2m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1fb40518613b4a5ea3060c13091390e2 : SUCCESS in 3m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d20840639824408cb9cedb20f7bb1a70 : FAILURE in 6m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8ca006a3a7adcba31f662c34c127747ad6c036f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"real_author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-02-27 03:58:01.000000000","message":"Patch Set 4: Patch Set 3 was rebased on behalf of \u003cGERRIT_ACCOUNT_15343\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"_revision_number":4},{"id":"0a5d0b40db47862c1a2e811a2c21e2fb746910b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 04:03:10.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89ec91eee1334d80b8ae62a588e5315f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98d170a0878d4b69b32e6a45707da7fc : SUCCESS in 3m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b2ffbc7d770a4ec9acdb82cf3425e5b2 : SUCCESS in 2m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c7bef28ec598417a99ae096aea4c1503 : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":4},{"id":"950c15ee8fa7bc19bc39ea41adfb14f845509b8d","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-02-27 04:05:10.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"7d76355abf46ef2bedd055d80dea47e2661ebed2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 04:05:32.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"5e38840d938684d94313cebec680dff8a9e63d1e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 04:17:06.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/346144de0f5f4e98a0d1fc9fafb5d116\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/007d4ae80c42405cb1a8c669cd80405d : SUCCESS in 3m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4725335aee4f4833b74e11735831e014 : SUCCESS in 3m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/93218494e6b04d60819ef67d08d58f7a : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":4},{"id":"44fc45bddd54ea061d7a4e6177d618578243d791","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 04:17:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b00714980dc016e38fe9a0f1d70c79bdabc2067a","revisions":{"6bce89964bbb6362009aabb76bf1831152c87cca":{"kind":"REWORK","_number":1,"created":"2023-02-17 23:46:49.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/41/874341/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift-bench","ref":"refs/changes/41/874341/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/41/874341/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/41/874341/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/41/874341/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift-bench refs/changes/41/874341/1"}}},"commit":{"parents":[{"commit":"1c0c75543153a89b31c91c89f08bc1b3a1ff47d4","subject":"refactor bin/bench into swiftbench/cli for testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/1c0c75543153a89b31c91c89f08bc1b3a1ff47d4"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-02-17 23:33:50.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-02-17 23:33:50.000000000","tz":-480},"subject":"Switch from optparse to argparse","message":"Switch from optparse to argparse\n\nChange-Id: Ie56dc0cdcc56577570e13e48732d7d72c63820e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/6bce89964bbb6362009aabb76bf1831152c87cca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/6bce89964bbb6362009aabb76bf1831152c87cca"}]},"branch":"refs/heads/master"},"a395f374897dd5af72a83fe1f5c92900331d67fb":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-07-10 22:42:47.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/41/874341/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift-bench","ref":"refs/changes/41/874341/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/41/874341/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/41/874341/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/41/874341/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift-bench refs/changes/41/874341/2"}}},"commit":{"parents":[{"commit":"5f803ae8fde7d5511e8b364ae7fd86fac09e48d3","subject":"refactor bin/bench into swiftbench/cli for testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/5f803ae8fde7d5511e8b364ae7fd86fac09e48d3"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-02-17 23:33:50.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-07-10 22:42:24.000000000","tz":-420},"subject":"Switch from optparse to argparse","message":"Switch from optparse to argparse\n\nChange-Id: Ie56dc0cdcc56577570e13e48732d7d72c63820e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/a395f374897dd5af72a83fe1f5c92900331d67fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/a395f374897dd5af72a83fe1f5c92900331d67fb"}]},"branch":"refs/heads/master"},"7eb2f89b119d88a555db3047010f5431fe987ef7":{"kind":"REWORK","_number":3,"created":"2024-06-27 17:16:09.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/41/874341/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift-bench","ref":"refs/changes/41/874341/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/41/874341/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/41/874341/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/41/874341/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift-bench refs/changes/41/874341/3"}}},"commit":{"parents":[{"commit":"666fc883fa9db6f54765585460707da3e6127b9a","subject":"Removed Python 2 support and six","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/666fc883fa9db6f54765585460707da3e6127b9a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-02-17 23:33:50.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-27 17:15:07.000000000","tz":-420},"subject":"Switch from optparse to argparse","message":"Switch from optparse to argparse\n\nChange-Id: Ie56dc0cdcc56577570e13e48732d7d72c63820e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/7eb2f89b119d88a555db3047010f5431fe987ef7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/7eb2f89b119d88a555db3047010f5431fe987ef7"}]},"branch":"refs/heads/master"},"b00714980dc016e38fe9a0f1d70c79bdabc2067a":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-02-27 03:58:01.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"real_uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/41/874341/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift-bench","ref":"refs/changes/41/874341/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/41/874341/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/41/874341/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/41/874341/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift-bench refs/changes/41/874341/4"}}},"commit":{"parents":[{"commit":"dee1430b423ede5eab8fe79a7c6ac295c8522735","subject":"Up-rev hacking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/dee1430b423ede5eab8fe79a7c6ac295c8522735"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-02-17 23:33:50.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-27 03:58:01.000000000","tz":0},"subject":"Switch from optparse to argparse","message":"Switch from optparse to argparse\n\nChange-Id: Ie56dc0cdcc56577570e13e48732d7d72c63820e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/b00714980dc016e38fe9a0f1d70c79bdabc2067a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/b00714980dc016e38fe9a0f1d70c79bdabc2067a"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
