)]}'
{"id":"openstack%2Fswift-bench~942873","triplet_id":"openstack%2Fswift-bench~master~I8ecb3143536b9c97438af7b6b2797fb22768e7d3","project":"openstack/swift-bench","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-02-27 03:54:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8ecb3143536b9c97438af7b6b2797fb22768e7d3","subject":"Up-rev hacking","status":"MERGED","created":"2025-02-27 00:21:47.000000000","updated":"2025-02-27 04:22:56.000000000","submitted":"2025-02-27 03:54:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"942873","meta_rev_id":"f3ccabf86cc67899469ca7eb4ee718ccfd0c1819","_number":942873,"virtual_id_number":942873,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-27 03:54:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2025-02-27 03:16:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2025-02-27 03:16:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-27 00:53:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-27 03:16:52.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"6211f254ce3785b990131eeda1afb26bf2cdac7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-27 00:21:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c1a1c940b98759bbb5cd0ead4c3880e7eeb8996a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 00:53:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e6e0c94ad9994b108c81ebffbef34336\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/74a19845a71a46dbbdeb1c00b3be599a : SUCCESS in 2m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd5a89d3aca14f719046a33aafdca0d0 : SUCCESS in 2m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/db290ee7ba8147809b62325a5421b3e0 : FAILURE in 2m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/99fa41a1036a4b038a7d8304606674fa : FAILURE in 3m 36s","accounts_in_message":[],"_revision_number":1},{"id":"11df16e054f7ab8ebd00181339a8320bbda94306","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-02-27 01:39:59.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"03bbf545ae17e56ce246adf0cf6ff3df76ebdc12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 01:48:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4bd15e14efd4607aaf823333f405e64\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/5026b983045646999434554690853b94 : SUCCESS in 2m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b113816e37554ce89a55dbee74b7b614 : SUCCESS in 2m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d82827880d44af8819eb89862f6c3fe : SUCCESS in 3m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/163f6864f415499aba9471bd2c9b8a75 : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":2},{"id":"5408911578332204229f7c4fff797ead7c5b0049","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-02-27 03:16:52.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"594dc81ea35b1306c7502db6c18014c601986b94","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 03:17:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"13f5988b3057517b871547de6e53681caf2fb4b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 03:54:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39ccd9442bc346e6ba57e06c2db2d9d1\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/50c43f923ed04a04a9e30c72f59eecea : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9807a5b8a6d4b37bee762ddaef4d56e : SUCCESS in 7m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9298b55517d5470f8755717989e4145f : SUCCESS in 4m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7382a098ca284d4c80022e36098d49a2 : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":2},{"id":"a64e8ec2af2866e28cdbc676baec83e7a8d6f153","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 03:54:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f3ccabf86cc67899469ca7eb4ee718ccfd0c1819","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-27 04:22:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"dee1430b423ede5eab8fe79a7c6ac295c8522735","revisions":{"515440f4a607c87c5cfc10908076228987763604":{"kind":"REWORK","_number":1,"created":"2025-02-27 00:21:47.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/73/942873/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift-bench","ref":"refs/changes/73/942873/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/73/942873/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/73/942873/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/73/942873/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift-bench refs/changes/73/942873/1"}}},"commit":{"parents":[{"commit":"666fc883fa9db6f54765585460707da3e6127b9a","subject":"Removed Python 2 support and six","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/666fc883fa9db6f54765585460707da3e6127b9a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-27 00:21:35.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-27 00:21:35.000000000","tz":-480},"subject":"Up-rev hacking","message":"Up-rev hacking\n\nChange-Id: I8ecb3143536b9c97438af7b6b2797fb22768e7d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/515440f4a607c87c5cfc10908076228987763604"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/515440f4a607c87c5cfc10908076228987763604"}]},"branch":"refs/heads/master"},"dee1430b423ede5eab8fe79a7c6ac295c8522735":{"kind":"REWORK","_number":2,"created":"2025-02-27 01:39:59.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/73/942873/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift-bench","ref":"refs/changes/73/942873/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/73/942873/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/73/942873/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/73/942873/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift-bench refs/changes/73/942873/2"}}},"commit":{"parents":[{"commit":"666fc883fa9db6f54765585460707da3e6127b9a","subject":"Removed Python 2 support and six","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/666fc883fa9db6f54765585460707da3e6127b9a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-27 00:21:35.000000000","tz":-480},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-02-27 01:39:13.000000000","tz":660},"subject":"Up-rev hacking","message":"Up-rev hacking\n\nAnd fix readfp to read_file and make sure it reads from the\nstart of any filelike object.\n\nChange-Id: I8ecb3143536b9c97438af7b6b2797fb22768e7d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/dee1430b423ede5eab8fe79a7c6ac295c8522735"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/dee1430b423ede5eab8fe79a7c6ac295c8522735"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
