)]}'
{"id":"openstack%2Fswift-bench~943032","triplet_id":"openstack%2Fswift-bench~master~Ia876bf3270dd7d598248bbf93c71a0c70e9c97f4","project":"openstack/swift-bench","branch":"master","hashtags":[],"change_id":"Ia876bf3270dd7d598248bbf93c71a0c70e9c97f4","subject":"Use entry_points for console_scripts; get rid of bin/","status":"NEW","created":"2025-02-28 16:44:44.000000000","updated":"2025-02-28 18:29:31.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9a5aa2e581a73dae3a3a489c47456455d2b817e1","_number":943032,"virtual_id_number":943032,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-02-28 18:29:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-28 18:29:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5c826dcdff0d986fb3f181c34a8a092bea33d053","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-28 16:44:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bdef94cba5fe116974ccbbafe17b58527fb605b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-28 18:15:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9a5aa2e581a73dae3a3a489c47456455d2b817e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 18:29:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7f3b8f225a8442580bc57565b5e9d09\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d08e8aafc09d4a88a6dae5ae2d4e64cc : SUCCESS in 1m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2dc4c637e034093961c8c8f1257ffc8 : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8086582cd4ca40ed9172f1906e77a686 : SUCCESS in 3m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/81ed3aeaf3c0429a9ae3e26d62cbaf09 : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"05b51e3df44e1f3d3e15655fc170e4fbef661d8a","revisions":{"8cb727beac7880525e4756226b220a35a56eca42":{"kind":"REWORK","_number":1,"created":"2025-02-28 16:44:44.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/32/943032/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift-bench","ref":"refs/changes/32/943032/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/32/943032/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/32/943032/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/32/943032/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift-bench refs/changes/32/943032/1"}}},"commit":{"parents":[{"commit":"fd76818d4110f043cb22ca406740ff3e2d9cac3e","subject":"Move swift-bench-client to swiftbench.client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/fd76818d4110f043cb22ca406740ff3e2d9cac3e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-28 16:44:21.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-28 16:44:21.000000000","tz":-480},"subject":"Use entry_points for console_scripts; get rid of bin/","message":"Use entry_points for console_scripts; get rid of bin/\n\nChange-Id: Ia876bf3270dd7d598248bbf93c71a0c70e9c97f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/8cb727beac7880525e4756226b220a35a56eca42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/8cb727beac7880525e4756226b220a35a56eca42"}]},"branch":"refs/heads/master"},"05b51e3df44e1f3d3e15655fc170e4fbef661d8a":{"kind":"REWORK","_number":2,"created":"2025-02-28 18:15:46.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/32/943032/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift-bench","ref":"refs/changes/32/943032/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/32/943032/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/32/943032/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift-bench refs/changes/32/943032/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift-bench refs/changes/32/943032/2"}}},"commit":{"parents":[{"commit":"fd76818d4110f043cb22ca406740ff3e2d9cac3e","subject":"Move swift-bench-client to swiftbench.client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/fd76818d4110f043cb22ca406740ff3e2d9cac3e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-28 16:44:21.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-28 18:15:36.000000000","tz":-480},"subject":"Use entry_points for console_scripts; get rid of bin/","message":"Use entry_points for console_scripts; get rid of bin/\n\nChange-Id: Ia876bf3270dd7d598248bbf93c71a0c70e9c97f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/05b51e3df44e1f3d3e15655fc170e4fbef661d8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift-bench/commit/05b51e3df44e1f3d3e15655fc170e4fbef661d8a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
