)]}'
{"id":"openstack%2Fswift~1000494","triplet_id":"openstack%2Fswift~master~I879790778c0339741d9f19bbb74b86b5506f5e20","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"6968":{"account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"last_update":"2026-08-13 14:38:31.000000000","reason":"Change was submitted"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2026-08-13 14:38:31.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-08-13 13:59:09.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I879790778c0339741d9f19bbb74b86b5506f5e20","subject":"CHANGELOG for 2.38.1","status":"MERGED","created":"2026-08-11 15:08:57.000000000","updated":"2026-08-13 14:39:54.000000000","submitted":"2026-08-13 14:38:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"1000494","meta_rev_id":"87634026b9fcdb5dd7262e78fb0c4b724111688f","_number":1000494,"virtual_id_number":1000494,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-13 14:38:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-08-13 13:59:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-08-13 13:59:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-11 15:39:57.000000000","updated_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"reviewer":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"state":"REVIEWER"},{"updated":"2026-08-11 15:41:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-12 18:28:25.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"cb4aa78a31ea82af3d60ec9f33715d7eea9088d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-08-11 15:08:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"02f6d1d7ea571880d870cb2b42d49228561f37e9","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-08-11 15:39:57.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"442d32738488adb9be2475f247039c0ed042996a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-11 15:41:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/066e3a397de34cbf8125972b668951eb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d889cca92cce42d6815d3403748423c0 : SUCCESS in 10m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b64adff56b3644039907086865e1e58e : SUCCESS in 9m 20s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/339ef50313e44e2b90fda46dfe08a2ed : SUCCESS in 5m 00s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c96934bb9ece4dd2a8a6f6c56759fd38 : SUCCESS in 26m 21s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/85c29ad76bf84e608c0f030eee70f6a3 : SUCCESS in 12m 16s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/78b9323793734d51ac3543dcd3cbfbef : SUCCESS in 12m 19s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/4cc4bf5fb4224faa9cf24e8cd2675720 : SUCCESS in 22m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cc3724db64794afa8a3401459df512c9 : SUCCESS in 11m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba25b897fb4e43f3a075f6d3b35bdaa5 : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":1},{"id":"50311d06f85b74756ade1dc72830ef8823f0770a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-11 16:23:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8a9fd7ea0524e559d64e386648b4107\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1956cfe257bb4e569d02b0cbf27d8e26 : SUCCESS in 1h 11m 12s","accounts_in_message":[],"_revision_number":1},{"id":"ed340b33ee5b17dfc65b86a683b1f314bdd50d6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-08-11 17:54:16.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"647c4c389612bd5e056abe86f6abf2e1c7af3d5b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-08-11 17:54:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3dda85226221b998c40caf8b6ca605f0de3fc943","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-11 18:36:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60d4aabc54b74f25ba40cc12a931f9a8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b221b7208c543519e43eb355f35a1d9 : SUCCESS in 10m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/00e9c136beec4c95ae6dce809b96ba4d : SUCCESS in 9m 59s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5da31383b2f14333a588fb35226c5775 : SUCCESS in 9m 49s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a3e3d23d18114250b808710fad53ea04 : SUCCESS in 26m 59s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/c050e33f94ff426097177baeee4467f3 : SUCCESS in 21m 57s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/6fad27561a3f465cb9e03cced486ab1c : SUCCESS in 21m 37s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/36593fddd5034daba913852526921834 : SUCCESS in 19m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f046de396754ae99072523c310301a3 : SUCCESS in 28m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2abb9f93512d489babd346a9d3b8e617 : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":2},{"id":"16f3c0e770656d3ed32a18ef7581ed3ea3159133","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-11 18:38:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b02d6b6b05f4e7c97bcd1b85dde9fab\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0bfe084a5d7f4daf8ad0c3afba41b08f : SUCCESS in 42m 37s","accounts_in_message":[],"_revision_number":2},{"id":"de932cc4c45e8701b5fbc27b4bb6e1f5396950d6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-08-11 19:01:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3d980528c0a95877d5e75552851159a896be8fc1","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-11 19:58:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc947b4967da4524ba8c3d14be8081a9\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/779f4aec451846b781c904deb727a178 : SUCCESS in 53m 05s\n- swift-tox-func-encryption-py312-arm64 https://zuul.opendev.org/t/openstack/build/1bd625d63de84b83beaaab84bb03da6f : SUCCESS in 40m 55s\n- swift-tox-func-py312-arm64 https://zuul.opendev.org/t/openstack/build/c5923c444af4435199fc64e8decfdda0 : SUCCESS in 43m 26s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/eb75a13f0d334a5a81ebf5de4293bd9e : SUCCESS in 9m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e3d799b5c59c4202a525b82759c4ca06 : SUCCESS in 19m 11s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/a29d1a465626472f8457cf866344833d : SUCCESS in 18m 42s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0b2ccba9a7464a06bc140377564fef62 : SUCCESS in 14m 02s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/293bab88ffa44b9485a9df1120809291 : SUCCESS in 10m 26s\n- swift-tox-func-encryption-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/c917975afae44e509f94e8fd5d6aafbd : SUCCESS in 10m 54s\n- swift-tox-func-ec-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/3a0a89bb423b457abe13fcdbc0d2cf28 : SUCCESS in 10m 52s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/d8e080bc3fea4c75ab590116380729fe : SUCCESS in 19m 05s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/f8834419fbb34a80b6e75d46d68214ac : SUCCESS in 24m 55s\n- swift-multinode-rolling-upgrade-yoga https://zuul.opendev.org/t/openstack/build/ab4eeaa445ef4c5891f201bde413e46d : SUCCESS in 25m 42s\n- swift-multinode-rolling-upgrade-zed https://zuul.opendev.org/t/openstack/build/be73ac2724b54753abf50d9ae6822170 : SUCCESS in 10m 56s\n- swift-multinode-rolling-upgrade-antelope https://zuul.opendev.org/t/openstack/build/71890a95a19c4f0e8719a0c074011e61 : SUCCESS in 18m 18s\n- swift-multinode-rolling-upgrade-bobcat https://zuul.opendev.org/t/openstack/build/b1bbf954a8074896aa2795ab3411f55c : SUCCESS in 8m 57s\n- swift-multinode-rolling-upgrade-caracal https://zuul.opendev.org/t/openstack/build/b75eafad0b0f4b6f9cf9cc1f725a6d7e : SUCCESS in 16m 23s\n- swift-multinode-rolling-upgrade-dalmatian https://zuul.opendev.org/t/openstack/build/2542b0e1de7f477bb81fa0b4380a35f2 : SUCCESS in 19m 09s\n- swift-multinode-rolling-upgrade-epoxy https://zuul.opendev.org/t/openstack/build/ce529e1b84bf459ab4e45fea77777672 : SUCCESS in 25m 14s\n- swift-multinode-rolling-upgrade-flamingo https://zuul.opendev.org/t/openstack/build/b7df678d1eca400281658f716c18baad : SUCCESS in 17m 12s\n- swift-multinode-rolling-upgrade-gazpacho https://zuul.opendev.org/t/openstack/build/59998f365a3246acb72edc17293607b4 : SUCCESS in 18m 17s\n- swift-multinode-rolling-upgrade-master https://zuul.opendev.org/t/openstack/build/d7ea63c69e314c9c84f4eee7b1a4431e : SUCCESS in 18m 30s","accounts_in_message":[],"_revision_number":2},{"id":"e4ec9168c5327fe74140db1fcf5e60f80b31c527","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-08-12 14:00:17.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"c185d39684f9e15df6fa73170016d15e27344fe8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-12 14:40:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c75c9d5b582245ed988a1912ddff501e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e9f6339ea72486ba485d28b44258314 : SUCCESS in 10m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21f3bfcc2c6b45c085e97645d02b65dd : SUCCESS in 9m 32s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/373262ac4f9844bdbc9c71a57453c538 : SUCCESS in 6m 54s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e89222e8efba4492b9fabacc16b6c98f : SUCCESS in 26m 44s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/77a9da802dc74205a91e978848ee95e6 : SUCCESS in 22m 07s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/3f8c1fc622504b90afadd45641081766 : SUCCESS in 31m 46s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/730dccef0865458fb936434cc8226754 : SUCCESS in 11m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1f58d2f04ead4e4a81ef7ed84a4521ca : SUCCESS in 30m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1543701ed6dd4bee98843c7f9113bd36 : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":3},{"id":"6666fd2b7a16c76da20566d6eda9c790dbb78b9b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-12 17:14:14.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3cc0dbee673c4acf885174d1c1e76654\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1abf1e2cbd214c9ca94fd2ce7c6b2691 : SUCCESS in 1h 13m 56s","accounts_in_message":[],"_revision_number":3},{"id":"93361d1f7314bd1a3ae2c6e15f179dacb1431643","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-08-12 18:28:25.000000000","message":"Patch Set 3: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"77d064048ff561007decc433d767424cd8a48d46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-08-12 22:02:46.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"15139cc59b4b7d948ba17e0fa5756ce3799112c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-12 22:34:04.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8101a9889fc45d588413647af1b0eec\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a306ebcd0df41c5b84add47e90fc0bd : SUCCESS in 5m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e2dd27aa746c4c32b1f91cad49a8d316 : SUCCESS in 8m 19s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3eafa75391ba425694a965c38283335e : SUCCESS in 6m 07s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f9cbe6ad0a574719a45884eb688e4667 : SUCCESS in 10m 13s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/302a10ac68304d699364bd7122e6e507 : SUCCESS in 20m 52s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/71e89fc07901460bba701cd3c0f6ea91 : SUCCESS in 23m 26s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/e1e5c836ad8e48e0b56632108ce12fbc : SUCCESS in 24m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b92554a34f6f4725961eaaede451a650 : SUCCESS in 24m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a0656cb5c554233b486c47bb869e00f : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":4},{"id":"018c78bfad32477bd8fad42f276eaa6515299bf5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-12 23:15:24.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/605ee83d896d44ea941f515dc9017632\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ec479380e94746da8648e39fbafcf76b : SUCCESS in 1h 05m 04s","accounts_in_message":[],"_revision_number":4},{"id":"8a3ea09847bb307ca9de611839899da834466402","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-08-13 13:59:09.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"367949f257b9bfcfae5400ab0ea73b48657b1761","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 14:00:04.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"91a9284bda27d1b71b55e47ee100e09fbb190374","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 14:38:30.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35d842e605c94fdeb9f73a777f8b446c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8db826aa7e24431898c413a7b1f330d9 : SUCCESS in 8m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e3e2b48d5afb42cd943bf24f83e2641b : SUCCESS in 7m 47s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/ed4802584d4142f3ac1cc1ed09f9e770 : SUCCESS in 5m 25s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6be892d0bb7e46aea112862fbaa1dc80 : SUCCESS in 18m 21s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/bfbb688cc60b4645980aed8b3f632bbd : SUCCESS in 19m 11s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/698260c5e69a444bb285f14536b2f46a : SUCCESS in 32m 22s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/7628f7ee3a754a11b8953eb4999a7ebe : SUCCESS in 22m 05s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a3cacc6d26434651ac0d2c0d9f5b1736 : SUCCESS in 16m 52s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/41bf063fd6b640fea5a2a4af1a710af6 : SUCCESS in 14m 06s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/db1d9f0e07ae4761b9db480ee6da5545 : SUCCESS in 16m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/561970bf503745fdac495b7fbb2b057c : SUCCESS in 9m 16s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/7dd3a1fbd40d4c468163db2705c9589b : SUCCESS in 9m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/398b603ed61c417cad8481812fef55cc : SUCCESS in 23m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/473f102a58f547bb9430219168807f77 : SUCCESS in 5m 41s","accounts_in_message":[],"_revision_number":4},{"id":"3ed07c96d85ff8575b5fb2548722f00fade7188f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 14:38:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"87634026b9fcdb5dd7262e78fb0c4b724111688f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 14:39:54.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4cec11e5df9949eca1ce721b176915d4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e959cfdd75648e58876a83c21d52062 : SUCCESS in 1m 01s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a28ea37a10a42e0a29bc4dc5325c8d3 : SUCCESS in 51s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/d3d82619a6bb4794ade608284bb2ef17 : SUCCESS in 56s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"3064444bf433dfb7ade3d91130b58f7bceae9add","revisions":{"ba90505b77ffd09d3051d2800970b6153df016a2":{"kind":"REWORK","_number":1,"created":"2026-08-11 15:08:57.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/94/1000494/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/94/1000494/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/1000494/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/1000494/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/1000494/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/94/1000494/1"}}},"commit":{"parents":[{"commit":"482a93529b5822e28e707bccd94037acf4c16d8d","subject":"Merge \"versioning: more tests for OPTIONS requests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/482a93529b5822e28e707bccd94037acf4c16d8d"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-08-11 14:58:14.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-08-11 15:08:37.000000000","tz":-420},"subject":"CHANGELOG for 2.38.1","message":"CHANGELOG for 2.38.1\n\nChange-Id: I879790778c0339741d9f19bbb74b86b5506f5e20\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ba90505b77ffd09d3051d2800970b6153df016a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ba90505b77ffd09d3051d2800970b6153df016a2"}]},"branch":"refs/heads/master"},"537a6b91ea47477de8782ee9d2cf0828a67ebd65":{"kind":"REWORK","_number":2,"created":"2026-08-11 17:54:16.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/94/1000494/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/94/1000494/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/1000494/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/1000494/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/1000494/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/94/1000494/2"}}},"commit":{"parents":[{"commit":"482a93529b5822e28e707bccd94037acf4c16d8d","subject":"Merge \"versioning: more tests for OPTIONS requests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/482a93529b5822e28e707bccd94037acf4c16d8d"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-08-11 14:58:14.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-08-11 17:54:03.000000000","tz":-420},"subject":"CHANGELOG for 2.38.1","message":"CHANGELOG for 2.38.1\n\nChange-Id: I879790778c0339741d9f19bbb74b86b5506f5e20\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/537a6b91ea47477de8782ee9d2cf0828a67ebd65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/537a6b91ea47477de8782ee9d2cf0828a67ebd65"}]},"branch":"refs/heads/master"},"558db8b367e967b50ce11ca20d35591eeffb0eb6":{"kind":"REWORK","_number":3,"created":"2026-08-12 14:00:17.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/94/1000494/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/94/1000494/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/1000494/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/1000494/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/1000494/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/94/1000494/3"}}},"commit":{"parents":[{"commit":"17a47d47e2dfc5d5d23396e194147060525aa34d","subject":"Merge \"Add parallelization to container_sync\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/17a47d47e2dfc5d5d23396e194147060525aa34d"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-08-11 14:58:14.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-08-12 14:00:11.000000000","tz":-420},"subject":"CHANGELOG for 2.38.1","message":"CHANGELOG for 2.38.1\n\nChange-Id: I879790778c0339741d9f19bbb74b86b5506f5e20\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/558db8b367e967b50ce11ca20d35591eeffb0eb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/558db8b367e967b50ce11ca20d35591eeffb0eb6"}]},"branch":"refs/heads/master"},"3064444bf433dfb7ade3d91130b58f7bceae9add":{"kind":"REWORK","_number":4,"created":"2026-08-12 22:02:46.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/94/1000494/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/94/1000494/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/1000494/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/1000494/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/1000494/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/94/1000494/4"}}},"commit":{"parents":[{"commit":"17a47d47e2dfc5d5d23396e194147060525aa34d","subject":"Merge \"Add parallelization to container_sync\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/17a47d47e2dfc5d5d23396e194147060525aa34d"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-08-11 14:58:14.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-08-12 22:02:40.000000000","tz":-420},"subject":"CHANGELOG for 2.38.1","message":"CHANGELOG for 2.38.1\n\nChange-Id: I879790778c0339741d9f19bbb74b86b5506f5e20\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3064444bf433dfb7ade3d91130b58f7bceae9add"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3064444bf433dfb7ade3d91130b58f7bceae9add"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
