)]}'
{"id":"openstack%2Fswift~1000528","triplet_id":"openstack%2Fswift~master~Ib2c7087b335b00a7ec6176dd23ae52ce8d270364","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-08-14 08:22:51.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ib2c7087b335b00a7ec6176dd23ae52ce8d270364","subject":"sq? WIP: MPUChecksumAttrs class??","status":"ABANDONED","created":"2026-08-11 17:13:07.000000000","updated":"2026-08-14 08:22:51.000000000","total_comment_count":31,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"4e6e75e94f410cc5007f4c08b30f1f3e163e8bdc","_number":1000528,"virtual_id_number":1000528,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-08-11 18:28:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-11 18:28:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-11 23:15:07.000000000","updated_by":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"reviewer":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"state":"CC"}],"messages":[{"id":"dbc1919dec6dfbf4ab3124c7e9f32c21051a32da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-08-11 17:13:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"38b59c2c72a758afd4350db5973f0ad1a2b8395b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-11 18:28:53.000000000","message":"Patch Set 1: Verified-1\n\n(28 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/150b1121b32345b7a545fe993c2de4e1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76688a5d78b44152b1c096675d9b6391 : SUCCESS in 7m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e9ed8e3fe48943a486fd748ab248d7e6 : SUCCESS in 8m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/5816f1afba5a4d85aac3c62d4264ced5 : SUCCESS in 52m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bb8fe4590e5b4d9aa858a71fb845c226 : SUCCESS in 1h 07m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d75b4b12657e4288a44428bc538869ea : SUCCESS in 49m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9fbade9199a34e878fdbd80bc74af8f7 : SUCCESS in 52m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/95dbfb7d2b61484b8c6ae9f12c308aef : FAILURE in 25m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/477869b41d6f4825b558542cec482301 : FAILURE in 21m 37s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ea8f96dd82224394ac4338f2d081dcfd : FAILURE in 24m 08s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/502549e15cd8443190ed2fbffbc30302 : SUCCESS in 8m 58s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/55ba78768bf2467c9522679b3c07135e : FAILURE in 29m 14s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b2a0e30b194041d4b08d031f89ae53a7 : FAILURE in 24m 49s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d6b8c64bd2d24c29ba152d8efc1216d9 : FAILURE in 35m 41s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/c05d480e3aa44eb5b1834de12900c669 : FAILURE in 22m 01s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/74091310351c439cbbe0782292fab495 : SUCCESS in 14m 32s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a30fb27050f04880897f59e4e46fdcb8 : SUCCESS in 15m 13s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7b4d574739b840f582e107015f1c3773 : SUCCESS in 15m 25s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/7d997a014a7043499ddb5f50ba9db892 : FAILURE in 21m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/73ce49caa8e147288e84fa35eafe2e97 : SUCCESS in 7m 01s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5666d1231f724206ba8d777e48c67d40 : FAILURE in 7m 08s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/869accb4064542f99b0afc9eb88f4fcf : FAILURE in 12m 21s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d2d8f394143e471880bae657256d1639 : FAILURE in 1h 01m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2435ab8fc67f4db98dfdbae8f0890f00 : FAILURE in 34m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/53a15c1e9e0543c09b7ed84e9452e84a : FAILURE in 36m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/13e308c8f752454080f0541a32875773 : FAILURE in 30m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbbf11e3152d4f5cb1a4bf5699446a1b : FAILURE in 3m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/231bee13b1f74225b0aeaab9619907f1 : SUCCESS in 25m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8cd3ce4f9ef44af4a5de624d96007cbd : SUCCESS in 55m 44s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  Comments left for invalid file swift/common/middleware/s3api/s3request.py","accounts_in_message":[],"_revision_number":1},{"id":"2f226fa7fe98da158bfa29fea48b28ffbdefb3ff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-11 18:33:29.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ea062e975c049099e4950808d25e030\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2573aace83e443d49711223e17d02c6b : FAILURE in 1h 13m 07s","accounts_in_message":[],"_revision_number":1},{"id":"90b53f60bda8c8fb9e7689a06e486700bf8558e6","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-11 23:15:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8e7e7f162b809fd3263b7fc905274e667fa7282b","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-11 23:33:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"df87df943fbe3db695c3ee3c3126cb5f774d3f81","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-11 23:41:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4e6e75e94f410cc5007f4c08b30f1f3e163e8bdc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-08-14 08:22:51.000000000","message":"Abandoned\n\nthe idea found its way into 993193: s3api: Checksum Persistence on MPUs | https://review.opendev.org/c/openstack/swift/+/993193","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"72fc0c807ec14d75dabcde6efbe2981da5df0758","revisions":{"72fc0c807ec14d75dabcde6efbe2981da5df0758":{"kind":"REWORK","_number":1,"created":"2026-08-11 17:13:07.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/28/1000528/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/28/1000528/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/1000528/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/1000528/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/1000528/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/28/1000528/1"}}},"commit":{"parents":[{"commit":"7c3a3eb95a7f684c44760f26484e999462eba581","subject":"sq? CreateMultipartUplaod: cleanup header validation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7c3a3eb95a7f684c44760f26484e999462eba581"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-08-11 17:05:07.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-08-11 17:11:57.000000000","tz":60},"subject":"sq? WIP: MPUChecksumAttrs class??","message":"sq? WIP: MPUChecksumAttrs class??\n\nNOT finished, code is all broken, just sharing the idea.\n\nTBH I haven\u0027t yet convinced myself how helpful this\nwould be, but my goals were:\n\n* encapsulate header/sysmeta parsing\n\n* reduce the number of vars that are being passed around.\n\n* have one place that normalises values e.g. crc-32 -\u003e CRC32\n\n* have some properties that encapsulate conditionals that can be\n  (self-)documented in one place.\n\nMaybe this would also be applicable to regular object checksums. IDK?\n\nChange-Id: Ib2c7087b335b00a7ec6176dd23ae52ce8d270364\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/72fc0c807ec14d75dabcde6efbe2981da5df0758"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/72fc0c807ec14d75dabcde6efbe2981da5df0758"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
