)]}'
{"id":"openstack%2Fswift~1000823","triplet_id":"openstack%2Fswift~master~I406bb3f5b103ca27096ebfc1acae45d5339a8520","project":"openstack/swift","branch":"master","topic":"s3-checksum-persistence","attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-08-13 12:16:20.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I406bb3f5b103ca27096ebfc1acae45d5339a8520","subject":"slo: refactor to simplify heartbeat response paths","status":"NEW","created":"2026-08-13 10:48:13.000000000","updated":"2026-08-13 12:16:20.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5fccead94f781bf290d6f532040f42271f0f1cca","_number":1000823,"virtual_id_number":1000823,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-08-13 12:16:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-13 11:23:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-13 12:16:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a7b1712df4c8530af9db3502f1c37186f492d3eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-08-13 10:48:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"21fd6fddaa7244814255aaef26d77d7ac2f6be46","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 11:23:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2158f64bcbb047fba5c9f9586334ba4b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1c76a04ac2d247b396a497859713b8fa : SUCCESS in 31m 45s","accounts_in_message":[],"_revision_number":1},{"id":"5fccead94f781bf290d6f532040f42271f0f1cca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 12:16:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/33009eb607c24f03a746ad406d5f1a4c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34c57e30b51e40ff82c5d30ae73776d2 : SUCCESS in 10m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a578ebac0ca4667970b318bb6d58be4 : SUCCESS in 9m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/1e8326215db348af86c56a69b7d684f9 : SUCCESS in 1h 04m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f54aeade944c43fbbd3753c10cad7b26 : SUCCESS in 1h 06m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fc9eb2fc2a064d408d8369f0eeee00c3 : FAILURE in 7m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/74d3ea650d1c4d7aac1a3cee15c1338f : POST_FAILURE in 11m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f2671394c68e4b3abd97084a5f824929 : SUCCESS in 23m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2e7d91e92de0440287616cc54ab6bdeb : SUCCESS in 22m 15s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5d1192371cee4cc9914a935f84bc97c7 : SUCCESS in 11m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f590b9d7a48544c3be5081918486c660 : SUCCESS in 5m 17s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5b46308af18d41ea8ff990e7fecda7fc : SUCCESS in 23m 42s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/cb6d3678de0245e79e0fc4ffb497a042 : SUCCESS in 23m 22s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/bf691245bdd14136a82b9798d99d473b : SUCCESS in 26m 02s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/99f13c5217004e009757f799a531cd75 : SUCCESS in 20m 55s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4f46d8418ba446e2a08a8d39a49cb5ed : SUCCESS in 15m 25s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/45e26af5ca6743ed81ec35727778fdad : SUCCESS in 15m 46s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9416e672b80048adb7c80c8c13fb7c48 : SUCCESS in 15m 27s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/93e0901fddb5409595bddd57bf8afd6f : SUCCESS in 11m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8195dfea9d0f44869d900233b76c8ec3 : SUCCESS in 4m 51s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a1e2019594874505b1e1be470c5283b8 : FAILURE in 7m 47s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/f25b6d912c454dada93a90c29fa746fc : SUCCESS in 9m 00s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c08babc9652a4661a1db2b1b09f8300a : SUCCESS in 40m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/493e5b943e3740e48c041ce587c74051 : SUCCESS in 52m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3e9bdf0a29064fc6a5da0df6f371680e : SUCCESS in 52m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/24b38b37e00442d380ff22fbc23cae92 : SUCCESS in 28m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a57c51ef01bc46f089cd16d39817fbe7 : SUCCESS in 4m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/92c648eba2fa4b26a23b499b407cf232 : SUCCESS in 9m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d623f08e9a594eae9bc4e079ff80b2e6 : FAILURE in 8m 47s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"503c7fff2d87cc2d2d82014f4ffb0310c1358e5a","revisions":{"503c7fff2d87cc2d2d82014f4ffb0310c1358e5a":{"kind":"REWORK","_number":1,"created":"2026-08-13 10:48:13.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/23/1000823/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/23/1000823/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/1000823/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/1000823/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/1000823/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/23/1000823/1"}}},"commit":{"parents":[{"commit":"0b95488b21e984bd9b5aed9bc7c18be39648d83f","subject":"slo: Pass segment HEAD headers to manifest hook","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0b95488b21e984bd9b5aed9bc7c18be39648d83f"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-08-13 09:37:41.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-08-13 10:45:23.000000000","tz":60},"subject":"slo: refactor to simplify heartbeat response paths","message":"slo: refactor to simplify heartbeat response paths\n\nMove some duplicated code segments to a single helper function.\n\nChange-Id: I406bb3f5b103ca27096ebfc1acae45d5339a8520\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/503c7fff2d87cc2d2d82014f4ffb0310c1358e5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/503c7fff2d87cc2d2d82014f4ffb0310c1358e5a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
