)]}'
{"id":"openstack%2Fswift~1000844","triplet_id":"openstack%2Fswift~stable%2F2026.1~Ic14839c82596ea8db89c8e5c7f27ccd705be7beb","project":"openstack/swift","branch":"stable/2026.1","hashtags":[],"change_id":"Ic14839c82596ea8db89c8e5c7f27ccd705be7beb","subject":"CHANGELOG for 2.37.3","status":"MERGED","created":"2026-08-13 14:12:05.000000000","updated":"2026-08-13 19:20:01.000000000","submitted":"2026-08-13 19:18:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1000844","meta_rev_id":"bb7a3a2165ff9a4191957b7be4944a76f4a69cd2","_number":1000844,"virtual_id_number":1000844,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-13 19:18:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-08-13 18:12:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-08-13 18:12:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-13 14:38:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e231434cf9b759d11ebf8ddb31ad70725696f427","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-08-13 14:12:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e8d2a0f43fb4a5ca5f3c57d43b920cc3d9af52e9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-08-13 14:31:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"332199104440d43c5f1c3b5f840b53c5a163f6cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 14:38:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1be649e76a6448318ce86c5d46fc67ce\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4fb6fc80b824f8ab62b1cdc0b386b86 : SUCCESS in 6m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/880c3ded83194fcb9be7aae787b7aace : SUCCESS in 5m 05s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/48bf2a9084764bf493921c712fdce40d : SUCCESS in 5m 59s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/bd4012d4f6444956a898278849af48c5 : SUCCESS in 10m 43s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/98d50ef9c8db4cc1a7c1abeb2afc3265 : SUCCESS in 11m 07s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/24c6d440906a47da939a4e160b3be967 : SUCCESS in 11m 30s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/dddc8867cceb410ba4fddf8874a7b915 : SUCCESS in 9m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5361dc405213483898fe5875bb37f557 : SUCCESS in 11m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71a9d8cf014f4b01b3f53470294010d0 : SUCCESS in 2m 41s","accounts_in_message":[],"_revision_number":1},{"id":"de142b6cfd4d3022cea1e418bd95a11b4258c58f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 15:11:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c02726a9407a4e468d60170ccbaa1ddd\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/befe36a1bbcd4ca19717d92b216951e3 : SUCCESS in 57m 11s","accounts_in_message":[],"_revision_number":1},{"id":"7ca779f28421a5160bcfa515629f755e1ccf1e1c","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 15:27:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d33bcddb410f4f8aa1c3e06de480d76e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3b92be244d644dbfaa0ba369bf10296a : SUCCESS in 50m 32s\n- swift-tox-func-encryption-py312-arm64 https://zuul.opendev.org/t/openstack/build/5bf608fd898c4f4ca88096f7812dfa0c : SUCCESS in 34m 18s\n- swift-tox-func-py312-arm64 https://zuul.opendev.org/t/openstack/build/9fd9187dd93643cdad06afc405c0d2d2 : SUCCESS in 34m 30s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9554d3528e204aeab1ea866183911218 : SUCCESS in 8m 28s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3ca96fb8f7cf4407976b503998ab9e5f : SUCCESS in 16m 24s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/fe4ad6c7b11e4bddb49ef37291756779 : SUCCESS in 21m 20s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/15091f6865de49d2be7dacd83320c50a : SUCCESS in 21m 13s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/5265397bfb1e48c8a6935765970584b7 : SUCCESS in 9m 52s\n- swift-tox-func-encryption-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/7dd6d47341f449c7a77b5b4aae4d0ff0 : SUCCESS in 15m 28s\n- swift-tox-func-ec-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/1fd242ba938a4773b3e49461584e3542 : SUCCESS in 12m 44s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/dcd48d7d96d84fa695e590e875d4d19c : SUCCESS in 9m 29s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/8219719bbd7c4841a4a8f98ba1921144 : SUCCESS in 17m 32s\n- swift-multinode-rolling-upgrade-yoga https://zuul.opendev.org/t/openstack/build/b3d130438b18491387356fa87e714835 : SUCCESS in 10m 10s\n- swift-multinode-rolling-upgrade-zed https://zuul.opendev.org/t/openstack/build/28d609757e874869991c75b2d9d95b14 : SUCCESS in 19m 37s\n- swift-multinode-rolling-upgrade-antelope https://zuul.opendev.org/t/openstack/build/0d540483653f4d47a38b5d201969be08 : SUCCESS in 19m 51s\n- swift-multinode-rolling-upgrade-bobcat https://zuul.opendev.org/t/openstack/build/e3f71fe507d941928938334ff99035b4 : SUCCESS in 16m 59s\n- swift-multinode-rolling-upgrade-caracal https://zuul.opendev.org/t/openstack/build/7af5ec05672a4fd199dbfff56ce22768 : SUCCESS in 18m 03s\n- swift-multinode-rolling-upgrade-dalmatian https://zuul.opendev.org/t/openstack/build/cc0dc31b5b2745fc91ef066b824df8c3 : SUCCESS in 21m 20s\n- swift-multinode-rolling-upgrade-epoxy https://zuul.opendev.org/t/openstack/build/bc05069b6bba464787915d578a42abb1 : SUCCESS in 17m 51s\n- swift-multinode-rolling-upgrade-flamingo https://zuul.opendev.org/t/openstack/build/1d58e9e130d6442bbbdc0aefda61da45 : SUCCESS in 17m 15s","accounts_in_message":[],"_revision_number":1},{"id":"b0ddfa2896cb8874b33752a9d9d7ae53c1ae7123","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-08-13 18:12:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"59da6bc141a29dbf049b8facbb5604c3db508ec2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 18:12:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4f8741730e49a00bfab8dd2fd5c66647524efb96","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 19:18:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2b592272a89417484294383a810655e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/590ead42258743b288fa94beb3a6f2a6 : SUCCESS in 10m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b59cf5e24f864bb3910bbd32f0aa889e : SUCCESS in 9m 13s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/d13de6e9bb9b47a7b6bd48ac8d0f92d9 : SUCCESS in 4m 25s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/760ab1a35fcb42a89a7119fa49472d23 : SUCCESS in 16m 43s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/7bd2dec739ab4c358d39606020fcfc00 : SUCCESS in 21m 13s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/9b41538e18b749b599fe7c77eb3bdd53 : SUCCESS in 25m 47s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/c96dc0c401964053ad746cf4bc06ad51 : SUCCESS in 9m 16s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9dcb367559cb4ff6891e098cf31f1a25 : SUCCESS in 12m 54s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ebe2c2ddcb554d54810cd68fa47a7056 : SUCCESS in 15m 11s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4001f2bee1a2425aa9f683c4aa704ef6 : SUCCESS in 14m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e83cc921ed9440938fb1237d3ee952ab : SUCCESS in 6m 26s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/0c2c9fa70dc54b728552018ef4950833 : SUCCESS in 11m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cc34628b33d746d6b8ce304ace74867c : SUCCESS in 26m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3af1a8445f6c42249186e17ec061f0de : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":1},{"id":"5ba67d6d118b45614028157e7942e339b146e62c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 19:18:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"bb7a3a2165ff9a4191957b7be4944a76f4a69cd2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 19:20:01.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/54a94b6274594ec5b8f64d581a0025a9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db7f3bf8a0e548e4b84402f155307b59 : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c1a0426dd6d249c29c9bb4640fdd1fa5 : FAILURE in 46s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/ca6349e9d657474281d88edde554e3eb : SUCCESS in 43s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5104042f12efcaf90c28b656a378da8959004826","revisions":{"5104042f12efcaf90c28b656a378da8959004826":{"kind":"REWORK","_number":1,"created":"2026-08-13 14:12:05.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/44/1000844/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/44/1000844/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/1000844/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/1000844/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/1000844/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/44/1000844/1"}}},"commit":{"parents":[{"commit":"f3753adc065cb4ae9b9fb342e3b386f2775b881a","subject":"checksum: Don\u0027t crash when AF_ALG is denied","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f3753adc065cb4ae9b9fb342e3b386f2775b881a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-08-13 14:11:44.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-08-13 14:11:44.000000000","tz":-420},"subject":"CHANGELOG for 2.37.3","message":"CHANGELOG for 2.37.3\n\nChange-Id: Ic14839c82596ea8db89c8e5c7f27ccd705be7beb\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5104042f12efcaf90c28b656a378da8959004826"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5104042f12efcaf90c28b656a378da8959004826"}]},"branch":"refs/heads/stable/2026.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
