)]}'
{"id":"openstack%2Fswift~1000845","triplet_id":"openstack%2Fswift~stable%2F2025.2~Ie28d7b6eaa78ac99e3e68e2bd17f80a99d34b648","project":"openstack/swift","branch":"stable/2025.2","hashtags":[],"change_id":"Ie28d7b6eaa78ac99e3e68e2bd17f80a99d34b648","subject":"CHANGELOG for 2.36.3","status":"MERGED","created":"2026-08-13 14:19:51.000000000","updated":"2026-08-13 19:20:16.000000000","submitted":"2026-08-13 19:18:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1000845","meta_rev_id":"03992f1b01098883f129b11e243d612b0abe7f7e","_number":1000845,"virtual_id_number":1000845,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-13 19:18:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-08-13 18:12:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-08-13 18:12:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-13 15:04:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6fa88465c414a348c7de697875d646378969a473","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-08-13 14:19:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cfba2d8f6ad2224b3494b6a552154052dfb2646a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-08-13 14:32:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"56d7eab225818891a595aef94d338237c896a19a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 15:04:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51e2747dbd08478f862fcdb1683e594f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47b976f53ad94b8f920d669e438918cf : SUCCESS in 7m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/247451da6abd49d79f98aecc235506b6 : SUCCESS in 9m 21s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a7a1eccc5bd74399805023906cd98ab3 : SUCCESS in 4m 33s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ea2298c474bd47ac90c87e0aa57acfe6 : SUCCESS in 26m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c304c873f2be42a192c732d25ca1ebb1 : SUCCESS in 11m 57s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8ceb23616328431ea1bd764eb2aee3f3 : SUCCESS in 14m 36s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c029651da5d74478aabe0a430640f92a : SUCCESS in 22m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c2bdd8e3faeb4ed7bbaf51175b69b068 : SUCCESS in 37m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe35776b8acd479fb6fdc7477b050a83 : SUCCESS in 2m 38s","accounts_in_message":[],"_revision_number":1},{"id":"29f4acf3fab6bd5b9ff8389d7bfd8f1858d9c0a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 15:28:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff9c7167694b400a88d484d685321e34\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0ce654426a95478299c49eb6d76fc11c : SUCCESS in 58m 33s","accounts_in_message":[],"_revision_number":1},{"id":"82de35d516c2a2de27d236dbcd62be6207675cf6","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 15:33:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ce7fe527452406599b929a3acb91595\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5a725c8afd8746fba9ac2cffe96ab4c8 : SUCCESS in 52m 47s\n- swift-tox-func-encryption-py312-arm64 https://zuul.opendev.org/t/openstack/build/15ab5494fd5848178d8aff8235996fe8 : SUCCESS in 23m 27s\n- swift-tox-func-py312-arm64 https://zuul.opendev.org/t/openstack/build/3e0377d51645437b9190572e6404b5e4 : SUCCESS in 31m 13s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/56dd3df61e884b0ca7da4039400d06b3 : SUCCESS in 22m 42s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/18d51ddd9ec5476f8a30b5fc7924cc30 : SUCCESS in 21m 09s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/4d7b25dbf9b54ec7ab790583f2dc546c : SUCCESS in 14m 34s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/bf1a09ef243840ca8218d9a3fa7bab33 : SUCCESS in 14m 28s\n- swift-tox-func-encryption-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/38eb41eb68be42bb9b25eabd223aa6cc : SUCCESS in 17m 21s\n- swift-tox-func-ec-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/b6e53b0263c34a8db20d2a6b4e5fca5b : SUCCESS in 17m 34s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/5e4dc3d8696a42de8076924b2dbebf31 : SUCCESS in 17m 07s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/4d3dbb1ab6314793813640b7a97bec43 : SUCCESS in 8m 58s\n- swift-multinode-rolling-upgrade-yoga https://zuul.opendev.org/t/openstack/build/c525659f41c94e2898382cb6d72f9acd : SUCCESS in 10m 45s\n- swift-multinode-rolling-upgrade-zed https://zuul.opendev.org/t/openstack/build/8d4432ee012c460289aef3353ebf3027 : SUCCESS in 16m 15s\n- swift-multinode-rolling-upgrade-antelope https://zuul.opendev.org/t/openstack/build/88714cab242145d7b9ba9806086dc8a8 : SUCCESS in 19m 14s\n- swift-multinode-rolling-upgrade-bobcat https://zuul.opendev.org/t/openstack/build/75f303fd95f6427c89be6e08d7dec571 : SUCCESS in 18m 46s\n- swift-multinode-rolling-upgrade-caracal https://zuul.opendev.org/t/openstack/build/491bc045acb14545b1d19e519bab0a6d : SUCCESS in 16m 45s\n- swift-multinode-rolling-upgrade-dalmatian https://zuul.opendev.org/t/openstack/build/ea1c2b09e24e41919eea5f039265c5b4 : SUCCESS in 9m 48s\n- swift-multinode-rolling-upgrade-epoxy https://zuul.opendev.org/t/openstack/build/e4b3a7d6b0a8409ba495c466b1bd9755 : SUCCESS in 20m 20s","accounts_in_message":[],"_revision_number":1},{"id":"341bf1eb76ec35d03d8df65a0ceef92a49dcf276","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-08-13 18:12:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cc0245f3ff9dedc62f87318a0f429c2beaa7b8d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 18:13:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e96dff03f545d790a70effbd02409d853020c85c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 19:18:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbf0fd1a86d242dab29a6d31828bbfcc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97090324e67047bdb46492fd1329407e : SUCCESS in 5m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a380190b2b24459591d38502155ec9d7 : SUCCESS in 9m 30s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/2696e1a982bd4fdea84213b628f85865 : RETRY_LIMIT in 12s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ea2eb42a84ce42ce8d366a9cfbb3f936 : SUCCESS in 35m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/bcdfe19a1be3449fa50be115ca59e591 : SUCCESS in 15m 51s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0cf9c6bfea5f408e8b4d01dc53d91dc9 : SUCCESS in 10m 54s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/37e17148bea14c138ddb5b4a838f73c6 : SUCCESS in 30m 41s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/740056aadef4417bb943622e2b02625a : SUCCESS in 15m 08s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/98780209dc98431ab989f55c6af333df : SUCCESS in 9m 21s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6c34fed85b6f497582d1cf382383283d : SUCCESS in 16m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/03368da3a8a940439f964abc3734e81d : SUCCESS in 6m 17s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/1a894c2fad684c20b9b73bff183ead88 : SUCCESS in 9m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f6e4f6d80f944b7da1c7fcd7defb80b3 : SUCCESS in 26m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f030c30de1c4cc5abc95c6ad77070e0 : SUCCESS in 2m 33s","accounts_in_message":[],"_revision_number":1},{"id":"02a25c9c6365ebfe1c906d0f024d21912c029289","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 19:18:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"03992f1b01098883f129b11e243d612b0abe7f7e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 19:20:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b9273fe58ad4ff4acf42679148d2dc0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8da88c8c8a7941bcbc860f96bab88520 : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53611e4fa23b42c99cf06b9992c9808c : SUCCESS in 39s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/2f1470290beb4f8a84604ac7c12c3026 : RETRY_LIMIT in 6s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"deaa3e4691bb19224f5ef57e7fbcd08b7eff5277","revisions":{"deaa3e4691bb19224f5ef57e7fbcd08b7eff5277":{"kind":"REWORK","_number":1,"created":"2026-08-13 14:19:51.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/45/1000845/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/45/1000845/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/45/1000845/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/45/1000845/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/45/1000845/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/45/1000845/1"}}},"commit":{"parents":[{"commit":"06650c8dbb8a8650804e620729bdf0ecb7d714fb","subject":"checksum: Don\u0027t crash when AF_ALG is denied","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/06650c8dbb8a8650804e620729bdf0ecb7d714fb"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-08-13 14:19:21.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-08-13 14:19:21.000000000","tz":-420},"subject":"CHANGELOG for 2.36.3","message":"CHANGELOG for 2.36.3\n\nChange-Id: Ie28d7b6eaa78ac99e3e68e2bd17f80a99d34b648\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/deaa3e4691bb19224f5ef57e7fbcd08b7eff5277"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/deaa3e4691bb19224f5ef57e7fbcd08b7eff5277"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
