)]}'
{"id":"openstack%2Fswift~1000847","triplet_id":"openstack%2Fswift~stable%2F2025.1~I61ff255511e6613a2766195315ea1cbc1eb1e169","project":"openstack/swift","branch":"stable/2025.1","hashtags":[],"change_id":"I61ff255511e6613a2766195315ea1cbc1eb1e169","subject":"CHANGELOG for 2.35.4","status":"MERGED","created":"2026-08-13 14:26:01.000000000","updated":"2026-08-13 19:20:16.000000000","submitted":"2026-08-13 19:18:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1000847","meta_rev_id":"c8fdd7eca34e952be713ac76736f7e16d226e7d0","_number":1000847,"virtual_id_number":1000847,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-13 19:18:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-08-13 18:12:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-08-13 18:12:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-13 15:07:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1869ae133692f7263124327145266dc14367fcf7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-08-13 14:26:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ed52d317edeafeaa615228b4787da52f9ece6548","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-08-13 14:32:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a6e4b0675933aca316cf9bb5634b469199bd427e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 15:07:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e7e31c4b2994a9583da87cd138afead\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1180caa96970447fa4cad2f0b2175abe : SUCCESS in 7m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/32a10503381f4a7eb206460ac4444823 : SUCCESS in 10m 55s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/789db9b5194d46e1a734e0c6c816cde1 : SUCCESS in 6m 28s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bc44af06f8604e0896fff4a1e2561be3 : SUCCESS in 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5a68c46c95bf43e3928d39155cbdd6b3 : SUCCESS in 9m 47s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9f54029218904e66b7616eb389cf92ca : SUCCESS in 33m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e1baa27394974e40b120344b1c190457 : SUCCESS in 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6130f94cf2e5484d87d890c6bab8dcc3 : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":1},{"id":"dd43fdae771662447a99df268d4cad205a9e921f","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 15:39:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8ef6cb396db496e9c8eb154db9fd59b\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/14b04a8e786946e0a0adc15be551bb9c : SUCCESS in 48m 26s\n- swift-tox-func-encryption-py312-arm64 https://zuul.opendev.org/t/openstack/build/01b6e72d80284f0cb6b3fe428c1ef74c : SUCCESS in 34m 18s\n- swift-tox-func-py312-arm64 https://zuul.opendev.org/t/openstack/build/3017c90d3f6445efb61de3cb1df1b26a : SUCCESS in 28m 06s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/18f291207c6a414f8ed5d5eaa1aebfa4 : SUCCESS in 28m 36s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/94356d3028564d6f996d4704a7b1136b : SUCCESS in 22m 48s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/a865e49b948840fcb47ca5914f66aeef : SUCCESS in 21m 39s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/b32f3592fe444f60be413e73b377f21c : SUCCESS in 21m 14s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/4f394548eb054ca3a5bdf8dded16e28e : SUCCESS in 14m 25s\n- swift-tox-func-encryption-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/463b7020852343be88141e09cbf92ac8 : SUCCESS in 20m 07s\n- swift-tox-func-ec-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/213617610b894db7955895329634cff3 : SUCCESS in 15m 13s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/60e71e9386ac48b380e2e0fd2e08074f : SUCCESS in 19m 34s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/e6e71a142e3b47638aed247142a6176d : SUCCESS in 18m 08s\n- swift-multinode-rolling-upgrade-yoga https://zuul.opendev.org/t/openstack/build/050955e2dc3d4cb287516dc6ed11ee8f : SUCCESS in 18m 40s\n- swift-multinode-rolling-upgrade-zed https://zuul.opendev.org/t/openstack/build/913e485c12d746f7a96e60effa8e6049 : SUCCESS in 19m 47s\n- swift-multinode-rolling-upgrade-antelope https://zuul.opendev.org/t/openstack/build/ececf7ba30c84e77823da007be9ceb04 : SUCCESS in 17m 11s\n- swift-multinode-rolling-upgrade-bobcat https://zuul.opendev.org/t/openstack/build/2041bd30810048a88d44dd202f3ac0be : SUCCESS in 18m 42s\n- swift-multinode-rolling-upgrade-caracal https://zuul.opendev.org/t/openstack/build/975b985b864944e5bb362d42feaacf5a : SUCCESS in 20m 53s\n- swift-multinode-rolling-upgrade-dalmatian https://zuul.opendev.org/t/openstack/build/02039438a5aa46588512b0337d8515c4 : SUCCESS in 19m 52s","accounts_in_message":[],"_revision_number":1},{"id":"e7707f464319c4efd9d836b83b39c6eb94542a28","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 16:26:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2d03c07f0054fc1bd25fc6d6a5424e6\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d1521261ca5f4842a095d64b259a2fff : SUCCESS in 55m 44s","accounts_in_message":[],"_revision_number":1},{"id":"429ae7dee8604e0b20e3fc7902ffb552d35974ca","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-08-13 18:12:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dba0ef32430f670e3f055bb14cf42a4634ecfccd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 18:13:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2a02c02c58672f1edb01f387a53006e77aeb6b7e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 19:18:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f02bea1d762417cbb09f7456e953145\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/785b91fc3fc84f049ea12d2d01406a3c : SUCCESS in 6m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/474d6a5343da4ea19b4e04822635f4fa : SUCCESS in 8m 46s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/1b4f739854754c89a1b11378e7a8c6f0 : RETRY_LIMIT in 16s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/747535c8c9bf430283902e5338a87482 : SUCCESS in 50s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/84b1dac44daf48a29125f4d50ed9dd99 : SUCCESS in 21m 17s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/efbaf7a973fb4e35a777669a0e804c66 : SUCCESS in 23m 35s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b1a8b809126b40978afcc18ec2670823 : SUCCESS in 15m 11s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/558f786b4a2647a38743d65fc887ed4f : SUCCESS in 14m 26s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a937f3b4e2b2444b8e5e4267dc3b75c2 : SUCCESS in 8m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b0adc6f4bb154a9191c0fe8c3256f46e : SUCCESS in 7m 01s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0cf2c1cad8ef4fed8aa8ec461fcc4561 : SUCCESS in 9m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e965c4339ab84a8c99d625e6bea1f6b1 : SUCCESS in 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/066eec265f36420eab2360d097eed3f4 : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":1},{"id":"dc944bf0ae5aec459e1278989f67a735993f0af7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 19:18:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c8fdd7eca34e952be713ac76736f7e16d226e7d0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 19:20:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b87772418871428a97ad6c0fc38b90cd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e40e04d67664066b10431be37d6baab : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af67172c7efa437cae384ec4d347e73a : SUCCESS in 41s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/dd50101b4c374ecf81e597ffb402b2be : RETRY_LIMIT in 6s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fc875f93e82edf98805d0eead3e3c961673c3ef8","revisions":{"fc875f93e82edf98805d0eead3e3c961673c3ef8":{"kind":"REWORK","_number":1,"created":"2026-08-13 14:26:01.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/47/1000847/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/1000847/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/1000847/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/1000847/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/1000847/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/1000847/1"}}},"commit":{"parents":[{"commit":"1b41fba6f34c49fdadd766ded1549d670f685c72","subject":"checksum: Don\u0027t crash when AF_ALG is denied","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1b41fba6f34c49fdadd766ded1549d670f685c72"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-08-13 14:25:30.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-08-13 14:25:30.000000000","tz":-420},"subject":"CHANGELOG for 2.35.4","message":"CHANGELOG for 2.35.4\n\nChange-Id: I61ff255511e6613a2766195315ea1cbc1eb1e169\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fc875f93e82edf98805d0eead3e3c961673c3ef8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fc875f93e82edf98805d0eead3e3c961673c3ef8"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
