)]}'
{"id":"openstack%2Fswift~1000853","triplet_id":"openstack%2Fswift~master~I84323a43b81d8440ee33c52c6473efa83e4c545e","project":"openstack/swift","branch":"master","topic":"s3-checksum-persistence","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-08-25 18:13:33.000000000","reason":"Change was abandoned"},"39146":{"account":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"last_update":"2026-08-14 19:59:17.000000000","reason":"\u003cGERRIT_ACCOUNT_39146\u003e replied on the change","reason_account":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"}}},"hashtags":[],"change_id":"I84323a43b81d8440ee33c52c6473efa83e4c545e","subject":"sq? make mpu manifest hook cache errors","status":"ABANDONED","created":"2026-08-13 15:16:52.000000000","updated":"2026-08-25 18:13:33.000000000","total_comment_count":49,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"ed56bad9855ea9ca41099bf7e58301ac8dc4f691","_number":1000853,"virtual_id_number":1000853,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-14 21:06:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-13 15:46:24.000000000","updated_by":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"reviewer":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"state":"CC"},{"updated":"2026-08-13 16:26:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-13 18:49:19.000000000","updated_by":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"reviewer":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"state":"REVIEWER"}],"messages":[{"id":"de4565b7f31202e0e8da9e02b576dfe2fc7a840b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-08-13 15:16:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cd34d9cc2b305c45262eb0e460c740173b3da992","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-13 15:46:24.000000000","message":"Patch Set 1:\n\n(8 comments)","accounts_in_message":[],"_revision_number":1},{"id":"9c580411b7e034314d5d12daa6b69d3d9b0d45e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 16:26:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cf03da63e080420eae4581644bb6f2af\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b619b32b4b742bb837b4aa48171e8c1 : SUCCESS in 5m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d3e30051845a4a1998850bbb965da152 : SUCCESS in 4m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/bd925dcd26254e79aacdbb5ef6621564 : SUCCESS in 1h 05m 07s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/499c7e025d864c288e92688d56479d02 : SUCCESS in 30m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f4e9e9eff4e6473d84af77e3e7082759 : SUCCESS in 52m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c7b52ec66b8041c6ace57ab492004fc9 : SUCCESS in 34m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/019cd2d5e75e48b28fed5a6ce37a6351 : SUCCESS in 20m 41s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8f900bca29ee4dd08702e0e3678ea005 : SUCCESS in 17m 16s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7ba262eb88b84e4a9074c5db8b8689a3 : SUCCESS in 13m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/61c231c863ce4a138d5ad44f8755cfbd : SUCCESS in 5m 29s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/61a425c2acb54c10984b7cd91e4c7d50 : FAILURE in 26m 38s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/5e8f46ae4c6142fab531d867b9dfc081 : FAILURE in 23m 28s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/2e48071a5d6140caab84433f84ed8afd : FAILURE in 15m 46s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/fbb7d641268b4a53a4ec619543d25ffd : FAILURE in 23m 37s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/43a928e2cd2a49d99d39709d305e7ea7 : SUCCESS in 15m 05s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/70d3db8368644999a872f7c1cf0e2528 : SUCCESS in 12m 03s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/df4dcf802c714d6daab670c9d88dd4e4 : SUCCESS in 14m 59s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/2ff33822a8ce4070856dc1adf114cf73 : SUCCESS in 18m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/60888164e1104373a49687a86c051bc0 : SUCCESS in 9m 05s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f7bf414148aa43f8b1f4bfae24cf163f : FAILURE in 5m 27s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/0e3471fbb7214b9290df2873bcf110b9 : SUCCESS in 5m 45s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a5f568d57d2e41bf98870b19a70fd720 : SUCCESS in 51m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/621dd8e819b64e52aa5ec6fc0de49ea5 : SUCCESS in 58m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7218292c0eaa4291b66a542c956ca06d : SUCCESS in 1h 01m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cf72067c87ae435fb7a59a7f481f7970 : SUCCESS in 15m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/758ff3ac90be42e091c77b265a90ecc9 : SUCCESS in 4m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/476becc85e194019833ae402eee82e3d : SUCCESS in 27m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c473e90977f645feae943bbdfa3b8dc5 : SUCCESS in 1h 02m 13s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"0d0c696c50716389086fac4808bb326149358860","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 17:57:47.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/822c92f4068d4e19ba3d7ea1f9488218\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4d204027febe4cf99be84621a05a96ea : FAILURE in 1h 07m 59s","accounts_in_message":[],"_revision_number":1},{"id":"cfb938baf051e96fd0af6be173c4fa91943737ce","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-13 18:49:19.000000000","message":"Patch Set 1: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d5efb234c83303714705f6c23006dfb1cc3e248d","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-13 19:18:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2b345e0328cab47d5037a03cda4a530b2358077e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-13 21:41:05.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"e9696f03951fb49670732f497c75bcd011ef67a9","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-13 21:59:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"45261c4291a38f36bbd5ba94cdb5ac7b67b5a76a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 22:16:24.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b73629e9666547d888330a1ddeb61f17\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7f06f3b1289540319cd8aee06014404a : FAILURE in 33m 17s","accounts_in_message":[],"_revision_number":2},{"id":"c41208774f013a74718aed6af81b68d70ef6fb72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 22:50:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/82665cfdd9ae4200800b7235cc05bf56\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcc3c3d0f9c94c69b48a0a92be45532e : SUCCESS in 5m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b04a0283f48446d9581afd3085a6399 : SUCCESS in 6m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/40e774d6d2774cd285a09b9c2489e7d6 : SUCCESS in 1h 04m 40s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1be02ad354d8483e9dad4457930a9b81 : SUCCESS in 36m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2f8f556e8e394f74bcacb846d36a6533 : SUCCESS in 50m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/58d8498bc83441e8b6b4fca208efc22a : SUCCESS in 34m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/70d9be91aded4740be2c6d240f4d13e1 : SUCCESS in 17m 41s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c590031bdda8419d94cd3ed23451ac11 : SUCCESS in 20m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a6ab7614bc4f4f519080fef14cb12648 : SUCCESS in 13m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/439768c59b394b8bb1eee622c09bf5ed : SUCCESS in 6m 29s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/179cbfd7e8cd46f0a17f25e9543ed222 : FAILURE in 25m 00s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/35d4d7c34c9d462ebd56e0bde20245dd : FAILURE in 25m 50s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/75bd436ca8b04a8b9fd52301454e54bc : FAILURE in 16m 52s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/debd908376ce4d8c9cc349cfd6636c2c : FAILURE in 9m 40s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e10abb4af73446c29770048d6a63d14d : SUCCESS in 8m 18s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5dcba0eef5b54bac9822aef1d86cd7cb : SUCCESS in 15m 57s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0cd54e6dea4b4ffe86c04d7cb3e76be9 : SUCCESS in 13m 43s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/65980eb4becf410a9d5dda552a2efc71 : SUCCESS in 12m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/01f42321ad9b43b892e2cfafa49ccaf8 : SUCCESS in 5m 50s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/87cb719ea37a4a6a8bcde1d5b4eb7397 : FAILURE in 9m 45s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/8682c29c7525448bbc090a738e6ed488 : SUCCESS in 8m 42s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a87bcecb141d4598b884a8659b80c805 : SUCCESS in 41m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/42b22d2441a04b65bba9de3dec6250c5 : SUCCESS in 50m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/298ba1a442ae4f8c8c99420d07e765f6 : SUCCESS in 55m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7c2dfe1adf3148aaae8dbe6474f91ed5 : SUCCESS in 28m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2aa448db4d3426783c096a97e7bfe94 : SUCCESS in 6m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8135fa107c1b4e03acf07d1768395f99 : SUCCESS in 18m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dbebb08e359e4b74998bcc0d1c375806 : SUCCESS in 43m 19s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"97e1e551473192f398d8ee1b1df7cacc7f4cb3cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-13 23:56:47.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"b0ac70ef6d890afeaedee9c153cce0b6d6d921af","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-14 00:00:08.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bcbdf00d08eb439ca1e6b0d6d8d48d4a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2262cd4bfeca4f68aae508b149e2dc8b : NODE_FAILURE Node(set) request 62cdd88034564ad69f3f4f31e8c98eff failed in 0s","accounts_in_message":[],"_revision_number":3},{"id":"7c7b6c0a4b940081f403a3a40fc22af67f209aab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-14 01:03:55.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0629a26209ba4ad79930c3fea2f4d8e4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1a06a39fe2d465ea31fad82c28b7484 : SUCCESS in 10m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/45fd799107e541f8991baced1302888a : SUCCESS in 8m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/2742d4fee517473e9ca47f25a5e41a02 : SUCCESS in 1h 00m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/22204b04e5994e59ab4394476155d509 : SUCCESS in 33m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2a4cdfc9a0cf4260a0c97b96bed43063 : SUCCESS in 50m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6d02d2d23845428abcd8c1ef32b11ad2 : SUCCESS in 27m 09s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/73749200660c445fb6360cf6de32476e : SUCCESS in 10m 14s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cebdb30286b04df3877a1e4a81656ba4 : SUCCESS in 11m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e6826579ea714c8abf82af39f7d87d8a : SUCCESS in 12m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/869dadec7d294ff49354e2994e2d27cc : SUCCESS in 8m 17s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/958e2831a15b4f7eb3ca49b3054b7cdd : FAILURE in 13m 17s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d0c314a45e4a4156a9237e3d5990348e : FAILURE in 21m 49s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/a33e7a589a064641bad18ccaa49f941a : FAILURE in 31m 41s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/434ac8c456e249a993727baa88ed5713 : FAILURE in 21m 05s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/efba97a146b3416f90dac58946b3a632 : SUCCESS in 14m 58s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d8bfa7315df5471c9144917df5f108c2 : SUCCESS in 15m 04s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/237fbbc0b8f04093a5e467c80db9ef80 : SUCCESS in 16m 17s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/9810b61b0e1f4768a35923e3da7b026d : SUCCESS in 17m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fbe4adac96fc4d6bb0a42b8ae65cf177 : SUCCESS in 3m 38s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b9d89f343a594ad9a435aa4e2a2e76df : FAILURE in 5m 02s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/b1e2a80721374a60af101d93d2f8f2e6 : SUCCESS in 9m 00s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7e2252e927054bceb4fbc25c73fedfeb : SUCCESS in 36m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/98b221a3311547ef8a7c13b936241607 : SUCCESS in 54m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c6253c79e6144de2b6422b90769f3571 : SUCCESS in 53m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1b63a8249d6940919fbdc695071e6a25 : SUCCESS in 28m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb291bcce5d54b89ad8d7b79d39bbea4 : SUCCESS in 4m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8f262b63c641464e8c1fefc06616744f : SUCCESS in 17m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e583e8f86be5407296a9de937c94f2e8 : SUCCESS in 1h 01m 28s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":3},{"id":"0eae5471aeb5b502bda562a13f08c046ebdcf222","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-14 01:37:48.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"57a3bf63faa58a2c492c94c0ce71cf02f784977e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-14 01:38:56.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ca95f785e784aa483eb7b3d7a5207fb\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/852a002634db46fdafecf3b7a5a59360 : NODE_FAILURE Node(set) request 568b063f3ed24176b84f5118ae78c71d failed in 0s","accounts_in_message":[],"_revision_number":4},{"id":"317d3f3a9b5c719be0ff42c61214a5a4671cbd23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-14 02:58:17.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c3a74c89d10f49c6a9f53aca399b49b4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8866b40354e4da492067fa4638d20dd : SUCCESS in 11m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/747ab5c4f8ac4bf5bc055e96751f0b7c : SUCCESS in 9m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/3db2ed777b734090b42375e6c62758ae : SUCCESS in 1h 14m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8e00ba9e529840c59d2bd785a0ad8bae : SUCCESS in 32m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/699af401fdb84107a4022cf3a3382844 : SUCCESS in 58m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/19c02950a0f84904be2f5088ca840fad : SUCCESS in 54m 10s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/50fe0dd3e1b2405cbaf8a18782a47452 : SUCCESS in 20m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0c1b80dfc41c4ae0af4ff1c0d37487f2 : SUCCESS in 20m 29s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/149501f3667f489ea4481836aede9f3e : SUCCESS in 21m 10s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0a087981ffa9401eb75986393f7625a6 : SUCCESS in 6m 31s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f438767377d34486b3fe61b660ff0bf1 : FAILURE in 30m 02s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/7b176e1cd52548a3bd69170103c74a9c : SUCCESS in 22m 59s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/395ff50c0bd349ddac094afaeb6b1939 : SUCCESS in 19m 08s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/51412a637ad243699504fac14e497dfd : SUCCESS in 24m 16s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/771feb848490414280e034faaedd7cdf : SUCCESS in 13m 47s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e86c6b73e6034f8f8a69648a16c0b7ef : SUCCESS in 14m 31s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6d4a8695742342faa697c56684bdc1d0 : SUCCESS in 16m 50s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/be073cf9ff4a444fa976597e1e8d1dc5 : SUCCESS in 9m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/de62400ca0404324912b6f7ee489a0eb : SUCCESS in 5m 44s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c190759678c8433baf561e107d716f69 : FAILURE in 7m 29s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/74c9b54df5c84b7eafbc7e4cda28c6c9 : SUCCESS in 9m 37s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/22d0599d9fe142c49fdf11ed0ff7997f : SUCCESS in 1h 04m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/34b620d8193546e291c6c73dca5a954f : SUCCESS in 43m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8ba6e1e2c0b942689db7139db6b63ad2 : SUCCESS in 54m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dd718307f2be42b6b8f4fb9442d55040 : FAILURE in 29m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0107743f257142dea3876d38af148bd9 : SUCCESS in 4m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0c1a6b963f15438484986bcdc26d7a1a : SUCCESS in 16m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ddb2f77caa55431bbf6298d922fa010c : SUCCESS in 27m 10s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":4},{"id":"42ba42ac7e1eadd6a6d8d443ce87e83f90a1a91b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-14 03:22:39.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"49ff76c53301299304fa9fb9f4cd018157413f4c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-14 03:24:03.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/957706532ff14a4d85907b18ac264355\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3203b178a5ec44c597aeb9f8ec4e4a5e : NODE_FAILURE Node(set) request 180b81983b714a11b0cd53ff2fa47bcd failed in 0s","accounts_in_message":[],"_revision_number":5},{"id":"585809cff320d43f2139fe02c4a7440a89753d46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-14 04:27:21.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f92db30fc9a44a579e8fdbf08dbf9e1f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/888277a22b18495489edb4abdc816002 : SUCCESS in 10m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d47cc79ae4f4308851b5ee765476a39 : SUCCESS in 9m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/a9e8443d05be4ecebd17be5478b8c464 : SUCCESS in 1h 03m 46s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6212deac6754423591815be2d61f52fc : SUCCESS in 1h 00m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6728e5ec30be4137a072c565497d0c51 : SUCCESS in 49m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7dc02ad305174620954136c17d0208e5 : SUCCESS in 1h 02m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1304dbc3f35840a5adcfa4615cd59a95 : SUCCESS in 12m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/691394e4c20e40a1b48c826611146575 : SUCCESS in 20m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f50224756a2f41f0a53dba8952d63b90 : SUCCESS in 17m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6ca0dc3ca40a4a6f9f2a8375afb19732 : SUCCESS in 6m 51s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b4da95c1db974b6e926511b9fdfa2f88 : SUCCESS in 26m 48s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/65f9d8abcf544a22b6a7aca381fe5aca : SUCCESS in 19m 37s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/13a0dae4fe5d458085c988d96b6c6786 : SUCCESS in 31m 24s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/e1d0d97c535f40a2a8bc9fb180129425 : SUCCESS in 16m 24s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3fdc0cb34bbc4cbdb8d8060ce3d4e232 : SUCCESS in 14m 58s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/81dbdefc437746ddb7ed1fb2063ba8df : SUCCESS in 15m 31s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7fe2d5bf5be741c3ab5e314c7d1740df : SUCCESS in 15m 27s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/d8a629d436b544ea849184c5ff04be7e : SUCCESS in 15m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/34ef31bddded47bab1c5d5ce4bdbec9b : SUCCESS in 8m 30s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3a3b9968d20a40729eb9eecb6955aeb5 : FAILURE in 4m 17s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/502e4d046b984a49b230af13197f74e7 : SUCCESS in 6m 01s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3bcdc3fb4dc8472cb41383cb5ce31359 : SUCCESS in 58m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/83b5f3d9bd6d47df8b8d070df32cbf1d : SUCCESS in 58m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/997366e2732247c2b58683018ac226fb : SUCCESS in 57m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e443779af8e142729b10aba291b8fe31 : SUCCESS in 21m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40f60471f2ab4f49a985e375192e52d3 : SUCCESS in 5m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/04634c2e393d4c75a72b88049c1f85e0 : SUCCESS in 26m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4f6e5f54555b40b99b2c31c1bfb59f3f : SUCCESS in 53m 37s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":5},{"id":"018326efb090a777c725a9a195fc21cb14334b71","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-08-14 13:28:38.000000000","message":"Patch Set 5:\n\n(6 comments)","accounts_in_message":[],"_revision_number":5},{"id":"7024fd3998ed7d50686eab4e94c2089998bb4fee","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-08-14 13:30:52.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"77af3f00b75c40a5125b4fbd0b802cb1de87ad84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-08-14 13:51:31.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"79e871f5b328b2469cc3f5c87e61b5d90a88959a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-14 13:52:50.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4cf437cd3d95479f86b290d8e9b0914b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0cd52faaa24040e7b28bee8d902f5b49 : NODE_FAILURE Node(set) request 9570b6699796471584a1c35e41c9faf6 failed in 0s","accounts_in_message":[],"_revision_number":6},{"id":"f96f89479f1a841be32a55b54c688df7228914ca","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-08-14 14:40:47.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"149d66328728cedf47e9a008ad2f920545f56c97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-14 15:13:18.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c0482adffd044ed2b30e77bc5b57055d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa1b3ffd79da4127b60ed3e7a3ceb7e4 : SUCCESS in 9m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/172cadf7fdf54040a0d02af06539a339 : SUCCESS in 9m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/36f9f5fcd6db4ef886e69fe1b3d3a86d : SUCCESS in 46m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3f4e59205ef64ab0be3fcf8680370180 : SUCCESS in 1h 00m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e48ddab0a6734854a3fbc4bec800b3a9 : SUCCESS in 53m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0346a23b42644ac0a97d35d8b53e34b1 : SUCCESS in 1h 10m 27s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e11ccb53b83c41489b944b3d171865ab : FAILURE in 15m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/830c8724de484bc3b5f4cace8772ce8a : FAILURE in 19m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/658c2ed28f1c4bd89867f7e16d134000 : FAILURE in 19m 50s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d778622c0eb848d99d44001f96502104 : SUCCESS in 7m 35s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7f34adbb7eec4c4f8cc5b7b04fe2eee7 : FAILURE in 26m 12s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/5c850f1dc57a45adbf8fb331ca95b390 : FAILURE in 9m 28s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/500cbb42f6b54055a359a441f9d171ba : FAILURE in 13m 34s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/ce89e8ed75ba4355b412b9c3b272bf19 : FAILURE in 19m 27s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d636819d1ae64a13a124d555ceba84c3 : SUCCESS in 8m 13s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/127d26657aad411eb8faa1a1795edab5 : SUCCESS in 9m 15s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6be3c758e547446481067e6bb00a5f94 : SUCCESS in 15m 26s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/e959883381b24312b6ce68ef643338d8 : FAILURE in 9m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/70e838930f10498d98cc0cdea233a5c0 : SUCCESS in 6m 28s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e619d56e8d9e4c63b708329fb84fc53c : FAILURE in 4m 16s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/8e67e020e58a4e3186c089d8e5ce70ee : FAILURE in 6m 20s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/729c9b3137de44eb8365fab521a2880a : SUCCESS in 1h 06m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/67cf7cf58b66497bbd557762df2fefc7 : FAILURE in 33m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/22440b2b143a4b56ab4c97a667d9c16a : FAILURE in 17m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8a6845133e57417d9771224936204ff4 : FAILURE in 29m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c75d4f6c3ca14e339dbd9470ce358f41 : SUCCESS in 3m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/66ecd7cac4c34e31b3218e2e069cf69e : SUCCESS in 27m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1b15ec87a1ba4d898b98be3de992402e : SUCCESS in 1h 04m 14s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":6},{"id":"bb6ea6dd5edd70788d5c3a93a3bfa556c737e5cf","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-08-14 15:50:52.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"8234efaa1e3b4886a92c22292bf78f8edd582467","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-14 16:59:02.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"706412008cafcb3ef635c1d7a910eb136e46775b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-14 17:13:37.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"1553c13185ab37e9dbf3bb7fdd0368d5fd8e851c","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-14 17:42:45.000000000","message":"Patch Set 8:\n\n(6 comments)","accounts_in_message":[],"_revision_number":8},{"id":"c79736d2cd7af632b52725d92327e0768fb2fb76","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-14 17:47:31.000000000","message":"Patch Set 8:\n\n(13 comments)","accounts_in_message":[],"_revision_number":8},{"id":"5ab2697df2e16a482180fe0914c4db28f7958ace","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-14 18:01:07.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d852c91d882e4deea9a0021d6891d46f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c9296b9ec9dd4e3d9ec890c6e3f33e8f : SUCCESS in 45m 40s","accounts_in_message":[],"_revision_number":8},{"id":"f157674f36e3700b33ebfd616d52c59ddc33f047","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-14 18:23:06.000000000","message":"Patch Set 8: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0b2783ea13404e93866a09c2d63b5c5a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5f1199f7dc247319099189c07428d23 : SUCCESS in 11m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e4c34699628341fa8c1e421c60446c40 : SUCCESS in 8m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/34188f3bd1bd4de3a6d7e953cd3ba73a : SUCCESS in 37m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/883ec69d7aff4c9cba3a10f6fe51c73b : SUCCESS in 1h 04m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5f7d30afd0944e7486b0c3f2833cfc3e : SUCCESS in 33m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/45902d1b39554e0ba112f79e3d1bda80 : SUCCESS in 1h 00m 05s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5ef8439ae7b6449c99b1cd8c06cf660d : FAILURE in 18m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/061bcd93d9cd439096fcdc44a6e0af27 : FAILURE in 11m 09s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/25f5eb4018044b799008f6128f6dbd06 : FAILURE in 13m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/75fa841cd57b482bb20c30b39ad62dfa : SUCCESS in 5m 02s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/de236385d391412cbfbe0890d208659f : SUCCESS in 24m 44s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/c1d0158041384b428a86e343fa9bb94f : SUCCESS in 19m 53s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/8b7131620ddf448d8118f8c13160d84d : SUCCESS in 14m 35s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/8fc05e74da104d5b8628e3acaac3da3a : SUCCESS in 11m 35s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4b8d220ad01d481e8710efcdba1f4127 : SUCCESS in 16m 48s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0bb58ccf469848449127e11b7f5a6426 : SUCCESS in 14m 58s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/925a6273f4904e90aa03d4ffa172b307 : SUCCESS in 13m 25s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/e893cb6261eb40baac5059311d9952ef : FAILURE in 17m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/20630f1413dc4bb8879f4b3963c28a16 : SUCCESS in 8m 23s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8ff72a3800604fa190203ddbaf7f556d : FAILURE in 10m 07s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/fa448f0e53a647d592d94441fc1611d6 : FAILURE in 9m 33s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6bc77d2905ce49dea74cb75a7d09fa60 : SUCCESS in 41m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5b1deaa97f6343e6ad545e5b167433b4 : FAILURE in 19m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d9360795bc294c5c8abeed4d105f8f96 : FAILURE in 14m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/352d77a81062470489e8c72f6aa58a3f : SUCCESS in 30m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80fb124f3b41485499a4e3d0952db3e6 : FAILURE in 3m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/54f8bf17bddd47d69d3eb81decd2a95b : SUCCESS in 11m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0de64a7f84ba4e5da7e042fc2dcc0c50 : SUCCESS in 25m 27s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":8},{"id":"fef59056b3efe0c64b88fb31fdd4e4292528e01b","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-08-14 18:36:49.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"a78129ae282549f3d8c2cd76bf66cddc9944b2f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-14 19:08:09.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"b070613faa725fbb914078f93d434365fcd658a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-14 19:58:40.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"1d764c92e6239b2040c52ebaa1f1887657759456","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-14 19:59:17.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"46c0819212540775764eb6fcef0baf69b73d4c47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-14 21:06:39.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe3fb95414f64a72ac1b34a82bfc8d62\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d53e46973b543259c49618010ffabef : SUCCESS in 10m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/362ce18b69c14061b2dc2c07cdefb947 : SUCCESS in 8m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/f5cb9bf2670545faa37d0220d9534525 : SUCCESS in 1h 00m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c4b9b13fa2ec4159ab9943b1a1c08040 : SUCCESS in 33m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f1d9704c42b84a689b64f71349c88a9f : SUCCESS in 26m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2d174fa53ac44e01addfd62919b3408a : SUCCESS in 27m 15s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6660fda93bd54e4c86a617d7456bc34f : SUCCESS in 20m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3f0c28d1fff04ad38a42dfa287619220 : SUCCESS in 10m 28s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b806da238aa548f8a9751af5438a8c7d : SUCCESS in 20m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9d49f3ac30124df6a71218a32b7223f9 : SUCCESS in 5m 07s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1529df4232f34c69a7dc224cf0496e2a : SUCCESS in 22m 44s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d7c55bddf0ed4fe1ae4a45b720a77e2f : SUCCESS in 22m 55s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/8e5adfc0e95542d4a1f0ddfd58587943 : SUCCESS in 11m 30s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/43ee4106b7004bb9a5d2c9c6a9051fe9 : SUCCESS in 9m 20s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6338788eba3048d0bf066b15add83c9c : SUCCESS in 14m 42s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1f4430abc1d64dfd9d11dd67958183fe : SUCCESS in 12m 39s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7060e25e5d64473ab43be7ee07c57031 : SUCCESS in 12m 20s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/6abdbf327908412b93ff2fb92d1d816a : SUCCESS in 18m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/20fc74d5e5f5428485271ab4f2e3f610 : SUCCESS in 5m 49s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fba2a5e81b6546808797fac3504895c1 : FAILURE in 7m 15s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/8d2760e4146b40e4b344cb8466ace64d : SUCCESS in 8m 55s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d079ca826a2e4a3ba49a6ea3a760146f : SUCCESS in 51m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/343e6a82c18540dbbfb86dff37ef6540 : SUCCESS in 53m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2fad6554d5d64b399f82908b136964c0 : SUCCESS in 35m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9c143b300ea14abea47dfce4a505b205 : SUCCESS in 27m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f636f3caa53b43cfb4ad382f82f22a5e : SUCCESS in 3m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e7c3e7a6c8fc454992a9796970eb49f7 : SUCCESS in 15m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f3bf020abf314a8d8d2619c43f5f79b0 : SUCCESS in 37m 34s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":10},{"id":"40d438fea7fd1a252c8e517fe449715590899366","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-14 21:13:27.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5858c33bdc654b1fb55f2a1c84fd2b34\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/974bebd7e87d41a5bb92d474e760a7b0 : SUCCESS in 1h 11m 37s","accounts_in_message":[],"_revision_number":10},{"id":"ed5a4276f7f840aaf4fa9e09425b90d6b25154db","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-14 21:59:14.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"4215f27ad397ce81373224d8ae38e56ffaaf45fd","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-14 23:27:00.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"ed56bad9855ea9ca41099bf7e58301ac8dc4f691","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-08-25 18:13:33.000000000","message":"Abandoned\n\nsquashed 993193: s3api: Checksum Persistence on MPUs | https://review.opendev.org/c/openstack/swift/+/993193","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"2eb8c7f18db63c8030245d7ca5057290deb7e851","revisions":{"71edf1c5eb6520d6c42b526c26bd210848cdc432":{"kind":"REWORK","_number":1,"created":"2026-08-13 15:16:52.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/53/1000853/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/53/1000853/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/1000853/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/1000853/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/1000853/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/53/1000853/1"}}},"commit":{"parents":[{"commit":"1734fb8c60562d4f346cb4fb38612206211c46fe","subject":"s3api: Checksum Persistence on MPUs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1734fb8c60562d4f346cb4fb38612206211c46fe"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-08-13 14:00:31.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-08-13 15:08:10.000000000","tz":60},"subject":"sq? make mpu manifest hook cache errors","message":"sq? make mpu manifest hook cache errors\n\nAvoid raising exceptions inside the slo resp iter by statshing them in\nthe manifet hook and replaying when the SLO resp iter feeds an error\nbody back to MPU.\n\nChange-Id: I84323a43b81d8440ee33c52c6473efa83e4c545e\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/71edf1c5eb6520d6c42b526c26bd210848cdc432"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/71edf1c5eb6520d6c42b526c26bd210848cdc432"}]},"branch":"refs/heads/master"},"3b7cf771813203f03f1120babc40d6735accee9c":{"kind":"REWORK","_number":2,"created":"2026-08-13 21:41:05.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/53/1000853/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/53/1000853/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/1000853/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/1000853/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/1000853/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/53/1000853/2"}}},"commit":{"parents":[{"commit":"1734fb8c60562d4f346cb4fb38612206211c46fe","subject":"s3api: Checksum Persistence on MPUs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1734fb8c60562d4f346cb4fb38612206211c46fe"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-08-13 14:00:31.000000000","tz":60},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-08-13 21:40:52.000000000","tz":-420},"subject":"sq? make mpu manifest hook cache errors","message":"sq? make mpu manifest hook cache errors\n\nAvoid raising exceptions inside the slo resp iter by statshing them in\nthe manifet hook and replaying when the SLO resp iter feeds an error\nbody back to MPU.\n\nChange-Id: I84323a43b81d8440ee33c52c6473efa83e4c545e\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3b7cf771813203f03f1120babc40d6735accee9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3b7cf771813203f03f1120babc40d6735accee9c"}]},"branch":"refs/heads/master"},"0e8f91c321ee928bcc727b4d02301bdc1adf547a":{"kind":"REWORK","_number":3,"created":"2026-08-13 23:56:47.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/53/1000853/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/53/1000853/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/1000853/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/1000853/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/1000853/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/53/1000853/3"}}},"commit":{"parents":[{"commit":"5b5c8a2378c798025d8f185f69adecfe98a1c3f3","subject":"s3api: Checksum Persistence on MPUs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5b5c8a2378c798025d8f185f69adecfe98a1c3f3"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-08-13 14:00:31.000000000","tz":60},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-08-13 23:51:40.000000000","tz":-420},"subject":"sq? make mpu manifest hook cache errors","message":"sq? make mpu manifest hook cache errors\n\nAvoid raising exceptions inside the slo resp iter by statshing them in\nthe manifet hook and replaying when the SLO resp iter feeds an error\nbody back to MPU.\n\nChange-Id: I84323a43b81d8440ee33c52c6473efa83e4c545e\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0e8f91c321ee928bcc727b4d02301bdc1adf547a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0e8f91c321ee928bcc727b4d02301bdc1adf547a"}]},"branch":"refs/heads/master"},"848f679b0b9a2bc25fceee62f64f1e0cd054a421":{"kind":"REWORK","_number":4,"created":"2026-08-14 01:37:48.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/53/1000853/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/53/1000853/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/1000853/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/1000853/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/1000853/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/53/1000853/4"}}},"commit":{"parents":[{"commit":"5b5c8a2378c798025d8f185f69adecfe98a1c3f3","subject":"s3api: Checksum Persistence on MPUs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5b5c8a2378c798025d8f185f69adecfe98a1c3f3"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-08-13 14:00:31.000000000","tz":60},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-08-14 01:37:38.000000000","tz":-420},"subject":"sq? make mpu manifest hook cache errors","message":"sq? make mpu manifest hook cache errors\n\nAvoid raising exceptions inside the slo resp iter by statshing them in\nthe manifet hook and replaying when the SLO resp iter feeds an error\nbody back to MPU.\n\nChange-Id: I84323a43b81d8440ee33c52c6473efa83e4c545e\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/848f679b0b9a2bc25fceee62f64f1e0cd054a421"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/848f679b0b9a2bc25fceee62f64f1e0cd054a421"}]},"branch":"refs/heads/master"},"fcd70336a9aec8bf67eb01440004c14eea5b40ee":{"kind":"REWORK","_number":5,"created":"2026-08-14 03:22:39.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/53/1000853/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/53/1000853/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/1000853/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/1000853/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/1000853/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/53/1000853/5"}}},"commit":{"parents":[{"commit":"5b5c8a2378c798025d8f185f69adecfe98a1c3f3","subject":"s3api: Checksum Persistence on MPUs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5b5c8a2378c798025d8f185f69adecfe98a1c3f3"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-08-13 14:00:31.000000000","tz":60},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-08-14 03:22:27.000000000","tz":-420},"subject":"sq? make mpu manifest hook cache errors","message":"sq? make mpu manifest hook cache errors\n\nAvoid raising exceptions inside the slo resp iter by statshing them in\nthe manifet hook and replaying when the SLO resp iter feeds an error\nbody back to MPU.\n\nChange-Id: I84323a43b81d8440ee33c52c6473efa83e4c545e\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fcd70336a9aec8bf67eb01440004c14eea5b40ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fcd70336a9aec8bf67eb01440004c14eea5b40ee"}]},"branch":"refs/heads/master"},"64ec178fb6439b9f98b3e8e0d35d68f843cd9a7e":{"kind":"REWORK","_number":6,"created":"2026-08-14 13:51:31.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/53/1000853/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/53/1000853/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/1000853/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/1000853/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/1000853/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/53/1000853/6"}}},"commit":{"parents":[{"commit":"5b5c8a2378c798025d8f185f69adecfe98a1c3f3","subject":"s3api: Checksum Persistence on MPUs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5b5c8a2378c798025d8f185f69adecfe98a1c3f3"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-08-13 14:00:31.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-08-14 13:18:51.000000000","tz":60},"subject":"sq? make mpu manifest hook cache errors","message":"sq? make mpu manifest hook cache errors\n\nAvoid raising exceptions inside the slo resp iter by statshing them in\nthe manifet hook and replaying when the SLO resp iter feeds an error\nbody back to MPU.\n\nChange-Id: I84323a43b81d8440ee33c52c6473efa83e4c545e\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/64ec178fb6439b9f98b3e8e0d35d68f843cd9a7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/64ec178fb6439b9f98b3e8e0d35d68f843cd9a7e"}]},"branch":"refs/heads/master"},"1f06c0cba921572608b6455c6149c1aaa5d17d89":{"kind":"REWORK","_number":7,"created":"2026-08-14 16:59:02.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/53/1000853/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/53/1000853/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/1000853/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/1000853/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/1000853/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/53/1000853/7"}}},"commit":{"parents":[{"commit":"5b5c8a2378c798025d8f185f69adecfe98a1c3f3","subject":"s3api: Checksum Persistence on MPUs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5b5c8a2378c798025d8f185f69adecfe98a1c3f3"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-08-13 14:00:31.000000000","tz":60},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-08-14 16:58:50.000000000","tz":-420},"subject":"sq? make mpu manifest hook cache errors","message":"sq? make mpu manifest hook cache errors\n\nAvoid raising exceptions inside the slo resp iter by statshing them in\nthe manifet hook and replaying when the SLO resp iter feeds an error\nbody back to MPU.\n\nChange-Id: I84323a43b81d8440ee33c52c6473efa83e4c545e\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1f06c0cba921572608b6455c6149c1aaa5d17d89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1f06c0cba921572608b6455c6149c1aaa5d17d89"}]},"branch":"refs/heads/master"},"c0cec353f95db6b9853d892bb24374a70e5612c2":{"kind":"REWORK","_number":8,"created":"2026-08-14 17:13:37.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/53/1000853/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/53/1000853/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/1000853/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/1000853/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/1000853/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/53/1000853/8"}}},"commit":{"parents":[{"commit":"5b5c8a2378c798025d8f185f69adecfe98a1c3f3","subject":"s3api: Checksum Persistence on MPUs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5b5c8a2378c798025d8f185f69adecfe98a1c3f3"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-08-13 14:00:31.000000000","tz":60},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-08-14 17:12:29.000000000","tz":-420},"subject":"sq? make mpu manifest hook cache errors","message":"sq? make mpu manifest hook cache errors\n\nAvoid raising exceptions inside the slo resp iter by statshing them in\nthe manifet hook and replaying when the SLO resp iter feeds an error\nbody back to MPU.\n\nChange-Id: I84323a43b81d8440ee33c52c6473efa83e4c545e\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c0cec353f95db6b9853d892bb24374a70e5612c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c0cec353f95db6b9853d892bb24374a70e5612c2"}]},"branch":"refs/heads/master"},"acf498a2f10c994de3802263001cb9e1dcd95866":{"kind":"REWORK","_number":9,"created":"2026-08-14 19:08:09.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/53/1000853/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/53/1000853/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/1000853/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/1000853/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/1000853/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/53/1000853/9"}}},"commit":{"parents":[{"commit":"5b5c8a2378c798025d8f185f69adecfe98a1c3f3","subject":"s3api: Checksum Persistence on MPUs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5b5c8a2378c798025d8f185f69adecfe98a1c3f3"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-08-13 14:00:31.000000000","tz":60},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-08-14 19:08:00.000000000","tz":-420},"subject":"sq? make mpu manifest hook cache errors","message":"sq? make mpu manifest hook cache errors\n\nAvoid raising exceptions inside the slo resp iter by statshing them in\nthe manifet hook and replaying when the SLO resp iter feeds an error\nbody back to MPU.\n\nChange-Id: I84323a43b81d8440ee33c52c6473efa83e4c545e\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/acf498a2f10c994de3802263001cb9e1dcd95866"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/acf498a2f10c994de3802263001cb9e1dcd95866"}]},"branch":"refs/heads/master"},"2eb8c7f18db63c8030245d7ca5057290deb7e851":{"kind":"REWORK","_number":10,"created":"2026-08-14 19:58:40.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/53/1000853/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/53/1000853/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/1000853/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/1000853/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/1000853/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/53/1000853/10"}}},"commit":{"parents":[{"commit":"5b5c8a2378c798025d8f185f69adecfe98a1c3f3","subject":"s3api: Checksum Persistence on MPUs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5b5c8a2378c798025d8f185f69adecfe98a1c3f3"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-08-13 14:00:31.000000000","tz":60},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-08-14 19:58:32.000000000","tz":-420},"subject":"sq? make mpu manifest hook cache errors","message":"sq? make mpu manifest hook cache errors\n\nAvoid raising exceptions inside the slo resp iter by statshing them in\nthe manifet hook and replaying when the SLO resp iter feeds an error\nbody back to MPU.\n\nChange-Id: I84323a43b81d8440ee33c52c6473efa83e4c545e\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2eb8c7f18db63c8030245d7ca5057290deb7e851"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2eb8c7f18db63c8030245d7ca5057290deb7e851"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
