)]}'
{"id":"openstack%2Fswift~1001446","triplet_id":"openstack%2Fswift~master~I9c44aad3e316d520cb8721e206221fd262fb77ef","project":"openstack/swift","branch":"master","topic":"relinker-audit-default","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2026-09-02 20:33:39.000000000","reason":"Change was abandoned"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2026-09-02 18:07:29.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},"38767":{"account":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"last_update":"2026-09-02 20:33:39.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I9c44aad3e316d520cb8721e206221fd262fb77ef","subject":"relinker: Require specifying a policy for audit","status":"ABANDONED","created":"2026-08-19 00:09:01.000000000","updated":"2026-09-10 16:20:03.000000000","total_comment_count":15,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a58577bab60631e4388ee242b402aaed0eb06953","_number":1001446,"virtual_id_number":1001446,"owner":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-22 00:16:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":1,"date":"2026-09-02 18:07:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-19 00:45:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-19 01:14:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-20 18:53:09.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"},{"updated":"2026-08-21 07:24:37.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2026-09-02 18:07:29.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"6c9655ed9e5e5e13f88243fad0d9b9ab6426b88f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-08-19 00:09:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"96c7e91927a34a0f2658bc70bb35905787bc6be1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 00:45:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec91e4375a7542879e6e966c2e444322\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/aefa32474cc24d1bb26978fd33e2971a : SUCCESS in 34m 31s","accounts_in_message":[],"_revision_number":1},{"id":"352a019c8c471bdec9f0c251a2eeadca2e2415b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 01:14:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9dd4491b22d40e9829b9b8167247f52\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/417bf2fec534479980def063e8f14df6 : SUCCESS in 8m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa192afd3ddb4d86ae2ce20a79904b78 : SUCCESS in 4m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/ab05f4b302134b4489446ef52141d4af : SUCCESS in 30m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/90be64badef24a31b866282d5a9f3aa2 : SUCCESS in 43m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4191a1a579fb46c1af5c37c5c4f59494 : SUCCESS in 50m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1cf24a16ceef4399adf21c7cdb6f662b : SUCCESS in 32m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c9700f832b9542fc98cd7ad6134a848c : SUCCESS in 18m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/be423e0eb63b4ad98ce8360d04be5330 : SUCCESS in 18m 37s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/91c4c788e6614af7be121388b0605671 : SUCCESS in 12m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8b6f4c4bd764408faf4f725607760510 : SUCCESS in 5m 50s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1e37256839d444ba97a72916936f4039 : SUCCESS in 18m 10s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ac1cded2e43d4b5a8da2f7b8fe5692d1 : SUCCESS in 15m 53s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/10ce64888ccc4c9e8f5fe0f65bca710a : SUCCESS in 32m 56s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/c663bb2bd5814d68934705d19ffb59f3 : SUCCESS in 14m 34s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e6a0dea5dacc443fb5fdff3bd64ffc0c : SUCCESS in 14m 40s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/22089a58789b4b5d9a153f05015e002d : SUCCESS in 14m 33s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d457a34c24c24191895abee410e251c1 : SUCCESS in 8m 32s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/aa908f4206754d88a48474acbd0ab1b8 : SUCCESS in 10m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1ef8d2c7a94e432b81cd438bf28679c3 : SUCCESS in 4m 11s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d7d7235c2947420e9d1595e084528ead : FAILURE in 4m 56s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/93e3fc025dd54adc9738d4925acf67b6 : SUCCESS in 6m 21s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/645a99562a34443880791de177aef714 : SUCCESS in 1h 04m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fe9f302b3df44556ad4463cc05388c89 : SUCCESS in 26m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9d9e2f0d07684819be36bc9b262eb041 : SUCCESS in 35m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ae1d44f12ce4f288c7dce74fd2254e4 : SUCCESS in 12m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cfc5f9c513f46dcbb6ceeb66ca78197 : SUCCESS in 2m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/92c00009b9f940fc9cbd06bb2fd692cc : SUCCESS in 14m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b5e59663658c4c019dccb4a5ce441213 : SUCCESS in 31m 00s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"3129158b34f681a7c8f6b1a48868e4e0f699a519","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-08-20 00:37:27.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"87183c4a278fa4db509fe6bf5dc80d1343bacefb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 01:52:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6deb376572140f99e14c23483aae0f4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91d329045e6f4e9cae571a9ae06f472f : SUCCESS in 10m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e80b127822564eb98490f1153969fe4d : SUCCESS in 8m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/a63dd9ab73b644838596b31279e4d6a6 : SUCCESS in 1h 08m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5b959e4801784b6e874ed5bd19997d14 : SUCCESS in 1h 02m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ea4918e9eb904dd99ffc346585959cf4 : SUCCESS in 29m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/49058110aef54e79b19ee8572255b103 : SUCCESS in 57m 23s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d808e403c36f4c08a1b8aedba570a1bb : SUCCESS in 20m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d26b4a07ba914110875bb416483ef924 : SUCCESS in 19m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/922b2f58c76c46c69c2536c5934f188d : SUCCESS in 19m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/869a16a99a5d4cecbba5820bf2dfbe7e : SUCCESS in 5m 53s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1f28c348c5e647cc89bbf2de3c068148 : SUCCESS in 26m 15s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/c278a72a10764b70bf71ee23a025ea97 : SUCCESS in 10m 31s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/1c7ea6dd46814214991c7c1de8d0ca93 : SUCCESS in 33m 22s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/5c129b3a36054e6a8b0c243e272951f0 : SUCCESS in 10m 17s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c48acf23bd59443e8eca92b531f338fb : SUCCESS in 8m 58s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3be9f9c2a12f4bb99b43f54586b2b25c : SUCCESS in 15m 06s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f3e020bfea32404babac08579b2a2a92 : SUCCESS in 8m 25s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/b8b9103c27ec45e8adf5dd58920ab889 : SUCCESS in 10m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/85455de268aa41c695a17391090a8ff9 : SUCCESS in 4m 21s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dbbf1031b9ae4f53a627e818e2de0e12 : FAILURE in 10m 09s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/62587d550d1845f3bc9e74ae81a1a143 : SUCCESS in 5m 29s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/83e2f2aad50e4c2ebc3f17c1295f87b4 : SUCCESS in 58m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1e73417744b94a7f8ce8fd5f0e85a332 : SUCCESS in 49m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/293a919f25424843bd966ce9502b6749 : SUCCESS in 57m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/403c91c266144a778065fd4fbc8d1403 : SUCCESS in 29m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7a8e670931d46bbaaf498d513151320 : SUCCESS in 3m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/76b0a97b4df84fbeb9aadbedb712b223 : SUCCESS in 21m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a33a1e6545e04726b67a27908d05ae67 : SUCCESS in 57m 54s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"7308c4c2c949213739b70afef4879ac57798c0cd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 02:08:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc112b5fdb0d4192b4c2e7b776d5cd53\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b71d2c4348054d8f87b2c0b4ac4a312a : SUCCESS in 1h 23m 51s","accounts_in_message":[],"_revision_number":2},{"id":"5dcd51a6172b5f79a5fd4bd1c937d0c93ea43331","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-08-20 18:53:09.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"b706e898e340404ea78b60aa9e8c7b1b1273d04b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-08-20 22:48:03.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"54eada75b38e5ddbadfc76c1f857fd6f80e16ada","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-08-20 22:49:04.000000000","message":"Patch Set 3:\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"bd78f32fa76f743ce6a99a0d7d17f533112feed0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 23:52:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fb440ba82b141f5815ef3a48ccd693c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a0e0900a982a4a7c8a645e055d1d1d2a : SUCCESS in 1h 02m 24s","accounts_in_message":[],"_revision_number":3},{"id":"5bdfc7c37101457234f7eba99bddf97be90f1e5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 23:53:17.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/972be576adbe465d8de06f52ca13132e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81c3b14302794c19bff59ff2eeffc178 : SUCCESS in 5m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3fe1ce2611ac4cf8b5fb9153b2d2aa0e : SUCCESS in 4m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/3e43a4a61e67452092147bc3210ce83e : POST_FAILURE in 39m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b0da5ebee6954d3ea3a8d2f5eb137e5d : SUCCESS in 30m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d2a47fd4e1da433180f1b2a2614bb610 : SUCCESS in 27m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a4cda096d37d421fa93e7838fe2ce6ed : SUCCESS in 34m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/53a33fee22924aa8b16ad397f48df5bd : SUCCESS in 12m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/83df5c85344245bfa48c43f19b8818ae : SUCCESS in 19m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b48e28cdca4243118288316d36708fd5 : SUCCESS in 11m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d31a4e60b57d4f2eaf6c8839c1f2610b : SUCCESS in 5m 38s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/180b1362c348483c85401995dba3154f : SUCCESS in 26m 09s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/c3fb8429b8624a078205418a365e6b8d : SUCCESS in 16m 51s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/ee740cb2cdd84ed7845a0f6d72470010 : SUCCESS in 30m 11s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/09c3a977fddb47228ceda508b5995546 : SUCCESS in 21m 03s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/bc556d19f7374694b07c2ae78bdb88ac : SUCCESS in 14m 46s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/829b479a38644f60bd6c504ccd076aac : SUCCESS in 15m 04s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/58e66e3c7b1e4cc2a7ca007d7e277653 : SUCCESS in 8m 31s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/ebbc037403c64e908dd0bab63668f7ce : SUCCESS in 9m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f57a95fcdae04b43a52ca805cca568f4 : SUCCESS in 8m 06s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c1fe02518ab8455c846ba93b91d7df60 : FAILURE in 6m 11s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/fdebb39c245d4a0db4f8085693beaf8d : SUCCESS in 5m 44s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0e6b6f1666c844c5b17db630735082f2 : SUCCESS in 1h 00m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2a11cb36643f45f69a93011208dd6f54 : SUCCESS in 23m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/24efbb14a43d4b7c9ca18810e96f672f : SUCCESS in 53m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/55350177c5694f31a8dcb6ccbf4f285d : SUCCESS in 25m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/780d6c0cf2c5442983d9d859c1e0940e : SUCCESS in 3m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6b517b944ecc43aa9a72c99ce2e6398b : SUCCESS in 19m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ff5ae660cad7431da2e8472468bfdb00 : POST_FAILURE in 46m 23s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":3},{"id":"6f89070d0e3ca819c50824103886c26401806f2d","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-08-21 00:10:59.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"db4df87a63dd20bc4e9e236d9e0e15664316af51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 01:21:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd935c7aa6af467babc8e0a1c72deda4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11e0e14d430d48c5a7c1e4c28f190ec1 : SUCCESS in 7m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f5d46607ee146e9b8a8afde5cd48522 : SUCCESS in 10m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/8c6f127e506e4b5ab942a08863e52962 : SUCCESS in 44m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3fc6bb935aad4a22a0414eb18a15dafa : SUCCESS in 1h 04m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/58efb3d4cce94dd2bad8f5dec6474f1d : SUCCESS in 48m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e449c5c99f3741f8b9e7e1d4694fe22b : SUCCESS in 57m 33s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e0cdc9549a154fafb68ab0e4ad280ad2 : SUCCESS in 20m 01s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7c2e2dc54b134494bc14f694e4fdf58f : SUCCESS in 10m 20s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2b55ed5d397b4ffb8855d136e836ec16 : SUCCESS in 17m 19s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/838ede50be8d443997a51f2618beab78 : SUCCESS in 3m 57s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e4dbd3c5fb6f4d9582c2caf48170f4a8 : SUCCESS in 12m 32s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/2bddb097c4364c8298d3e1b5c39bbf8b : SUCCESS in 9m 01s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/8f6daf966d9d4520893125b4b3ebee85 : SUCCESS in 14m 09s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/2cdeeefe85e2475bb5d402eb2ac11267 : SUCCESS in 20m 37s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4765169e95da43f3af355946662c4262 : SUCCESS in 14m 12s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0a53d233d3d9443c98111af901f3deee : SUCCESS in 9m 13s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8a13315933db4aee93f92f283acc2cd5 : SUCCESS in 15m 17s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/b7df9ed742b84785b1a0ec8eeca107a5 : SUCCESS in 9m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a3064d86a784408cb2e0e80f8f8187f2 : SUCCESS in 5m 41s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2ceacfc71fc3416a913b537556bc1553 : FAILURE in 10m 08s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/1dc8d5804f554da4adb950776ea34d7e : SUCCESS in 7m 47s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/80da68262b4645d8a7b5bc846a31148a : SUCCESS in 1h 02m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/df69628d4ddb4e93985cabef36c055a1 : SUCCESS in 31m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/72a7f218606246a692ccf758da6d70f4 : SUCCESS in 32m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83f3fbaa122a4003abc7b5310dbaa462 : SUCCESS in 20m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dba04de0236043d5b04713d3e375990d : SUCCESS in 2m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/13d1329824ec4768b5bf8ff165dc5bb0 : SUCCESS in 24m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/19e483a2581548b1a3a4984a1aeb76fb : SUCCESS in 36m 26s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":3},{"id":"87d740cc53c0d1066b521dbbcdc9b0307df3ceac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 01:24:52.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77f76dafca794823a62beca5f6d641f2\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1200718444f14183802b39cdcef0916f : SUCCESS in 1h 11m 39s","accounts_in_message":[],"_revision_number":3},{"id":"03d18307b278d47d6001773da4ee48724ef415fc","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-08-21 07:24:37.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e2b46b6dea5bc742b3625116225f60498cac07b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-08-21 22:59:03.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"69f0232d8770d60ac0a7918f882fb55f8fe90e82","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-08-21 23:05:04.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"76878e6333c52e367e5fba83d5bd56ebdb666d1a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-22 00:11:35.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f0ee8293a784466ae3a624325676f9a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/54f0792e27cb4414a982e33615c23567 : SUCCESS in 1h 09m 57s","accounts_in_message":[],"_revision_number":4},{"id":"5811b2a54db9860a84f5bc71f4019e3efb575598","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-22 00:16:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/681d5cbbc51749d0af6b79fa0beba40d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c07cf15752d46c0994053102a8fc816 : SUCCESS in 7m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca0d6c7902964745854c37cf67a2a59f : SUCCESS in 4m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/ae6c880dff124942b348c46205ffc814 : SUCCESS in 29m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fd75767e9e5148db8d38f5126960afd6 : SUCCESS in 33m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f4ebe1b0eed145dbaffc5fb0dc79e786 : SUCCESS in 33m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5951a713a3094c0296600c4a34ff1daf : SUCCESS in 31m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/43333dfdeebd4e6d840d1f591ad86b19 : SUCCESS in 10m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6e8b3f5b9e974210adea7e8834069fe0 : SUCCESS in 19m 56s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/55f83835f15f4d6eb26811d934905b9b : SUCCESS in 12m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/537179520b3441f3adc9b9dbd40f93a4 : SUCCESS in 5m 52s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7bc2120fd91a49efa16cfb26f6af85b1 : SUCCESS in 12m 10s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/4e7a8547d31c4d1c87ac19d6132e6970 : SUCCESS in 14m 28s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/b4e71cbc720546fe933916246e79c632 : SUCCESS in 13m 41s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/120cdd9df3094b9e8719aa67bf58d018 : SUCCESS in 9m 19s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/72ce592d2aea4d33ac0107e10bc26dfb : SUCCESS in 8m 05s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3ecfdfef3e8345ebac959cc46b30afb4 : SUCCESS in 12m 55s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/21277e6925df4b5f8d03c63689296cd6 : SUCCESS in 15m 41s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/ff2fe909a10a45a6a24129ac94369ffb : SUCCESS in 14m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b43e5237d07a4aaab26a9c63d33e2430 : SUCCESS in 8m 30s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dd10ba7d8f36467e91577e2cbba121bc : FAILURE in 4m 03s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/5e12a23a495544e99f9f1decad7d3162 : SUCCESS in 9m 21s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4eb600259b1e4aabba2848f03e958977 : SUCCESS in 35m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7fdddc598bc84472b15acc08eb729fd1 : SUCCESS in 59m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/26dad803944d4a1491169be0c607184e : SUCCESS in 25m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fb9896905a45400b87cbc625a62f90c7 : SUCCESS in 24m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b411d15ae0e4e39b57f749a94370f07 : SUCCESS in 3m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/934a099cb6bb456ba9afb086b50a92b4 : SUCCESS in 26m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ab0a218e19c84cb69d075b41f3b4223b : SUCCESS in 1h 09m 16s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":4},{"id":"7cc209e7278c18afa8a90f9f721655c269e1b505","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-09-02 18:07:29.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ee576fd357731a172d9246ad2888e3a7bcd7b04a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-09-02 20:33:39.000000000","message":"Abandoned\n\nsq\u0027ed into [966980](https://review.opendev.org/c/openstack/swift/+/966980).","accounts_in_message":[],"_revision_number":4},{"id":"a58577bab60631e4388ee242b402aaed0eb06953","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-09-10 16:20:03.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"34b1f469d6925c6ba5099cee3aadb1b00181b392","revisions":{"c8d24079054f12f9d4c0cc2a87a774272b9b9245":{"kind":"REWORK","_number":1,"created":"2026-08-19 00:09:01.000000000","uploader":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"ref":"refs/changes/46/1001446/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/46/1001446/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/1001446/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/1001446/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/1001446/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/46/1001446/1"}}},"commit":{"parents":[{"commit":"f6cbd18402755ff8c87966cf6011fe26f1df19e3","subject":"relinker: Add audit step to quarantine stale hash directories","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f6cbd18402755ff8c87966cf6011fe26f1df19e3"}]}],"author":{"name":"Wael Halbawi","email":"whalbawi@nvidia.com","date":"2026-08-19 00:05:03.000000000","tz":-420},"committer":{"name":"Wael Halbawi","email":"whalbawi@nvidia.com","date":"2026-08-19 00:08:50.000000000","tz":-420},"subject":"relinker: Audit runs default to EC storage policy 1","message":"relinker: Audit runs default to EC storage policy 1\n\nAfter a part-power increase, uncleaned hash directories on an object\nserver likely manifest as object reconstructor errors, as their sibling\nfragments are already cleaned up.\n\nDefault an unrestricted audit run to storage policy 1 when it is an EC\npolicy. Select no policy if policy 1 is absent or is not EC. The\n`--policy` flag can be used to override this choice to a user-specified\nset of policies.\n\nThe relinker and cleanup actions continue to retain their all-policy\ndefault behavior.\n\nChange-Id: I9c44aad3e316d520cb8721e206221fd262fb77ef\nSigned-off-by: Wael Halbawi \u003cwhalbawi@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c8d24079054f12f9d4c0cc2a87a774272b9b9245"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c8d24079054f12f9d4c0cc2a87a774272b9b9245"}]},"branch":"refs/heads/master"},"0080342facf3f6193556ec1e23287596b5a23ad4":{"kind":"REWORK","_number":2,"created":"2026-08-20 00:37:27.000000000","uploader":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"ref":"refs/changes/46/1001446/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/46/1001446/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/1001446/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/1001446/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/1001446/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/46/1001446/2"}}},"commit":{"parents":[{"commit":"2a9d1644047c6ed06ef59e7bd689ecfc2f774209","subject":"relinker: Add audit step to quarantine stale hash directories","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2a9d1644047c6ed06ef59e7bd689ecfc2f774209"}]}],"author":{"name":"Wael Halbawi","email":"whalbawi@nvidia.com","date":"2026-08-20 00:37:18.000000000","tz":-420},"committer":{"name":"Wael Halbawi","email":"whalbawi@nvidia.com","date":"2026-08-20 00:37:18.000000000","tz":-420},"subject":"relinker: Audit runs default to EC storage policy 1","message":"relinker: Audit runs default to EC storage policy 1\n\nAfter a part-power increase, uncleaned hash directories on an object\nserver likely manifest as object reconstructor errors, as their sibling\nfragments are already cleaned up.\n\nDefault an unrestricted audit run to storage policy 1 when it is an EC\npolicy. Select no policy if policy 1 is absent or is not EC. The\n`--policy` flag can be used to override this choice to a user-specified\nset of policies.\n\nThe relinker and cleanup actions continue to retain their all-policy\ndefault behavior.\n\nChange-Id: I9c44aad3e316d520cb8721e206221fd262fb77ef\nSigned-off-by: Wael Halbawi \u003cwhalbawi@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0080342facf3f6193556ec1e23287596b5a23ad4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0080342facf3f6193556ec1e23287596b5a23ad4"}]},"branch":"refs/heads/master"},"788605eb9fc9f01d7e91ce8a051b5200a042a2bc":{"kind":"REWORK","_number":3,"created":"2026-08-20 22:48:03.000000000","uploader":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"ref":"refs/changes/46/1001446/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/46/1001446/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/1001446/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/1001446/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/1001446/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/46/1001446/3"}}},"commit":{"parents":[{"commit":"2a9d1644047c6ed06ef59e7bd689ecfc2f774209","subject":"relinker: Add audit step to quarantine stale hash directories","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2a9d1644047c6ed06ef59e7bd689ecfc2f774209"}]}],"author":{"name":"Wael Halbawi","email":"whalbawi@nvidia.com","date":"2026-08-20 22:36:11.000000000","tz":-420},"committer":{"name":"Wael Halbawi","email":"whalbawi@nvidia.com","date":"2026-08-20 22:37:27.000000000","tz":-420},"subject":"relinker: Require specifying a policy for audit","message":"relinker: Require specifying a policy for audit\n\nRunning an audit against a replicated storage policy may have unintended\nconsequences: object-replicator may process a mis-homed hash directory\nas a handoff before audit gets to quarantining it.\n\nAdd a guard rail by refusing an unrestricted audit run and require\nselecting a storage policy using the `--policy` flag.\n\nThe relinker and cleanup actions continue to retain their all-policy\ndefault behavior.\n\nChange-Id: I9c44aad3e316d520cb8721e206221fd262fb77ef\nSigned-off-by: Wael Halbawi \u003cwhalbawi@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/788605eb9fc9f01d7e91ce8a051b5200a042a2bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/788605eb9fc9f01d7e91ce8a051b5200a042a2bc"}]},"branch":"refs/heads/master"},"34b1f469d6925c6ba5099cee3aadb1b00181b392":{"kind":"REWORK","_number":4,"created":"2026-08-21 22:59:03.000000000","uploader":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"ref":"refs/changes/46/1001446/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/46/1001446/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/1001446/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/1001446/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/1001446/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/46/1001446/4"}}},"commit":{"parents":[{"commit":"2a9d1644047c6ed06ef59e7bd689ecfc2f774209","subject":"relinker: Add audit step to quarantine stale hash directories","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2a9d1644047c6ed06ef59e7bd689ecfc2f774209"}]}],"author":{"name":"Wael Halbawi","email":"whalbawi@nvidia.com","date":"2026-08-21 22:58:57.000000000","tz":-420},"committer":{"name":"Wael Halbawi","email":"whalbawi@nvidia.com","date":"2026-08-21 22:58:57.000000000","tz":-420},"subject":"relinker: Require specifying a policy for audit","message":"relinker: Require specifying a policy for audit\n\nRunning an audit against a replicated storage policy may have unintended\nconsequences: object-replicator may process a mis-homed hash directory\nas a handoff before audit gets to quarantining it.\n\nAdd a guard rail by refusing an unrestricted audit run and require\nselecting a storage policy using the `--policy` flag.\n\nThe relinker and cleanup actions continue to retain their all-policy\ndefault behavior.\n\nChange-Id: I9c44aad3e316d520cb8721e206221fd262fb77ef\nSigned-off-by: Wael Halbawi \u003cwhalbawi@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/34b1f469d6925c6ba5099cee3aadb1b00181b392"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/34b1f469d6925c6ba5099cee3aadb1b00181b392"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
