)]}'
{"id":"openstack%2Fswift~1001512","triplet_id":"openstack%2Fswift~master~I466b0cecf02935e99eebba9de9b5cf3e39465c87","project":"openstack/swift","branch":"master","topic":"p-s3api-mpu-copy-if-match","attention_set":{},"removed_from_attention_set":{"38368":{"account":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"last_update":"2026-08-20 11:25:21.000000000","reason":"Change was submitted"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-08-20 11:25:21.000000000","reason":"Change was submitted"},"6968":{"account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"last_update":"2026-08-20 09:31:28.000000000","reason":"\u003cGERRIT_ACCOUNT_6968\u003e replied on the change","reason_account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}}},"hashtags":[],"change_id":"I466b0cecf02935e99eebba9de9b5cf3e39465c87","subject":"s3api: restrict headers sent with copy check HEAD request","status":"MERGED","created":"2026-08-19 14:19:48.000000000","updated":"2026-08-20 11:29:36.000000000","submitted":"2026-08-20 11:25:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"1001512-p-s3api-mpu-copy-if-match","meta_rev_id":"d7481318d332409f562c16a92f9db247341bf829","_number":1001512,"virtual_id_number":1001512,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},{"value":0,"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-20 11:25:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"all":[{"value":0,"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},{"value":2,"date":"2026-08-20 09:31:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"all":[{"value":0,"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},{"value":1,"date":"2026-08-20 09:31:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-19 16:00:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-19 21:37:19.000000000","updated_by":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"reviewer":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"state":"REVIEWER"},{"updated":"2026-08-20 08:38:41.000000000","updated_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"reviewer":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"state":"REVIEWER"}],"messages":[{"id":"56f5034ed510bd0bd36eb7e37a7300753df58256","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-08-19 14:19:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a6810a8154280cbd496edc7cb32e09b556c27287","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-08-19 14:46:26.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"df055ecae13517a66fa17674418c23df51039e3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 16:00:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e45dd4cdbb0b41ed8e7129766ec18578\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc7bf1e838a344a38f48b87d748afb77 : SUCCESS in 5m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3172317a906c4f75be2827a206d79192 : SUCCESS in 8m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/b1f1493574d14faa89e78c83f740149d : SUCCESS in 1h 06m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fcc794177aaa4595a48a0c94f5275215 : SUCCESS in 46m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/71545a444360441d99758ba7b0aef73f : SUCCESS in 1h 03m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8f317d4eb2834b759153c72c73af5209 : SUCCESS in 28m 53s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2e044d4eade04077b53d182ecdd91bbc : SUCCESS in 21m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/65db3c784e114d5dbc78082d4df6ede0 : SUCCESS in 14m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/612a59cf692e482b9572629f85386a22 : SUCCESS in 15m 28s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d61a25ab027a4dfba22c22e76a05908d : SUCCESS in 6m 20s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d4f65281cc544770aaa76bd4e19cff45 : SUCCESS in 20m 51s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/740185615cfa41a1b34c199f39ad0519 : SUCCESS in 25m 16s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/71d2037302064e9991d60754e09fe0b6 : SUCCESS in 33m 17s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/2a05d8ec50684e2ca92f31ad1894410e : SUCCESS in 18m 33s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/1d2803b7a3d2437c9dccc6bc8a39b73a : SUCCESS in 11m 05s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/715ef8b8b85c41a7b76ff712b74b6e73 : SUCCESS in 14m 36s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/51e2ac02c13d43938db496e3639ac246 : SUCCESS in 11m 43s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/94906bcf2ad946818c4b1154965860b7 : SUCCESS in 17m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/dbb9a2cb53204d8d9becf34d0a7b16a8 : SUCCESS in 5m 56s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/78237f0a3ed7495ab6970187f296303a : FAILURE in 8m 22s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/d0451161dcc94a9692c4a6781f72622f : SUCCESS in 8m 51s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/20a5b4241629451e83009b8583e13ee4 : SUCCESS in 1h 00m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c8e21ab81f2344f09b25e24b423521e9 : SUCCESS in 24m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/58f5ad94f69b42699b35950a3bed72ec : SUCCESS in 57m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ae8d07d5b89f49749c8d54167d8cc1a7 : SUCCESS in 25m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/544e7b003e64426ca1e7438cf041cf43 : SUCCESS in 2m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a9afbf7f0b87459487edcf301f43945a : SUCCESS in 19m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/08f179a7a5d84f43af0d6de076183775 : SUCCESS in 26m 32s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"b8f0ca2414ce115c39074e6a741ed5bb99b023de","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 16:03:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b71e5dd6c2e64496af2221cacb854d66\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/62252ce23b1c42c59e4f521b669e36e5 : SUCCESS in 59m 03s","accounts_in_message":[],"_revision_number":2},{"id":"84cf6c8f30120b69ef22958b3559579c422e94b1","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2026-08-19 21:37:19.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"eee2585b2ddf516a2f60f2c3ff24da205301880d","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-08-20 08:38:41.000000000","message":"Patch Set 2: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"d5cff68cb63daf2efe9f3d63181bc24e7e1d23fc","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-08-20 09:16:15.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"1d11bd388f8702d59b54aba6e72bd39a01628b32","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-08-20 09:24:55.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"17bf4b85ba73d12a00f2afa63cd2fc9719c08ac9","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-08-20 09:31:28.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1e1464d74034cacb585634ae2c5e9909fa44f07b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 10:23:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c431ab396eb433aa5430918093f1b6d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/59531aa791b541c0b324f6d1793df9ef : SUCCESS in 55m 19s","accounts_in_message":[],"_revision_number":3},{"id":"01eb45a5046e46945377531fc0eb6b200c86023e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 10:38:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/460f360984b14a4fbf5f1e531b06992c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b13f775df9743b9a929e0debc9c4cd6 : SUCCESS in 5m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c2887331adbe417ebc5f36409621a2df : SUCCESS in 4m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/8d66f355ec6a4c8184281e21f7e0989d : SUCCESS in 59m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/575d020748f24971b18aa4fc238d0e27 : SUCCESS in 32m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/750fe59e489045b690352cbef1dae34d : SUCCESS in 51m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3ad40d40030e4ef2b2ffe873653f9d6d : SUCCESS in 28m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f775bafc149f4df9bbf0ea6caaac031a : SUCCESS in 19m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a63ef056331b4be38bb97c1a2565fdd6 : SUCCESS in 11m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/30653dcea5a74ccd8d1872b094f8fd2c : SUCCESS in 22m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f88c14012fb84a30900fe526d7cdb624 : SUCCESS in 6m 53s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/cf71eb336ad84be988e8aaf122654b83 : SUCCESS in 11m 41s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/66c37b8a3dd44e0aad361779d898afd9 : SUCCESS in 22m 52s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/3cbc658aec9f468699edc79c8a29095c : SUCCESS in 32m 51s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/7a519aa18bf64a9cab0eeabe0c1aed95 : SUCCESS in 19m 30s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d9e5109f35a14114b33808a1795d3291 : SUCCESS in 8m 50s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d6ce0e2130544dacbab0dc1cdf9b4537 : SUCCESS in 8m 18s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/24d83b7e919f4b9d87d6661195897ad1 : SUCCESS in 9m 03s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/523b0ff51b3d49ff8aa28e6d2d12326f : SUCCESS in 17m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/aef46a8c68bb4937bc514c9d91ed3336 : SUCCESS in 6m 10s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/073e2b5fe23f4e94b2c365778ff07500 : FAILURE in 8m 39s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/419a14d44a174f478f12c3065ec4a63c : SUCCESS in 5m 39s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/97aa44c5750d49a7a92407da30d3d164 : SUCCESS in 35m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f92e8dfd204a4924b45ebc7c82b61e08 : SUCCESS in 56m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/53a602a415374698bf1c9652293ef91e : SUCCESS in 53m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/18c412ac55954fbcb3d4427b2780a274 : SUCCESS in 10m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddc89fd94eab44c0a03b33577dabba52 : SUCCESS in 5m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c4e82cc4d6f5478aa4fae77181edf481 : SUCCESS in 29m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9dcccef41ec944a38f859303cca9f7ac : SUCCESS in 1h 06m 52s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":3},{"id":"eb4ccf84c98cb3271fe8082e72694cc9b7fb816f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 10:38:39.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"00dd30c071ab2340df93715c917a9b553e900976","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 11:25:21.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5e6ec938d9545549868cff145de6992\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01728217c0234206845a2adf708b6226 : SUCCESS in 6m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce5a6ff90fa54815a316d93e6b44e4bc : SUCCESS in 6m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/6c31f277cebc44be8593b1c617863edb : SUCCESS in 34m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/aa417e5157e74c3cb83fc5a6c0a8e82e : SUCCESS in 32m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0e22b7019c8a45dd812bddad680a60c9 : SUCCESS in 32m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e2c605b342814a18a9dba163241c73cd : SUCCESS in 29m 33s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/7aac8d774d7c4d3cacbcb4248476b7ef : SUCCESS in 5m 06s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/fcf957ee343c47c396943ebe9a5dabde : SUCCESS in 11m 21s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/66f70e9f040548d4b8bab50c0e6e0944 : SUCCESS in 10m 27s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/9a7d58466fcb4daba65ceec42ac8590a : SUCCESS in 23m 08s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/c85b15b4b3264b3ea2967fa5cfcba39a : SUCCESS in 10m 39s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/64837074cf394527bcf5c82ac61449d7 : SUCCESS in 9m 34s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e6d882effaed4343a85f01e014c60d92 : SUCCESS in 9m 51s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7fbee6b155c2474cb18eb1f4931fe0f0 : SUCCESS in 9m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6c17e8fb2f284a9986c16eb4f3075ba5 : SUCCESS in 4m 08s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/dd5dc45b571746ce9d02f3f367eb9cb5 : SUCCESS in 6m 15s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/530d9de921df4eb782738efec3e923d7 : SUCCESS in 39m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d9e53ebfe90f4f1e8ef51836d356a2db : SUCCESS in 26m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cd98acbce246495697c2fb6f26780023 : SUCCESS in 24m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7b04cdd29d904d7da80f9a946bd6e1b6 : SUCCESS in 11m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/704308f73bc248e9b187756d1209836f : SUCCESS in 3m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/57da3f57b38d4067b5364ab82d1c7dd2 : SUCCESS in 11m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f8f4bbb61ce14450b91eff041353b15b : SUCCESS in 28m 53s","accounts_in_message":[],"_revision_number":3},{"id":"cd91f21260afa066410056c9a32a343aee97750b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 11:25:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"d7481318d332409f562c16a92f9db247341bf829","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 11:29:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b79063595afe4710aff7a2d5201af0fb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ba6076de07045d293411437cb14bc75 : SUCCESS in 55s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/d36ab9f0972f43afa1429ec223a63ea9 : SUCCESS in 59s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2d873f6c01717d8d60017f91d4e54ea5d3b57e05","revisions":{"41a250f42e951e2a6291011081f0224d1473a004":{"kind":"REWORK","_number":1,"created":"2026-08-19 14:19:48.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/12/1001512/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/1001512/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/1001512/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/1001512/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/1001512/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/1001512/1"}}},"commit":{"parents":[{"commit":"936199aafba7efc19063df9d492557283a7689cc","subject":"s3api: fix mpu copy error with x-amz-copy-source-if-match","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/936199aafba7efc19063df9d492557283a7689cc"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-08-19 14:07:41.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-08-19 14:15:19.000000000","tz":60},"subject":"s3api: restrict headers sent with copy check HEAD request","message":"s3api: restrict headers sent with copy check HEAD request\n\nS3Api makes a backend HEAD request to check the source of a CopyObject\nor UploadPartCopy request before making the copy request. Previously\ns3api translated all client request headers named\nx-amz-copy-source-\u003csuffix\u003e to a header named \u003csuffix\u003e in the backend\nHEAD request. This allowed unrecognised headers to be propagated to\nthe backend. This patch restricts the set of translated headers to\nonly those expected.\n\nChange-Id: I466b0cecf02935e99eebba9de9b5cf3e39465c87\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\nAssisted-By: Claude Opus 5 \u003cnoreply@anthropic.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/41a250f42e951e2a6291011081f0224d1473a004"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/41a250f42e951e2a6291011081f0224d1473a004"}]},"branch":"refs/heads/master"},"3aab4448a83acb12b3b82be39a107dcc37125ea5":{"kind":"NO_CHANGE","_number":2,"created":"2026-08-19 14:46:26.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/12/1001512/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/1001512/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/1001512/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/1001512/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/1001512/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/1001512/2"}}},"commit":{"parents":[{"commit":"e32bc963df516d6465912e5d2d6924925549c1da","subject":"s3api: fix mpu copy error with x-amz-copy-source-if-match","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e32bc963df516d6465912e5d2d6924925549c1da"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-08-19 14:07:41.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-08-19 14:46:26.000000000","tz":0},"subject":"s3api: restrict headers sent with copy check HEAD request","message":"s3api: restrict headers sent with copy check HEAD request\n\nS3Api makes a backend HEAD request to check the source of a CopyObject\nor UploadPartCopy request before making the copy request. Previously\ns3api translated all client request headers named\nx-amz-copy-source-\u003csuffix\u003e to a header named \u003csuffix\u003e in the backend\nHEAD request. This allowed unrecognised headers to be propagated to\nthe backend. This patch restricts the set of translated headers to\nonly those expected.\n\nChange-Id: I466b0cecf02935e99eebba9de9b5cf3e39465c87\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\nAssisted-By: Claude Opus 5 \u003cnoreply@anthropic.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3aab4448a83acb12b3b82be39a107dcc37125ea5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3aab4448a83acb12b3b82be39a107dcc37125ea5"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"936199aafba7efc19063df9d492557283a7689cc","ours":"41a250f42e951e2a6291011081f0224d1473a004","theirs":"e32bc963df516d6465912e5d2d6924925549c1da","merge_strategy":"recursive","contains_conflicts":false}},"2d873f6c01717d8d60017f91d4e54ea5d3b57e05":{"kind":"REWORK","_number":3,"created":"2026-08-20 09:24:55.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/12/1001512/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/1001512/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/1001512/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/1001512/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/1001512/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/1001512/3"}}},"commit":{"parents":[{"commit":"e32bc963df516d6465912e5d2d6924925549c1da","subject":"s3api: fix mpu copy error with x-amz-copy-source-if-match","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e32bc963df516d6465912e5d2d6924925549c1da"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-08-19 14:07:41.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-08-20 09:19:34.000000000","tz":60},"subject":"s3api: restrict headers sent with copy check HEAD request","message":"s3api: restrict headers sent with copy check HEAD request\n\nS3Api makes a backend HEAD request to check the source of a CopyObject\nor UploadPartCopy request before making the copy request. Previously\ns3api translated all client request headers named\nx-amz-copy-source-\u003csuffix\u003e to a header named \u003csuffix\u003e in the backend\nHEAD request. This allowed unrecognised headers to be propagated to\nthe backend. This patch restricts the set of translated headers to\nonly those expected.\n\nChange-Id: I466b0cecf02935e99eebba9de9b5cf3e39465c87\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\nAssisted-By: Claude Opus 5 \u003cnoreply@anthropic.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2d873f6c01717d8d60017f91d4e54ea5d3b57e05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2d873f6c01717d8d60017f91d4e54ea5d3b57e05"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
