)]}'
{"id":"openstack%2Fswift~1001845","triplet_id":"openstack%2Fswift~master~Ida05387c10588e8921ba6d896a7e8451e23833f8","project":"openstack/swift","branch":"master","topic":"s3-checksum-persistence","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-08-25 18:12:15.000000000","reason":"Change was abandoned"},"39146":{"account":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"last_update":"2026-08-25 18:12:15.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ida05387c10588e8921ba6d896a7e8451e23833f8","subject":"sq? object controller can install checksum resp headers","status":"ABANDONED","created":"2026-08-21 16:01:52.000000000","updated":"2026-08-25 18:12:15.000000000","total_comment_count":7,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"68090cfb32b6816d5bdbf05feba211e7cdd91361","_number":1001845,"virtual_id_number":1001845,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-08-21 21:26:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-21 17:18:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-21 17:24:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-21 19:41:51.000000000","updated_by":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"reviewer":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"state":"REVIEWER"}],"messages":[{"id":"b0b166b29307030dd78ec7153cfb720bf4b126dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-08-21 16:01:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"85c9036eb51fe3908a1083d30a583b6121b4e32f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 17:18:22.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ecdebbb3204c48c98eca0231ce690de7\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b66cd3d1c34a4742a1790ea196725955 : FAILURE in 1h 14m 20s","accounts_in_message":[],"_revision_number":1},{"id":"1b789a5fcab69137a1bb835715aae9fb350d1d84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 17:24:22.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6eb8b8a1b9454d1abb295070c8ec1ec0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ea867bbb8c443a8a34094b2991ef302 : SUCCESS in 9m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7307376fcf7465c92a603a48e8ab0e8 : SUCCESS in 10m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/fd2f27c3e5ba480b95523ff9ca33976b : SUCCESS in 1h 03m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4958ec47ebec4d0d8533ced8d34cbc44 : SUCCESS in 1h 06m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c3028c52f0ca4135be777347503a2f73 : SUCCESS in 35m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/400e6baa8a7b4f4e95af479d6f8b14fa : SUCCESS in 34m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f52852063e3e47208452805c53a23324 : SUCCESS in 13m 35s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2d050c74f7ef4cd78a9a985e15b0263a : SUCCESS in 20m 33s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4fbe1cadce0247da9e3b4261fff47feb : SUCCESS in 21m 10s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f1e0d6a743e745c59339d779850b33f0 : SUCCESS in 7m 10s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/606805d676554bfaa9a551f72f81a06d : FAILURE in 24m 47s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/457b581c7a16495d9189c062406bfab8 : FAILURE in 23m 24s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/f09c598140684c1892e95c921f943814 : FAILURE in 15m 59s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/4944fdf02a2c4435b98773ed3a0bdd79 : FAILURE in 16m 22s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/2fbf5b25020d4fa3bb1329f26663017d : SUCCESS in 16m 27s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1698474a543940308ae2f361ebde0725 : SUCCESS in 10m 21s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e224e8c2d10246d0a3e5123dd7c4d205 : SUCCESS in 14m 20s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/104d2bf6c4914539be95087ef42bc064 : SUCCESS in 18m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/913a5eb32bc14e0787ff109bbf55f831 : SUCCESS in 6m 40s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/77dc0ed4e43e4c9d812143ade2b2ee1d : FAILURE in 4m 31s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/1e3dcfba45904495bce122f04e12a955 : SUCCESS in 8m 40s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2311279c94624a339aa868781c9ccaf8 : SUCCESS in 57m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f1d43443cf974a0ca8c92a10f39508b5 : FAILURE in 16m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6ba5b6619a764dcc867a39fd7cde0418 : SUCCESS in 55m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/81ac318261424ada960650cbf04c20b4 : FAILURE in 33m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b27690d914840f48d041b4c5d51b253 : FAILURE in 3m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cd3f568fc4904957bc6ec2a504a297fe : SUCCESS in 18m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/baaaad78c39f4c05a12d7c901cde3aba : SUCCESS in 40m 42s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"669c8bd9e9a4d38b3a70800beafa5b1d7d25ca30","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-21 19:41:51.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"92c9099953114141dfc44453351e40344ea57d17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-21 20:13:31.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"cc5f1a98b099d6adc9ba32c887d50d2b3ebb5959","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-21 20:19:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6e4ea49d7e83523f6693e5419601e5ad7aa3e02f","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-21 21:10:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"26dc429d9722c1e391d6d194b77111703b93d310","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-21 21:11:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8f4418e77f9ca18a065e7a9cb6c09edb32e209cb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 21:13:56.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5cea489c5b1742588c80c6b6ef49982c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/26bed75627a74b7b874ca530146fd27c : FAILURE in 57m 18s","accounts_in_message":[],"_revision_number":2},{"id":"90e2c37904f29047aa0fa7d6a54c1f2288d1fc5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 21:26:24.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8d7d2d91eb2140198d982de765ec3828\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d473328960645c8845afc7a432e6588 : SUCCESS in 9m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/19ff2b065abd45309fbdf63378001fd4 : SUCCESS in 8m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/ddb11b6c1b1a4929b24546d5a1c11d6e : SUCCESS in 1h 04m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dfa7c7f538394a3eb86f33c5a8e1078b : SUCCESS in 34m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a07a8ec1189c4d1d9c8eeb27df9d110f : SUCCESS in 31m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e9aba5f9cf89418db9880109972a2063 : SUCCESS in 59m 26s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e53c8e171809498bbd15dec6f3e36b38 : SUCCESS in 19m 17s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a071196bf06d428d95885dbef678543b : SUCCESS in 20m 42s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a1a4a1fbfd5945df99400de851420a16 : SUCCESS in 17m 47s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/222ebd1009d54751a9667eb216fb535c : SUCCESS in 5m 49s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9033b4684aab4ac5ad269dffd2540299 : FAILURE in 26m 33s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b653a4f4779f4e148b033f1ef2490d1a : FAILURE in 9m 21s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/6faf9c296a574e1dae6487fcec1c1c85 : FAILURE in 33m 23s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/8f6a19c4229748bea255e1b2bf33f265 : FAILURE in 18m 52s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c314403d6d2943129100a2653d54c075 : SUCCESS in 8m 55s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/18bc6002cd31496c9de58c90976e02d5 : SUCCESS in 9m 10s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/83682554ea734a53ab24a1f06c2f643b : SUCCESS in 8m 37s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/be8bbe8a4bf3470f968be79640496dc2 : SUCCESS in 17m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b875d69e5924448d8dd02e3acdd2cf73 : SUCCESS in 3m 38s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/85a3b886e4f74908bff494ec51e63b76 : FAILURE in 7m 57s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/f1d48d06d7fd4875b11ca018603b4b7c : SUCCESS in 7m 32s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ef0bd3b0853a4e90b648f8b0e0b6270d : SUCCESS in 1h 00m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cc613e5623b54b949ca33a55f24ee45e : SUCCESS in 26m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/de17cec2e3964262b1d53827aceba52b : SUCCESS in 25m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/46b944d660e441e7863080cb421118eb : FAILURE in 27m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d3ccc575dbd48b19ede8b9c78e73005 : FAILURE in 4m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4551621a09ff4ab3a1dc7e4760238df8 : SUCCESS in 18m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c3ee0240ecc447138f57c2d838f16329 : SUCCESS in 39m 59s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"68090cfb32b6816d5bdbf05feba211e7cdd91361","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-08-25 18:12:15.000000000","message":"Abandoned\n\nsquashed 991516: s3api: Checksum Persistence for Full Object Upload PUTs | https://review.opendev.org/c/openstack/swift/+/991516","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"196a7b42bb1e9ee80bf7b62434437e9bb63f6ef6","revisions":{"6d8820223168d4a0d0bddd5107ed78fcd041fed7":{"kind":"REWORK","_number":1,"created":"2026-08-21 16:01:52.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/45/1001845/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/45/1001845/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/45/1001845/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/45/1001845/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/45/1001845/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/45/1001845/1"}}},"commit":{"parents":[{"commit":"0571359e2be37d8f78a4c06b33555b06384174b2","subject":"s3api: Checksum Persistence for Full Object Upload PUTs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0571359e2be37d8f78a4c06b33555b06384174b2"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-08-21 15:46:00.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-08-21 16:00:31.000000000","tz":60},"subject":"sq? object controller can install checksum resp headers","message":"sq? object controller can install checksum resp headers\n\nBetter separate concerns: S3Request stores sent request s3 sysmeta,\nS3Response stores received s3 sysmeta, ObjectController uses one or\nthe other to install client response checksum headers.\n\nSome tests will need updating.\n\nChange-Id: Ida05387c10588e8921ba6d896a7e8451e23833f8\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6d8820223168d4a0d0bddd5107ed78fcd041fed7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6d8820223168d4a0d0bddd5107ed78fcd041fed7"}]},"branch":"refs/heads/master"},"196a7b42bb1e9ee80bf7b62434437e9bb63f6ef6":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-08-21 20:13:31.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/45/1001845/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/45/1001845/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/45/1001845/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/45/1001845/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/45/1001845/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/45/1001845/2"}}},"commit":{"parents":[{"commit":"0571359e2be37d8f78a4c06b33555b06384174b2","subject":"s3api: Checksum Persistence for Full Object Upload PUTs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0571359e2be37d8f78a4c06b33555b06384174b2"}]}],"author":{"name":"Nate Martes","email":"nmartes@NVIDIA.com","date":"2026-08-21 20:12:54.000000000","tz":0},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-08-21 20:13:29.000000000","tz":0},"subject":"sq? object controller can install checksum resp headers","message":"sq? object controller can install checksum resp headers\n\nBetter separate concerns: S3Request stores sent request s3 sysmeta,\nS3Response stores received s3 sysmeta, ObjectController uses one or\nthe other to install client response checksum headers.\n\nSome tests will need updating.\n\nChange-Id: Ida05387c10588e8921ba6d896a7e8451e23833f8\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/196a7b42bb1e9ee80bf7b62434437e9bb63f6ef6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/196a7b42bb1e9ee80bf7b62434437e9bb63f6ef6"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
