)]}'
{"id":"openstack%2Fswift~1001893","triplet_id":"openstack%2Fswift~master~Iadc8ac8b9791b23f2186b8b0de370631b0ddaac4","project":"openstack/swift","branch":"master","attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2026-08-31 21:21:09.000000000","reason":"\u003cGERRIT_ACCOUNT_34930\u003e replied on the change","reason_account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Iadc8ac8b9791b23f2186b8b0de370631b0ddaac4","subject":"sq? Test chunked checksum errors in S3 API","status":"NEW","created":"2026-08-21 19:27:38.000000000","updated":"2026-08-31 21:21:09.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8acaa9df55333572449143002bb379fdb36457a8","_number":1001893,"virtual_id_number":1001893,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-21 21:44:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"all":[{"value":1,"date":"2026-08-31 21:21:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-21 19:43:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-31 21:21:09.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"REVIEWER"}],"messages":[{"id":"bb44c568e7903596ba4d80db6d47f698d78a2814","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-08-21 19:27:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"848f99acb9df7b1a35d0ac06ba032704015be805","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 19:43:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1eb27e1a3c7c411195914bbb8cdeaeb2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5eb5c71202a8489a9845383d4b5303d1 : SUCCESS in 4m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10c63593dda147058e6886857594a05e : SUCCESS in 9m 55s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d2b5e1e40e08447cb394fb3d11601e18 : SUCCESS in 6m 07s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/938e0222f3ba44bc9fe1fd764ff2f2f1 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bdc0e62522f45a5a5af75bae0ac9582 : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":1},{"id":"48a4396ad08b5675b3e2a2c9bae7147b56728e7e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 20:15:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a881da8948c4effbf01d00ade4ddd01\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/059968b97b784a78a7b2de22fedc1f31 : SUCCESS in 45m 25s","accounts_in_message":[],"_revision_number":1},{"id":"77e93c5226b0cc773ce3edd9b5e95146588d6acb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-08-21 20:35:48.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"14a174d5bd975496ee5180017aba499288b84d4a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 21:42:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fe83fe2231c458ba181386d9e3ecd5d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ac15d4e4efa94e9ab05f0ae2135e5c1d : SUCCESS in 1h 04m 11s","accounts_in_message":[],"_revision_number":2},{"id":"8982161314148b27fba79bd6b00bc0e53916524c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 21:44:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/091663a43ac04783a2f0fc0fb26be95c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60ee28bf63ad4ec9bba25f890e1f6a9a : SUCCESS in 6m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4773e9a0d42848d182b8a283bea2b602 : SUCCESS in 9m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/2a47f156c7314e6a9a40a9e006e0acda : SUCCESS in 1h 03m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1199b2be7627450a9eeef9533161b74b : SUCCESS in 29m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0468287d872f445d9e09c17f4f557b1b : SUCCESS in 25m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8d85a40cc24f4482973929efe24bab14 : SUCCESS in 56m 22s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/55e333e833d14eaea8bba7e0d3360fa3 : SUCCESS in 16m 35s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/01bc59e81465438698938f46591907cf : SUCCESS in 17m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4dbba04278d14783b1f47db0a3c65866 : SUCCESS in 18m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c31b438f1ab142ffb47721ec995003c6 : SUCCESS in 3m 51s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0580fcef95ad446fa8a538d0dc786296 : SUCCESS in 12m 23s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/6552a552e77b4956ad3bb361b99d1047 : SUCCESS in 14m 46s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/a77131e4bdd84b65bb601a0b6603eed3 : SUCCESS in 13m 06s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/6bdda5e9bd6f48c4b06f6fb9a8c6addc : SUCCESS in 20m 21s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4c48bf9b9e0e43db8e3555e1b4ce56c3 : SUCCESS in 14m 20s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/826fe667ec2e4b2696c72e1472810229 : SUCCESS in 15m 42s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d14ef181042e460e9d05fbad0e1c1d27 : SUCCESS in 15m 39s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/dcbbc9ffe6524fa3b28f501a65388f68 : SUCCESS in 14m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/70968e20774c49e6bf1e0538a854e884 : SUCCESS in 3m 32s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f3d951a7d6c4450ea0b63bdd1698867a : FAILURE in 9m 55s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/cf0cd01a608c427fa1a2e16fa2e2c7c3 : SUCCESS in 7m 53s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/11f1a3e80bd6437f9015e9d9d27d415e : SUCCESS in 50m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/df09218a68414681ae13090bf10313ed : SUCCESS in 33m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c2de1925613e4706819fe19af3add145 : SUCCESS in 51m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1461189874fa4228abda294f75570f6a : SUCCESS in 22m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/576794970066405f996dbb82b7e3c45a : SUCCESS in 4m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7435e6bc350f49b3b075d13d5a4632aa : SUCCESS in 19m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/04641214d110434b95088116da3520d1 : SUCCESS in 36m 26s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"8acaa9df55333572449143002bb379fdb36457a8","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-08-31 21:21:09.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b3d1efd5bac3b75cfb56360c7751ef9980284abd","revisions":{"2468b151d8699fc05cbdadb727070e39fb822914":{"kind":"REWORK","_number":1,"created":"2026-08-21 19:27:38.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/93/1001893/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/93/1001893/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/1001893/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/1001893/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/1001893/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/93/1001893/1"}}},"commit":{"parents":[{"commit":"779d0a3336e94995d14689e16c6995ab88f46568","subject":"s3api: reject checksummed HTTP-chunked PUTs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/779d0a3336e94995d14689e16c6995ab88f46568"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-08-21 19:07:34.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-08-21 19:25:23.000000000","tz":-300},"subject":"sq? Test chunked checksum errors in S3 API","message":"sq? Test chunked checksum errors in S3 API\n\nAdd cross-compatibility coverage for checksummed HTTP-chunked PUT and\nmulti-object delete requests. Verify that S3API reports NotImplemented\nand identifies Transfer-Encoding as the unsupported header.\n\nAssisted-By: Codex:GPT-5\nChange-Id: Iadc8ac8b9791b23f2186b8b0de370631b0ddaac4\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2468b151d8699fc05cbdadb727070e39fb822914"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2468b151d8699fc05cbdadb727070e39fb822914"}]},"branch":"refs/heads/master"},"b3d1efd5bac3b75cfb56360c7751ef9980284abd":{"kind":"REWORK","_number":2,"created":"2026-08-21 20:35:48.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/93/1001893/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/93/1001893/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/1001893/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/1001893/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/1001893/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/93/1001893/2"}}},"commit":{"parents":[{"commit":"779d0a3336e94995d14689e16c6995ab88f46568","subject":"s3api: reject checksummed HTTP-chunked PUTs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/779d0a3336e94995d14689e16c6995ab88f46568"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-08-21 19:07:34.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-08-21 20:34:40.000000000","tz":-300},"subject":"sq? Test chunked checksum errors in S3 API","message":"sq? Test chunked checksum errors in S3 API\n\nAdd cross-compatibility coverage for checksummed HTTP-chunked PUT and\nmulti-object delete requests. Verify that S3API reports NotImplemented\nand identifies Transfer-Encoding as the unsupported header.\n\nUse the request\u0027s chunked-transfer predicate so Transfer-Encoding takes\nprecedence over Content-Length when both headers are present.\n\nAssisted-By: Codex:GPT-5\nChange-Id: Iadc8ac8b9791b23f2186b8b0de370631b0ddaac4\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b3d1efd5bac3b75cfb56360c7751ef9980284abd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b3d1efd5bac3b75cfb56360c7751ef9980284abd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
