)]}'
{"id":"openstack%2Fswift~1001905","triplet_id":"openstack%2Fswift~master~I949114d9e37f7e42f70dd265926289ee63984e32","project":"openstack/swift","branch":"master","attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2026-08-22 01:07:59.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I949114d9e37f7e42f70dd265926289ee63984e32","subject":"Let SLO own copy sysmeta cleanup","status":"NEW","created":"2026-08-22 00:04:09.000000000","updated":"2026-09-18 23:52:14.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":10,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"74a9a2189a0d5f5cef2066baf2292da776412a54","_number":1001905,"virtual_id_number":1001905,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-26 04:34:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-22 00:57:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-22 01:07:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-24 14:24:50.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2026-08-28 00:46:32.000000000","updated_by":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"reviewer":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"state":"CC"},{"updated":"2026-09-18 23:52:14.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"ca78aedb607da5ac572c02b0a2bbca0f4ea1054d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-08-22 00:04:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6b7642d3ae2db627f900cb07b995fdb4c6fbf097","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-22 00:57:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70db1293ab2e469cba3a85c187b22b4b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4706d3ea5fcb4c498395564e432e3280 : SUCCESS in 49m 24s","accounts_in_message":[],"_revision_number":1},{"id":"ac94cd36d73b2130404599366e13e435cbbe3aac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-22 01:07:59.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ba638a9da3004d7b8f454c3066b408f3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40221da28deb40c89bb0250f52a7165d : SUCCESS in 10m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ab727eef994468a88c752394494d355 : SUCCESS in 4m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/ec244760b965418d87a24b4f0658b7ed : SUCCESS in 44m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/95141c75f4e9458fa0390bd57e1ee851 : SUCCESS in 1h 01m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a27426145675451787c1fc9a2ad5bf5e : SUCCESS in 26m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b64666d80c604bfa892ad0269db27417 : SUCCESS in 41m 15s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a50896180c8846abb0a70822c964cd2c : SUCCESS in 16m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/56eb44c5187b45cd9555e8bcfea0ea95 : SUCCESS in 10m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ce33c71d87c747dd9d0f9f562b3a7e54 : SUCCESS in 12m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/79be5781d2ff4f05b61ae63e7219fcfe : SUCCESS in 4m 01s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0b14ce99e3944cc6891230584068a88a : SUCCESS in 19m 24s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/2a55ca9f8f98472b8c81d9f671d9ef24 : SUCCESS in 18m 45s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/78a27109aecb4268ad01c93103680374 : SUCCESS in 20m 12s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/9e28d5144be04e1d9cfcac32efdf5260 : SUCCESS in 17m 41s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c92af85e491840439da9eabd4f24a745 : SUCCESS in 8m 11s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a1c7eba2d1e6485b8bcf94f09a12d55f : SUCCESS in 15m 43s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/09ff4e452bc240b3a6b20ae2e793210a : SUCCESS in 11m 43s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/30d5aef3ed0d4f83825f53e2ecf4d340 : SUCCESS in 9m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8b6925d98e1a49d3825caabe58eaf71d : SUCCESS in 3m 29s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a5c1c5b93ab8481a95acdae22254c5c2 : FAILURE in 4m 14s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/c26bc7c1644f40d5844b66d483f75f81 : SUCCESS in 4m 58s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/901aa8b80f3648f185070f826da17628 : FAILURE in 6m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8518ae4d57494be892f624b89a9a5afe : SUCCESS in 33m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/99002853e9fe48d288a013631bc7c657 : SUCCESS in 33m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c1937a6f032f4fd7ad052e01b4fbcf2b : SUCCESS in 28m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ad217cc573e4a22ae2bd94f39739255 : FAILURE in 3m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/48a5a03c19784f4cb5e2e71116114632 : SUCCESS in 15m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2a9577dd5f494946a7de46f00cc52edb : SUCCESS in 37m 43s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"0e9d6cf04b38ab4f41e943ca886e247fde1b22bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-08-22 03:13:51.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"acb2e57f9588ef34f1d9d2fcf7d68a75b699340f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-22 04:27:13.000000000","message":"Patch Set 2: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/875143c7189f4a3f928f3992c4771277\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db3388112a01429d9249410070f7b78b : SUCCESS in 12m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf36112a32884687a741678e03bc6c8d : SUCCESS in 9m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/987fb1efd02b421e8bb3d499c8348c6a : SUCCESS in 33m 51s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/577b3364c0ee4166b1c2a8dfb4f860b9 : SUCCESS in 1h 01m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/14b69f5f5a484fc6b64cf914a7825b86 : SUCCESS in 48m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e0cbab8ad4af43ed804310c4633ee4cc : SUCCESS in 54m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dc0bf48785254c9c9e06468a2114b8c1 : SUCCESS in 12m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7ecfc5034b194186a9155e19c19d9ddf : SUCCESS in 20m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e307909be04b4299b332e5f6fe05cd52 : SUCCESS in 12m 31s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fa4af9e757f34dc59667ab0cd82f6384 : SUCCESS in 4m 13s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a5e1258b55484a888a1aa4e120f755dc : SUCCESS in 28m 27s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/34e63be2924c41d1ad78392bc1bf2fa0 : SUCCESS in 15m 54s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c1806f1f89764b40ac38a99e9fa18e2c : SUCCESS in 22m 20s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/cc853a57a4fd4cf09457ebe05f897281 : SUCCESS in 21m 55s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/93a05f17e7284f609cfbd43143d4994e : SUCCESS in 14m 54s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3bdbd0bf2d554286b3caad74c1a9fc95 : SUCCESS in 15m 32s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8cf3708e7e6643349977e65c1143662f : SUCCESS in 16m 38s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/53a9f3ca969d48a1b642429792b66617 : SUCCESS in 10m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5b564b16730f479a9e678414ea53448b : SUCCESS in 4m 39s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6758406937c548be967f6329437a88a7 : FAILURE in 8m 31s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/607974625a02494bb63a96d89ce8477f : SUCCESS in 8m 36s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6b145bf485f746c083474566c49dcf40 : SUCCESS in 1h 04m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c65b2b4d30464c84a9996c97a9a00d18 : SUCCESS in 53m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5f616c11980b401287639c0a5e6544a6 : SUCCESS in 53m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0b857a8a51404d73836d31625d0ccb82 : SUCCESS in 21m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75d433ea0dc344809aa68d0074842676 : FAILURE in 4m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8cd59ad18a354787adade2df6329b9f2 : SUCCESS in 20m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/696bcde009cb46a3a37f4016c4cc19d3 : SUCCESS in 53m 21s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"d6e898657a3ca2e6f5def2987c0363dfbb41ea83","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-22 05:26:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bfb88375e8e243a399b9acb602f9a47b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0e68e412c9214cb8bba6069f84e66868 : SUCCESS in 53m 37s","accounts_in_message":[],"_revision_number":2},{"id":"dbf5d5267c52cb9a975b1b5e9f4b939a60e9e8dd","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-08-24 14:24:50.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"5966718fdfd9ab02b6a9614bb479fa8e8d1515f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-08-25 20:58:36.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"624f3b6bf9f5406522ff79b8efed91a012c75126","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 22:05:38.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bc7fa598673b4484a75623e735012d59\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/078e68beb8a34979a7f1604e4c7a7d12 : SUCCESS in 10m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae83a23ccf1543cfb79693cd85ef282d : SUCCESS in 6m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/1f8cf2188dde4bd18613eabe9e454421 : SUCCESS in 1h 03m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/eaf4811da1fe4be09461e6cc5d8bd412 : SUCCESS in 1h 04m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/42f193bcbede4cb9994596df7621c8eb : SUCCESS in 28m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/35ddf61dab5b4accacc3dc1aeaf5fb15 : SUCCESS in 1h 03m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/064d3f26f9d74f29b11f9de7eba8386a : SUCCESS in 19m 16s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ede6be003feb4ce0a3e86b15735c19a6 : SUCCESS in 17m 08s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0b8c3bd9adac44a4ae3030a54b138756 : SUCCESS in 17m 24s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6c2ecdffe70142c7bb508e9a7fb51a43 : SUCCESS in 4m 54s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6818aa6a3dc34832b6269fb482becde5 : SUCCESS in 15m 06s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/e5bc25dd513e4821bf813b85c59df622 : SUCCESS in 13m 09s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/88e7642c9f3d4471883e29b6b8f54dad : SUCCESS in 25m 07s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/e1da2d0ed6204d1b8dfcd08932d08e46 : SUCCESS in 19m 24s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ac1edf28c7024065812cf338ca51f95b : SUCCESS in 14m 06s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/bfb7ae0178c04368ac392536d5190ccd : SUCCESS in 10m 41s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c7f5b9185c49435caa7305b1f9190dcc : SUCCESS in 15m 51s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/ffd511a7083448788bc1963241742886 : SUCCESS in 14m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e1cdd2d7ff9d4b8490522086477dec34 : SUCCESS in 4m 59s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f13136b6cf794ca79f15cf5241911833 : FAILURE in 4m 30s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/18389a1153fe4f22bd57e86709f726a3 : SUCCESS in 8m 08s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c49648d2e33942c7ab3b46fbb3916628 : FAILURE in 1h 00m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/97829dc68bd4458088adfb28c2d5b560 : SUCCESS in 53m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/348713cc71664b5bb5f1b28a581874e3 : SUCCESS in 32m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83485b61565c437f975b2ed782054c87 : SUCCESS in 28m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0086dff1eaef424caf117a072c95e15d : SUCCESS in 4m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bd138d394e8c4fe2a1a8bf478aeca43d : SUCCESS in 18m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bdb864f96398420a9de9a716f2068d9e : SUCCESS in 35m 58s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":3},{"id":"881cd9bb4f2969707b452789b381502b325ae29a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 22:37:30.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1eebafc86c44f57aa0503d533d560e8\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a4e53e05c763456995306f6f3a4445cd : SUCCESS in 1h 01m 53s","accounts_in_message":[],"_revision_number":3},{"id":"e04beeae3e915943ec7950b048c33367ff430583","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-08-26 03:15:45.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"df1d0512e8cb93605f765b8dcee69a066e0a0b6c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 03:16:02.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/swift for 1001905,4","accounts_in_message":[],"_revision_number":4},{"id":"81d60c4dbb29f38469e33c4efc4a3666a4dbdb75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-08-26 03:21:21.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"25816fa1926362fc35524c3aefc3605c269c3044","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 04:34:04.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fc29686f8ce42a28b5aee5d260a9a7c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3978803dd45849e6b2eba1fda8f447fd : SUCCESS in 10m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a84cf76c104b444ea0983078bfb85b91 : SUCCESS in 9m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/d12f1a725ee3420ea0393fdcabee3240 : SUCCESS in 1h 02m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/df4b6964283643d394cce5dedde872fc : SUCCESS in 55m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/69b86d8364964e9793b3456f4d95a0c4 : SUCCESS in 40m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a5c30dbda79d49d29125522a729a947d : SUCCESS in 1h 03m 19s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9cf3c0dad8a44025b44a1d2dc2478b9f : SUCCESS in 10m 35s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ac487db405684c7ca6a719c8693a58e8 : SUCCESS in 17m 33s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b78f723aff694c2687cf7ccc06c9d311 : SUCCESS in 18m 52s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e5a4511647ed42979f3a12e4a6ac656a : SUCCESS in 6m 47s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e728c1fa439a457bb3c3eeb1c8ccedcc : SUCCESS in 20m 16s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b16f1efc461e467a8f51c651c0e5d999 : SUCCESS in 22m 13s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/71f6cd4de3184d6c98fc387a78ac63f9 : SUCCESS in 24m 22s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/cd5eb4cb5dcc40a0823165d6334a2512 : SUCCESS in 18m 09s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/05efc2ac402a47d299b72124820a2dae : SUCCESS in 13m 41s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0a9d75171f234d238b8f34238cc9bfb3 : SUCCESS in 15m 40s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6e5bbe00258d4e29a85e3e3038acb25a : SUCCESS in 16m 11s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/2b65f559141042058ac59f1ea9efb325 : SUCCESS in 17m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/69ba5d7bc98d4ea3a039415a16e1b5bd : SUCCESS in 6m 59s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0f7ad5259964491abdd0f94bb265dd93 : FAILURE in 7m 00s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/dba3ef8b2a684dd4a9b0f81d07b2f2a0 : SUCCESS in 9m 07s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b85cc2185853407a9d733d15388214f4 : SUCCESS in 41m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6e4d69363d7540d3b8693597788b0d1d : SUCCESS in 57m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ab5e2728b9814b40a1af2d1e26463fc0 : SUCCESS in 56m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0c63818f9915480eb70a9bdebf8a0eff : SUCCESS in 13m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a33959993f54a439b2250444588a22d : SUCCESS in 5m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/60b6875e97b24394a76843df6a6d8dca : SUCCESS in 27m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e72b0ee6ae92465d820c95594dbd3325 : SUCCESS in 1h 00m 49s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":5},{"id":"3ceeaf6f3a779fe0c8b7115504723b909009c113","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 08:24:40.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76bd0d24260a4f64902098750a226509\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/eb0f68bda8e4403fb6dc1be4854959c0 : SUCCESS in 53m 57s","accounts_in_message":[],"_revision_number":5},{"id":"32f0c57a4275082bde1bbff41f6cdbcd35b33793","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-28 00:46:32.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"74a9a2189a0d5f5cef2066baf2292da776412a54","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-18 23:52:14.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"71e43b790e47708549ea2322f663a97f0353cd7f","revisions":{"e1ef370a864c5c60dcb8acb97a9560c3e39f576c":{"kind":"REWORK","_number":1,"created":"2026-08-22 00:04:09.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/05/1001905/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/1001905/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/1001905/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/1001905/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/1001905/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/1001905/1"}}},"commit":{"parents":[{"commit":"e7d5e5d701e20a0fcba956c59f410d56ad6aa369","subject":"sq? add s3api copy sysmeta probe","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e7d5e5d701e20a0fcba956c59f410d56ad6aa369"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-08-21 23:35:49.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-08-21 23:35:49.000000000","tz":-300},"subject":"WIP: let SLO own copy sysmeta cleanup","message":"WIP: let SLO own copy sysmeta cleanup\n\nLet middleware that served a copy source attach a callback to the\nsource response environ. Run that callback before the callback from\nthe original request.\n\nUse the interface in SLO to remove its source sysmeta. This lets\ns3api limit its hook to S3-owned sysmeta.\n\nAssisted-By: Codex:gpt-5\nChange-Id: I949114d9e37f7e42f70dd265926289ee63984e32\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e1ef370a864c5c60dcb8acb97a9560c3e39f576c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e1ef370a864c5c60dcb8acb97a9560c3e39f576c"}]},"branch":"refs/heads/master"},"afbc3c6f9ecdecace34f9456e647d04a5d44de76":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-08-22 03:13:51.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/05/1001905/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/1001905/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/1001905/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/1001905/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/1001905/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/1001905/2"}}},"commit":{"parents":[{"commit":"223372e29e84741cce9a690e742efe8f2eb73b63","subject":"sq? add s3api copy sysmeta probe","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/223372e29e84741cce9a690e742efe8f2eb73b63"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-08-21 23:35:49.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-08-22 02:40:38.000000000","tz":-300},"subject":"WIP: let SLO own copy sysmeta cleanup","message":"WIP: let SLO own copy sysmeta cleanup\n\nLet middleware that served a copy source attach a callback to the\nsource response environ. Run that callback before the callback from\nthe original request.\n\nUse the interface in SLO to remove its source sysmeta. This lets\ns3api limit its hook to S3-owned sysmeta.\n\nAssisted-By: Codex:gpt-5\nChange-Id: I949114d9e37f7e42f70dd265926289ee63984e32\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/afbc3c6f9ecdecace34f9456e647d04a5d44de76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/afbc3c6f9ecdecace34f9456e647d04a5d44de76"}]},"branch":"refs/heads/master"},"e663fdcf6505ed5092bf99f97fce35aa8110d1d9":{"kind":"REWORK","_number":3,"created":"2026-08-25 20:58:36.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/05/1001905/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/1001905/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/1001905/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/1001905/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/1001905/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/1001905/3"}}},"commit":{"parents":[{"commit":"3c821b84177f8c95e7e402f60a016f56cb04101f","subject":"probe tests: extend SLO/MPU copy probes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3c821b84177f8c95e7e402f60a016f56cb04101f"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-08-21 23:35:49.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-08-25 20:57:29.000000000","tz":-300},"subject":"Let SLO own copy sysmeta cleanup","message":"Let SLO own copy sysmeta cleanup\n\nA server-side copy of an SLO as a regular object must not preserve the\nsource\u0027s SLO sysmeta. s3api used to remove that metadata while handling\nmultipart uploads, which made an SLO concern depend on the s3api\nmiddleware.\n\nAllow source-serving middleware to add a copy callback to the source\nresponse environ, and invoke it before the original-request callback.\nKeep source and original request hooks separate so each runs once.\nHave SLO use its hook to remove its own source sysmeta, leaving s3api\nto clean only S3-owned metadata.\n\nAssisted-By: Codex:gpt-5\nChange-Id: I949114d9e37f7e42f70dd265926289ee63984e32\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e663fdcf6505ed5092bf99f97fce35aa8110d1d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e663fdcf6505ed5092bf99f97fce35aa8110d1d9"}]},"branch":"refs/heads/master"},"c85c6ae051eec69bab3af82c479b1275a2ec3e98":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-08-26 03:15:45.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/05/1001905/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/1001905/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/1001905/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/1001905/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/1001905/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/1001905/4"}}},"commit":{"parents":[{"commit":"e05cfd56422d07a29d7f52493ced20d26c65d699","subject":"probe tests: extend SLO/MPU copy probes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e05cfd56422d07a29d7f52493ced20d26c65d699"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-08-21 23:35:49.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-08-26 03:15:45.000000000","tz":0},"subject":"Let SLO own copy sysmeta cleanup","message":"Let SLO own copy sysmeta cleanup\n\nA server-side copy of an SLO as a regular object must not preserve the\nsource\u0027s SLO sysmeta. s3api used to remove that metadata while handling\nmultipart uploads, which made an SLO concern depend on the s3api\nmiddleware.\n\nAllow source-serving middleware to add a copy callback to the source\nresponse environ, and invoke it before the original-request callback.\nKeep source and original request hooks separate so each runs once.\nHave SLO use its hook to remove its own source sysmeta, leaving s3api\nto clean only S3-owned metadata.\n\nAssisted-By: Codex:gpt-5\nChange-Id: I949114d9e37f7e42f70dd265926289ee63984e32\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c85c6ae051eec69bab3af82c479b1275a2ec3e98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c85c6ae051eec69bab3af82c479b1275a2ec3e98"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"3c821b84177f8c95e7e402f60a016f56cb04101f","ours":"e663fdcf6505ed5092bf99f97fce35aa8110d1d9","theirs":"e05cfd56422d07a29d7f52493ced20d26c65d699","merge_strategy":"recursive","contains_conflicts":false}},"71e43b790e47708549ea2322f663a97f0353cd7f":{"kind":"REWORK","_number":5,"created":"2026-08-26 03:21:21.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/05/1001905/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/1001905/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/1001905/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/1001905/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/1001905/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/1001905/5"}}},"commit":{"parents":[{"commit":"95138237d92c1eed34a6d9cdca7afb691229e795","subject":"probe tests: extend SLO/MPU copy probes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/95138237d92c1eed34a6d9cdca7afb691229e795"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-08-21 23:35:49.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-08-26 03:17:16.000000000","tz":-300},"subject":"Let SLO own copy sysmeta cleanup","message":"Let SLO own copy sysmeta cleanup\n\nA server-side copy of an SLO as a regular object must not preserve the\nsource\u0027s SLO sysmeta. s3api used to remove that metadata while handling\nmultipart uploads, which made an SLO concern depend on the s3api\nmiddleware.\n\nAllow source-serving middleware to add a copy callback to the source\nresponse environ, and invoke it before the original-request callback.\nKeep source and original request hooks separate so each runs once.\nHave SLO use its hook to remove its own source sysmeta, leaving s3api\nto clean only S3-owned metadata.\n\nAssisted-By: Codex:gpt-5\nChange-Id: I949114d9e37f7e42f70dd265926289ee63984e32\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/71e43b790e47708549ea2322f663a97f0353cd7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/71e43b790e47708549ea2322f663a97f0353cd7f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
