)]}'
{"id":"openstack%2Fswift~1002427","triplet_id":"openstack%2Fswift~master~I64b4ac85436d8a6cceb1ed046ce45f70b3dce94c","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I64b4ac85436d8a6cceb1ed046ce45f70b3dce94c","subject":"zuul: pin s3api compat jobs to SigV2 or SigV4 by name","status":"NEW","created":"2026-08-26 00:53:33.000000000","updated":"2026-08-26 02:21:46.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"81b7164ef959323d158bbccc10fd3792adccffe7","_number":1002427,"virtual_id_number":1002427,"owner":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-08-26 02:04:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2026-08-26 02:04:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"02fd474f9369a2596ee3251b83bcce303a71b062","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2026-08-26 00:53:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"04826f16e80597172ba7ae8637ae0566f0a32137","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 02:04:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a981e21ad60b49409138ed370046f6cf\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/eb05d1f863224902bf4c6ee116497758 : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/858dd7fd8da14dd8a6a8e1c9b9731ff5 : SUCCESS in 9m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/75f5c9b9521d4281a09a0121105f160d : SUCCESS in 9m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/db71439f54f5470cb82abdc5f7b684b2 : SUCCESS in 1h 06m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/82112823b244419aa5fb7a101b9a0d9a : SUCCESS in 1h 06m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0f220c48e90a4102b2950be81236c11c : SUCCESS in 51m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bdceee1cb464467c9fda279ae4fcbc35 : SUCCESS in 59m 05s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ffea1b98500841d5b2230520d6d19903 : FAILURE in 23m 13s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/38e63018e9be45f3bdf96ff83ee9fdfc : FAILURE in 23m 14s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3ea5cb8f997042e1aa773203699a6a20 : FAILURE in 25m 26s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/dd47a33dba9a49ae9f9dc0d80a1787c1 : SUCCESS in 6m 46s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e0c77ecc2b76445794aefb6eb645e210 : SUCCESS in 16m 01s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/0436ec1b95cb43809bab1d1172f305d6 : SUCCESS in 20m 54s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/618d5277b7424e278b47f91d12446a9d : SUCCESS in 32m 31s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/c1345d4d8ee84edb82a6b8bffc4eb890 : SUCCESS in 21m 00s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a08e01456f7a47f79218e7d215b2c8a3 : FAILURE in 19m 54s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/49056128e8004241870a0376e82ffe81 : FAILURE in 20m 29s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0646df90cd124c419e5e361583fd3996 : FAILURE in 18m 51s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/5ece17f6ac6c4b1e9a194926d12ea2fd : FAILURE in 18m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cea1c08709254cfca54bd600ba94b895 : SUCCESS in 7m 09s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/54ce6d76a91d45228ff33dc4d870352c : FAILURE in 7m 39s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth-sigv4 https://zuul.opendev.org/t/openstack/build/e2ae7ec2186c499f8f5fbe8483ba387d : SUCCESS in 8m 56s\n- swift-tox-func-s3api-compat-tests-tempauth-sigv2 https://zuul.opendev.org/t/openstack/build/f3e2c6fb77204ded9c767ca77bf59522 : SUCCESS in 6m 06s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/cc7cd48f7b354a8282c15776aca4544d : SUCCESS in 1h 03m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e663644dffdb4d42abc0468c4e21492f : FAILURE in 33m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8c8acc9644ec4105bb9f4abaa0cf7674 : FAILURE in 33m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e1f62e9aba1b4bfab7c9903994710e3e : SUCCESS in 29m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/200fb831a3c143c0944ef6d3be7dfe79 : FAILURE in 4m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ac65a8d543f34fc7ad32736b34dddc4a : SUCCESS in 18m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/316075fa199049b890c657c993edce05 : SUCCESS in 55m 17s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  Comments left for invalid file test/functional/s3api/test_bucket.py","accounts_in_message":[],"_revision_number":1},{"id":"81b7164ef959323d158bbccc10fd3792adccffe7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 02:21:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fcc66647a32748a3a771f93d5299fbdf\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4e17f7847c5e4972b6890db8ee723058 : SUCCESS in 35m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"88a841c1605288cfa8f77b8a3015e50144ac7a13","revisions":{"88a841c1605288cfa8f77b8a3015e50144ac7a13":{"kind":"REWORK","_number":1,"created":"2026-08-26 00:53:33.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/27/1002427/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/27/1002427/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/1002427/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/1002427/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/1002427/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/27/1002427/1"}}},"commit":{"parents":[{"commit":"eaf061d95310f692633b87f3486ab06eb913d752","subject":"tests: Port boto2 tests to boto3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eaf061d95310f692633b87f3486ab06eb913d752"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2026-08-26 00:50:58.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2026-08-26 00:52:57.000000000","tz":-420},"subject":"zuul: pin s3api compat jobs to SigV2 or SigV4 by name","message":"zuul: pin s3api compat jobs to SigV2 or SigV4 by name\n\nMake the signature version used by s3api func/compat tests explicit so\nCI is not an implicit mix of SigV4 (tox -e s3api default) and SigV2\n(functional s3api default). Jobs set SWIFT_TEST_S3_SIGNATURE; mismatching\nSigV2/SigV4 subclasses skip. s3_acl stays enabled on these SAIO jobs.\n\nDepends-On: I0104ad018b845e70caf8bc6368aaff8fc6156354\nChange-Id: I64b4ac85436d8a6cceb1ed046ce45f70b3dce94c\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/88a841c1605288cfa8f77b8a3015e50144ac7a13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/88a841c1605288cfa8f77b8a3015e50144ac7a13"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
