)]}'
{"id":"openstack%2Fswift~1002677","triplet_id":"openstack%2Fswift~master~Ic3fe665c1c8536400c84b1364f8ae855de7bf018","project":"openstack/swift","branch":"master","topic":"s3-checksum-persistence","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-08-28 08:51:20.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ic3fe665c1c8536400c84b1364f8ae855de7bf018","subject":"sq? s3request validate POST checksum fixup","status":"ABANDONED","created":"2026-08-27 17:51:16.000000000","updated":"2026-08-28 08:51:20.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ffbce7cfe7e095c05bc2f5a6d10bbad89d8e67b8","_number":1002677,"virtual_id_number":1002677,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-27 19:05:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"all":[{"value":1,"date":"2026-08-27 19:05:15.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-27 19:05:15.000000000","updated_by":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"reviewer":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"state":"REVIEWER"},{"updated":"2026-08-27 19:05:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7db7aa230cfcde902fbc22c73fb7b662fdf886d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-08-27 17:51:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e3b981cd08369290c9ddead105b318a5f3b0208e","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-27 19:05:15.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"9455947870d20a257089b9a30d7c4d3dedff7e6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 19:05:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/320c0f81b5d3486eb193fab69c5d9fbc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b01454f9462459884e117832478aeb0 : SUCCESS in 5m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d1119d71e074f858ee04b86a46bd037 : SUCCESS in 9m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/843fa0c51c78487f977daa36caa2f1e5 : SUCCESS in 1h 01m 40s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1d6394ba33344f548d849ef30b55ea76 : SUCCESS in 1h 07m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/73bb1b00c8ab4386a0249729f4962e5e : SUCCESS in 53m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dab54e5d863f4c4ea8278f9349899bc6 : SUCCESS in 58m 34s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ec82e48c0b64461880f1314fad2a3e66 : SUCCESS in 10m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/318846a03792451ca1e17b97d8f84721 : SUCCESS in 23m 39s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dd964491c02c48c6bca379ad73fec230 : SUCCESS in 12m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a8896b13ba384d409d28423cb6422a46 : SUCCESS in 7m 10s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/69a7cc051e2542cb9dde457874a833d8 : SUCCESS in 24m 13s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ab163607e67e4d5d8e22ea32d03f7528 : SUCCESS in 17m 37s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/70c8490c2fe34d728831656c917155e9 : SUCCESS in 22m 27s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/97c181432d904519b0df2b83d06c2327 : SUCCESS in 20m 31s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/edb179803b1948e594be99c06dace591 : SUCCESS in 15m 10s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/61fc6829f8c4411aa248b581b6a525bc : SUCCESS in 15m 14s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ee52f864ffa14bf9b0804b85a2fb4753 : SUCCESS in 18m 36s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/40573899e901426783b7cf665808c18a : SUCCESS in 17m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0e0b49304b9d4f8ea1814adf9633f9e4 : SUCCESS in 6m 42s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b59957faea2b42e0bf6e15c21be9ac79 : FAILURE in 9m 46s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/65534a6a67b143e6b789d38077ab956b : SUCCESS in 8m 43s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b77a82dc957149dc97043700c6fccb23 : SUCCESS in 1h 05m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bad1cb83f3c54c78bcfd7850b5eac503 : SUCCESS in 49m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ecc020280a2a4b91a1b68c0ff75aa5a7 : SUCCESS in 42m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ca36270f6474f31b6075923d7a11fdf : SUCCESS in 30m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93cc543efba54047a83d47b04b4b337a : SUCCESS in 4m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/59211f9119fc4b42911be08214c650b9 : SUCCESS in 11m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8cee7dd5e07b4714b73699aba66e9bcb : SUCCESS in 51m 02s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"5560d68fda2080e1479e0ac925170466adcfa262","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 22:02:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb36fc09118642d698d124e95a4809f4\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a88791096f4c462bbb71a6215975b495 : SUCCESS in 1h 34m 47s","accounts_in_message":[],"_revision_number":1},{"id":"0c2b96782e4c1df06f0a55fe7ae3cf2dab26076e","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-08-27 22:03:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ffbce7cfe7e095c05bc2f5a6d10bbad89d8e67b8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-08-28 08:51:20.000000000","message":"Abandoned\n\nsquashed","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"296f5307aa09fbb01cfc4c823437249c021cb5ac","revisions":{"296f5307aa09fbb01cfc4c823437249c021cb5ac":{"kind":"REWORK","_number":1,"created":"2026-08-27 17:51:16.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/77/1002677/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/77/1002677/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/1002677/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/1002677/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/1002677/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/77/1002677/1"}}},"commit":{"parents":[{"commit":"62e4431de37cd3f85d71d535f7fdbfb9eb5742d5","subject":"s3api: don\u0027t base64-validate checksum values on POST","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/62e4431de37cd3f85d71d535f7fdbfb9eb5742d5"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-08-27 17:50:20.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-08-27 17:50:20.000000000","tz":60},"subject":"sq? s3request validate POST checksum fixup","message":"sq? s3request validate POST checksum fixup\n\nChange-Id: Ic3fe665c1c8536400c84b1364f8ae855de7bf018\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/296f5307aa09fbb01cfc4c823437249c021cb5ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/296f5307aa09fbb01cfc4c823437249c021cb5ac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
