)]}'
{"id":"openstack%2Fswift~1002695","triplet_id":"openstack%2Fswift~master~Ic5dcb5a554b45bfaa68e52959c2eae80a97f6776","project":"openstack/swift","branch":"master","attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2026-08-27 22:29:49.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ic5dcb5a554b45bfaa68e52959c2eae80a97f6776","subject":"WIP: reconstructor: cache suffix audits","status":"NEW","created":"2026-08-27 21:14:17.000000000","updated":"2026-08-27 22:37:21.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ec558bce809a2194e168850aa0b9e0cbb2cdad85","_number":1002695,"virtual_id_number":1002695,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-08-27 22:29:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-27 22:29:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"50dec6828cec4dafd898b2d7080c206aba116ca4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-08-27 21:14:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"81c2e2eb205df7ebe4051a7738159d7ef00fe097","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 22:29:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1d0fab0cdb5342c5ad97a8adeba78c83\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f48ebe4787a42d3a784676719e3506e : SUCCESS in 9m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/955229cfb9904eec84c4c94c7e6f535d : SUCCESS in 10m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/869371952a0d4df3a286b61b00bdb6cf : SUCCESS in 1h 04m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0ddfaebd251249d3bcdeac8fc35f312a : SUCCESS in 1h 10m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e04c1b24be37404dbe6a8cbf3ac24708 : SUCCESS in 54m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/62b818cc9a394284a1cc856383b5e5c1 : SUCCESS in 1h 02m 53s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0b5eae05b6b944998b5dbff3b1865520 : SUCCESS in 21m 02s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e9254c46f4ae46d688dc92af74978fe3 : SUCCESS in 21m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/19a9407ade4e4455a2027861bb46f079 : SUCCESS in 12m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/123fdda378fe42a5885a3f2de6a39da6 : SUCCESS in 5m 01s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/381223456cfd405aaba4ad854c9d3a49 : FAILURE in 11m 48s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/a2dd92960c9e4bd185c6bb9e7beec8de : SUCCESS in 21m 05s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/e297c924b864406eaf0612eba497150c : SUCCESS in 34m 39s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/40e314e18f974584a0b7f0a2a2f73675 : SUCCESS in 8m 48s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/676da93d61f54ae880e1e06948a448d1 : SUCCESS in 13m 14s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/02bb4097819a4c2789547be7c2e7f320 : SUCCESS in 15m 50s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d4bf675dc93449b6ba6224a251daad3d : SUCCESS in 8m 18s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/8d950cb675f1447b89777c097b04620a : SUCCESS in 18m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c2235375269a4f44b36cb0de80207e95 : SUCCESS in 5m 47s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c8ed6ac0f44f4a36bc67d73212a2dea5 : FAILURE in 7m 42s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/ca4596062f264b1aaf1c0c824dd01c56 : SUCCESS in 5m 37s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/65fdea747ba042698663a6e9c3bb0405 : SUCCESS in 39m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9b6a169c004e46669eda2af93f59944b : SUCCESS in 56m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4cc3e6ff3d9c4d32899b6ba3d5125a15 : SUCCESS in 53m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/99271c4afaab4375833530b8e65a16d7 : FAILURE in 22m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/408feaf6e23a438882179682c77f8ead : SUCCESS in 2m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8dfb757cfb3d4d62b54a1de9cb0b83ab : SUCCESS in 20m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aded02ded8554156a0d9fefc551066fe : SUCCESS in 27m 23s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"ec558bce809a2194e168850aa0b9e0cbb2cdad85","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 22:37:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47dce54bfefb40bfa8f3b02c9a47ffa9\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/dba53b6365a746979e565b23fcb7e605 : SUCCESS in 1h 19m 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4faa7559a20ada2c032e12a1c56c67aef221c1fb","revisions":{"4faa7559a20ada2c032e12a1c56c67aef221c1fb":{"kind":"REWORK","_number":1,"created":"2026-08-27 21:14:17.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/95/1002695/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/95/1002695/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/1002695/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/1002695/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/1002695/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/95/1002695/1"}}},"commit":{"parents":[{"commit":"7349b92c7c3f1304ff82f49c1e7c3b79b2621d69","subject":"reconstructor: rsync pure-handoff reverts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7349b92c7c3f1304ff82f49c1e7c3b79b2621d69"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-08-26 14:41:07.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-08-27 19:23:44.000000000","tz":-300},"subject":"WIP: reconstructor: cache suffix audits","message":"WIP: reconstructor: cache suffix audits\n\nThis WIP currently combines a context cleanup, a cache-safety change,\nand a rotated audit.\n\nThe part_info/path normalization is a prerequisite-only cleanup and\nshould land separately. Generalizing empty-partition cleanup protects\nno-job partitions from an incorrect empty cache; it is independent of\nrsync, and must precede a change that trusts the cache more often.\n\nThe rotating ten-percent audit shares the replicator\u0027s helper. That\nis also independent of rsync, but it is the performance improvement\nthat makes cached pure-handoff partitions cheap between audits.\n\nAssisted-By: Codex:gpt-5\nChange-Id: Ic5dcb5a554b45bfaa68e52959c2eae80a97f6776\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4faa7559a20ada2c032e12a1c56c67aef221c1fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4faa7559a20ada2c032e12a1c56c67aef221c1fb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
