)]}'
{"id":"openstack%2Fswift~1002711","triplet_id":"openstack%2Fswift~master~Iee4be4203ff5a3cc4449fddfe3220a66f5cfa44e","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"Iee4be4203ff5a3cc4449fddfe3220a66f5cfa44e","subject":"DNM: contrast backend path and base-label encoding","status":"NEW","created":"2026-08-28 00:01:34.000000000","updated":"2026-08-28 02:24:00.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a53a0be4db3074ec8124c3cc09f8fb886231431e","_number":1002711,"virtual_id_number":1002711,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-28 01:17:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-28 01:17:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7bd41c9300fa31aa617af5763c02b501f6ff7389","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-08-28 00:01:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d73d1d7ed91631d66949b47a20579e6a11e2c723","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-08-28 00:48:07.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"5af67454a0a663d770134300d4ce26a9a405ffae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 01:17:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea1f178973524a1588ca380c1b91b88f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88f93944e21d4536a3b264e2ea7a27fd : SUCCESS in 4m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d41626879d9e44a19903945118ec65b3 : SUCCESS in 4m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/79395e8028cf4d048982729d4071d998 : SUCCESS in 16m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2c232869a6484c3c93db9ede1032d1ae : SUCCESS in 10m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/650babbe3f114864bc03a63e7f4b5a04 : SUCCESS in 11m 52s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/74ca9cd85cf9498083a4bab9b65a00d2 : SUCCESS in 5m 46s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/574667c6bce74eb19d4e29bffa770cf8 : SUCCESS in 24m 29s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/0156b15f4ffc442e91358351d77a726f : SUCCESS in 19m 23s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c80d7a699743443c81a8566c33fc69bf : SUCCESS in 12m 46s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/e562751ec72149e2a554a6565382acbe : SUCCESS in 17m 20s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e6e073ade27748cdb86621a30d3618f6 : SUCCESS in 14m 13s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e1008620eebb405283409c2c530c5f03 : SUCCESS in 13m 06s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d6c7ca871e924a8f964966eee606bcc7 : SUCCESS in 15m 06s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/8aca6219045b46cb8ae268305f54a442 : SUCCESS in 10m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/633952650b04484ea4f5b8b268c8e47d : SUCCESS in 25m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/47e3af0923c34e93a3816664e6f3d49c : SUCCESS in 25m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/38ab92551f844f09a247353dedc17eb4 : SUCCESS in 10m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0946bc04b0f64af2acd751ac6fc7a9e5 : SUCCESS in 3m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/985f524e1f564364a1d1b864887557f5 : SUCCESS in 19m 05s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"a53a0be4db3074ec8124c3cc09f8fb886231431e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 02:24:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e443c6d0d542493db5d47fc79fb22306\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/56f72e2604e74281898b03b3a04b34a1 : SUCCESS in 1h 07m 11s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"215e00c554e6966e4712ad598e50b8d6388bc911","revisions":{"3d446365d96fbdb2fde1fb68a77036f2e7bf7f27":{"kind":"REWORK","_number":1,"created":"2026-08-28 00:01:34.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/11/1002711/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/1002711/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/1002711/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/1002711/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/1002711/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/1002711/1"}}},"commit":{"parents":[{"commit":"af0e6818c5f54d1ff6e1782e3ae042d64e36709e","subject":"Merge \"Clarify s3api.utils sysmeta helper names\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/af0e6818c5f54d1ff6e1782e3ae042d64e36709e"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-08-27 23:43:48.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-08-27 23:53:53.000000000","tz":-300},"subject":"test: contrast backend path and base-label encoding","message":"test: contrast backend path and base-label encoding\n\nShow that a Unicode account\u0027s URL-quoted base label differs from raw\nWSGI backend PATH_INFO after one access-log URL decoding. Add a\ncompanion request-path check, whose decoded log field matches the\nbase-label form.\n\nAssisted-By: Codex:gpt-5\nChange-Id: Iee4be4203ff5a3cc4449fddfe3220a66f5cfa44e\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3d446365d96fbdb2fde1fb68a77036f2e7bf7f27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3d446365d96fbdb2fde1fb68a77036f2e7bf7f27"}]},"branch":"refs/heads/master"},"215e00c554e6966e4712ad598e50b8d6388bc911":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-08-28 00:48:07.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/11/1002711/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/1002711/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/1002711/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/1002711/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/1002711/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/1002711/2"}}},"commit":{"parents":[{"commit":"af0e6818c5f54d1ff6e1782e3ae042d64e36709e","subject":"Merge \"Clarify s3api.utils sysmeta helper names\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/af0e6818c5f54d1ff6e1782e3ae042d64e36709e"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-08-27 23:43:48.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-08-28 00:48:07.000000000","tz":0},"subject":"DNM: contrast backend path and base-label encoding","message":"DNM: contrast backend path and base-label encoding\n\nShow that a Unicode account\u0027s URL-quoted base label differs from raw\nWSGI backend PATH_INFO after one access-log URL decoding. Add a\ncompanion request-path check, whose decoded log field matches the\nbase-label form.\n\nAssisted-By: Codex:gpt-5\nChange-Id: Iee4be4203ff5a3cc4449fddfe3220a66f5cfa44e\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/215e00c554e6966e4712ad598e50b8d6388bc911"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/215e00c554e6966e4712ad598e50b8d6388bc911"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
