)]}'
{"id":"openstack%2Fswift~1002714","triplet_id":"openstack%2Fswift~master~Ibea5e17f0bd1a91c7c9489d131df24d6f77af12d","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"Ibea5e17f0bd1a91c7c9489d131df24d6f77af12d","subject":"DNM: characterize unescaped DogStatsD labels","status":"NEW","created":"2026-08-28 00:55:06.000000000","updated":"2026-08-28 02:15:02.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fd9196d617eac6e1770b2be4766fd6737257466b","_number":1002714,"virtual_id_number":1002714,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-28 01:39:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-28 01:39:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0e638919c1686b219651cb097c8ddd546958668b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-08-28 00:55:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fca2694d5aa8e03413492f582b02b73fbd075aab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 01:39:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a00043b855d049789aa9be7b74e08e0c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/504d157b81d64c9f85c21b35cdf49321 : SUCCESS in 5m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f2ce21e06bd4512bd30484b55c265e8 : SUCCESS in 9m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1daa018c92ec41a1abbbe4fcbea8263d : SUCCESS in 19m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ef5e51e24f6b43bca3b7c047cfa31623 : SUCCESS in 22m 06s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e0b51cba95e44329a778cb8f947de4bc : SUCCESS in 11m 52s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/36898b29ea5f43e98010a08e97dfd3c6 : SUCCESS in 5m 46s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b5c7848b2fcc43dd98a6ae8cc301f6b4 : SUCCESS in 12m 28s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/68836058f5b54fefb318113d617f81fa : SUCCESS in 16m 03s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/0de6b9a409474fb38c840a2b345910c7 : SUCCESS in 24m 53s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/99fa059449b74fc09f192fcc6b4cd5df : SUCCESS in 15m 56s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5ad7f317c9424daab0cef6ecb2a8853d : SUCCESS in 13m 46s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/362485fde59543c1a1081186a83e6449 : SUCCESS in 12m 46s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c2f9c4e1828741709bd297bc960634b0 : SUCCESS in 12m 49s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/8f404bcac7654fa38546fca6af145bf9 : SUCCESS in 16m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b1b7cc2aece048058ea921b24eb8b795 : SUCCESS in 40m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0712addefc554bd79df8def8626317e0 : SUCCESS in 40m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c140ef9ffe3447389029c674005b9c00 : SUCCESS in 29m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/951e81ec9dac400ca4b79284994b9081 : SUCCESS in 3m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fae599925adc45cdb3e488544df6ceeb : SUCCESS in 8m 38s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"fd9196d617eac6e1770b2be4766fd6737257466b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 02:15:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2cf0ff4acc194b66a7aae2ae09cc8a95\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/59fd815587ba4473bbe81a87441c24ef : SUCCESS in 1h 05m 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2e851d7a705eee4cb5b354446e8e2d52f8951d8a","revisions":{"2e851d7a705eee4cb5b354446e8e2d52f8951d8a":{"kind":"REWORK","_number":1,"created":"2026-08-28 00:55:06.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/14/1002714/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/1002714/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/1002714/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/1002714/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/1002714/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/1002714/1"}}},"commit":{"parents":[{"commit":"af0e6818c5f54d1ff6e1782e3ae042d64e36709e","subject":"Merge \"Clarify s3api.utils sysmeta helper names\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/af0e6818c5f54d1ff6e1782e3ae042d64e36709e"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-08-28 00:53:28.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-08-28 00:53:28.000000000","tz":-300},"subject":"DNM: characterize unescaped DogStatsD labels","message":"DNM: characterize unescaped DogStatsD labels\n\nChange-Id: Ibea5e17f0bd1a91c7c9489d131df24d6f77af12d\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2e851d7a705eee4cb5b354446e8e2d52f8951d8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2e851d7a705eee4cb5b354446e8e2d52f8951d8a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
