)]}'
{"id":"openstack%2Fswift~1002830","triplet_id":"openstack%2Fswift~master~I76783b7c85dae1b40fc3abcebc03cd9915df4840","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-09-03 15:06:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I76783b7c85dae1b40fc3abcebc03cd9915df4840","subject":"trivial: Remove some py2 comments and validation","status":"MERGED","created":"2026-08-28 21:42:15.000000000","updated":"2026-09-03 15:09:09.000000000","submitted":"2026-09-03 15:06:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1002830","meta_rev_id":"4c29e7631b22dbc61c440c006d5fb7b1eb909649","_number":1002830,"virtual_id_number":1002830,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-03 15:06:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"all":[{"value":2,"date":"2026-09-03 13:36:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"all":[{"value":1,"date":"2026-09-03 13:36:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-28 22:22:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-28 22:55:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-03 13:36:53.000000000","updated_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"reviewer":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"state":"REVIEWER"}],"messages":[{"id":"c40cf33773bb497d191b56ee683f1ca4ba072633","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-08-28 21:42:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"10c4902251b038a5f3581195588215a03c99e17b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 22:22:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb3b916e597449c6a4963aecae7a13b2\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/390deba6449a4b8589b82c3dea2dfcf2 : SUCCESS in 37m 04s","accounts_in_message":[],"_revision_number":1},{"id":"d758926849078c9c0e1fca41a98685534023a300","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 22:55:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8ff6a356b61467a98edecad35d5a326\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6b4f701fe0e425b979b2be74109ec24 : SUCCESS in 7m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/44609575971049fbb0901a9751c846a7 : SUCCESS in 6m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/a8e7d0a13adc412c82dd1fd629189d2d : SUCCESS in 59m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fa8ba2520da943b1b3a065cb6e332290 : SUCCESS in 1h 06m 40s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6275f11e975a4230b6045d3f19eb4967 : SUCCESS in 32m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7d6d3393ee64470c9a053e68965f1c23 : SUCCESS in 1h 06m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/123d13f91aaa4068a92522b4931421e2 : SUCCESS in 19m 44s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/87f01632d866441cb157f0e69b89271a : SUCCESS in 15m 13s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4649c17dc4e447f4ad8297e43cfafc74 : SUCCESS in 10m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c7d7cabf8bff4504a5f68bea2d8a94c8 : SUCCESS in 6m 10s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/910c057e44094715919539c2a541c687 : SUCCESS in 24m 28s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/714609189beb4396be2ba650d360ddcc : SUCCESS in 21m 06s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/fa245de92c71415f98430616632401c4 : SUCCESS in 11m 24s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/6f16ef7e02ca4784bb813340004fba18 : SUCCESS in 14m 52s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d64b3b2c8002464b80673116f016c5f3 : SUCCESS in 14m 34s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3295af9cdb0847b49edf839bbffffe27 : SUCCESS in 11m 57s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/afad2af6004f48b38db115b3d7053f6c : SUCCESS in 8m 12s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/047dc7cac0da4b929190ddff3d480b23 : SUCCESS in 13m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2af3cf2007164d1dba99d92fe1090b9c : SUCCESS in 6m 19s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/716070f168f74c3387d24b734e1a6460 : FAILURE in 7m 22s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/f74bf3ca98de4d02a329cccd26f549db : SUCCESS in 8m 35s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4aecc2bd35d143958bf08037643e93fc : SUCCESS in 38m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3ba12e16531c494eb261e705e58985a4 : SUCCESS in 55m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/324b2e7b511b435ca26048753bb327d8 : SUCCESS in 28m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5b13e90a05aa446ea68f8d665ce3d6dd : SUCCESS in 13m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7655e26672f434e87aa2904309ada78 : SUCCESS in 4m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2040fe52f1694122ac9614ca754b738c : SUCCESS in 18m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0bfee8d47db347ff86dbac5dd28fdcd7 : SUCCESS in 25m 46s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"07154e667be64799d6b6005da7cef94723cf8a5d","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-09-03 13:36:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"071284f22b65d550d8d9bb0aa8b4147c374ae2f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 13:37:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d72f518e2b7cab3e0044c3ea9f61ebfc5003a828","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 15:06:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/afe42a127d994a9ea769031bc0d8632d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b254901a6d04861ae96ed0b50311849 : SUCCESS in 13m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b522ddd3633c49c2a5962680ed7e886c : SUCCESS in 8m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/46c48789c5a341cc948acc9fdf66b18a : SUCCESS in 1h 08m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a351dbcfb96040fd9fe974c53e678291 : SUCCESS in 1h 08m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/20ccf691e24344198606784a31eee038 : SUCCESS in 1h 10m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/151a94d251714aff9c2ce820db88b746 : SUCCESS in 47m 42s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/596fe2057ecc4315b843ac543d468b23 : SUCCESS in 8m 15s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3d50fd25f494415f8f94b2b7818336a4 : SUCCESS in 29m 36s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/66dcfa5f29dc4dbda0d362e42a3e8670 : SUCCESS in 22m 51s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/b86b27caac644c84b809d3cdc9b1945e : SUCCESS in 23m 59s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/4c35dd6da9f24d70be6e120233f50cf9 : SUCCESS in 18m 08s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/29c13e3ad9894cd190fc714bfda4a172 : SUCCESS in 12m 51s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e93ab559f9a047dbaa05f4472e114ea8 : SUCCESS in 16m 35s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/346733c912ce47aca28802f8d8bc7df3 : SUCCESS in 17m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/40975a55322f405fa11d8f3d8d6fb4e6 : SUCCESS in 7m 12s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/028418ee35594b61ac351dbedd175032 : SUCCESS in 7m 55s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/dc2c964169f643b38fd03bc97dfd535c : SUCCESS in 44m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ef60b12d7461480c80b81222cb68e094 : SUCCESS in 58m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d637489c30a04d088bb4464eb9aaec84 : SUCCESS in 53m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7ffd3f27e8ec4478b5a87c6b6ca27e69 : SUCCESS in 28m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bf36d949124487a8d61990efad4a13e : SUCCESS in 3m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/130b9b4e57c64f479ccdb58c1b1fd6e7 : SUCCESS in 19m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f4a58682f6544d84bf42c679fc50ae2f : SUCCESS in 55m 03s","accounts_in_message":[],"_revision_number":1},{"id":"cb811831a612007bf6d8bef6093750f5cf7a5c64","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 15:06:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4c29e7631b22dbc61c440c006d5fb7b1eb909649","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 15:09:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1d459770a244d75817833cc121b63be\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0b32049a9cd404db1f40971ebf7d2c6 : SUCCESS in 1m 18s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/755e992ac3c14b939cbc8936bd59cc0e : SUCCESS in 1m 23s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"437f98742ebe5623b75185ef02b8e780939bf28a","revisions":{"437f98742ebe5623b75185ef02b8e780939bf28a":{"kind":"REWORK","_number":1,"created":"2026-08-28 21:42:15.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/30/1002830/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/30/1002830/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/1002830/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/1002830/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/1002830/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/30/1002830/1"}}},"commit":{"parents":[{"commit":"4ff57666f016d4054a46dc2dcaa3ca2ad7ed4766","subject":"s3api: add normalize_checksum_algorithm","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ff57666f016d4054a46dc2dcaa3ca2ad7ed4766"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-08-28 21:42:02.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-08-28 21:42:02.000000000","tz":0},"subject":"trivial: Remove some py2 comments and validation","message":"trivial: Remove some py2 comments and validation\n\nChange-Id: I76783b7c85dae1b40fc3abcebc03cd9915df4840\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/437f98742ebe5623b75185ef02b8e780939bf28a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/437f98742ebe5623b75185ef02b8e780939bf28a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
