)]}'
{"id":"openstack%2Fswift~1002834","triplet_id":"openstack%2Fswift~master~Ia31d262dad9bd074a764900cb15b2ba81de268f8","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-09-03 01:04:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia31d262dad9bd074a764900cb15b2ba81de268f8","subject":"Refactor handoff-deletion logic","status":"MERGED","created":"2026-08-28 22:13:35.000000000","updated":"2026-09-03 01:04:52.000000000","submitted":"2026-09-03 01:04:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"1002834","meta_rev_id":"cdb1790e20b5c5a449b400fb4fd791b69734ca9e","_number":1002834,"virtual_id_number":1002834,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-03 01:04:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"all":[{"value":2,"date":"2026-09-02 13:51:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"all":[{"value":1,"date":"2026-09-02 13:51:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-28 22:48:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-28 23:26:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-02 13:51:32.000000000","updated_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"reviewer":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"state":"REVIEWER"}],"messages":[{"id":"a57e54cbb398ecd7d8d9b64a7e189dee0283deff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-08-28 22:13:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"816f5aa6848d669d104e2cea46adc57016c50b32","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-08-28 22:14:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7ea86b2a5e2dea9d19a0d1c42832585b46a32355","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 22:48:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7808ebe4e858470b9178fa9ab6380986\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8e893553d007437882118aefb7930815 : SUCCESS in 32m 28s","accounts_in_message":[],"_revision_number":1},{"id":"15d8538b3ebca29dd987da180c34a93543398a65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 23:26:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5849906deae947eebfbb018633bb2ce0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86de6793f38746d6a3a88b6a01dc4f1f : SUCCESS in 8m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b7d87258501442ddaf67c446a2e6a907 : SUCCESS in 9m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/c367fd0cb8ca4ef08dccfe6ca48d3c1b : SUCCESS in 32m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4c22f2e0f7bd48a7a8468f83cffeeebf : SUCCESS in 35m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/671bf705b60647309e28e85fd87b0cc3 : SUCCESS in 40m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2db984b92b9d4455ade74ad7ee1e4089 : SUCCESS in 1h 02m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/691ced282ee84e33b72a098aff832ce7 : SUCCESS in 15m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a9ba3950cf1f4d1989584b404c1e4b7a : SUCCESS in 14m 52s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/80e6478cc823432c94edefec95e59e65 : SUCCESS in 16m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2f61e0040daf4004b03feff9d9c8ea21 : SUCCESS in 6m 00s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7794b2fccc034c3b8bf7b63636307e56 : SUCCESS in 18m 25s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/50e732fa4d6f4b108d50eaa5fba14166 : SUCCESS in 9m 06s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/291fe82ece6e4fb8a318c3c10315612c : SUCCESS in 27m 00s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/2c69db363c8243df86f2835473750355 : SUCCESS in 20m 41s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/dc1064d266404925970c938d2b7166ee : SUCCESS in 12m 42s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5563c5e0683441e1921e680b2ce24fca : SUCCESS in 8m 15s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1342256dd29d45e69bd2d045da8a3aa3 : SUCCESS in 15m 24s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/b432f7906e464b7887ed02418c605ae3 : SUCCESS in 19m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d2e3fc9fe408487cbe5623a974f20200 : SUCCESS in 9m 30s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9066a5f8126647ff873d2c9c2bdab2d5 : FAILURE in 29m 49s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/a7aa3a3343c64c6cab0235015ba65661 : SUCCESS in 5m 04s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3e929ed1213e491babb4a41a65f7aa48 : FAILURE in 1h 08m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/68889244d25243838c9e0c36a6b8b4ac : SUCCESS in 25m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5670e766c1e24a62b0b8ac635b6849d2 : SUCCESS in 53m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0c39c137715e4f10b8c574e39cf35e4c : SUCCESS in 27m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc00e47b54664a28bce76d2f6980736b : SUCCESS in 2m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/87aaef2cf66f40ee8cceb0cac8093d38 : SUCCESS in 16m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a5be07af64534605b57c7938f90bf1ad : SUCCESS in 43m 52s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"fe1886840fad4014232584a1c827aa0143f0885e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-08-31 22:39:44.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"bcc427739937c6e2d78c329a9936a1315193533d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 23:59:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37b21895ee79421ea5591f61de9d5ccc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acf4889acc714fbd8d9253fab3b51a6a : SUCCESS in 11m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/08ca8dce173b4b2db13e27ea7d7cfcbe : SUCCESS in 9m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/98b84ea6e3d64b36bae791cfb9634466 : SUCCESS in 1h 02m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3a14b221b02c422098830fd406513793 : SUCCESS in 32m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d980b17fc3124f12acaa1716eea82431 : SUCCESS in 36m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5429ab9aaeeb4cc288c3541bafafce5a : SUCCESS in 31m 27s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/eaba4a7506724acbb54c85020814cb37 : SUCCESS in 19m 50s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a1900d721d8b41da8ba6662a43eb4cdf : SUCCESS in 16m 54s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/31346acf4b5640eb9a224ac1e46a17b4 : SUCCESS in 19m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/27982c1825f843e29eeabdc67d566610 : SUCCESS in 7m 22s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4722b23b5f384c378d39c9c5b9e4294f : SUCCESS in 28m 43s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/1149f4a0edd54ec1be318d86c295a9be : SUCCESS in 22m 04s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/07e9d20bc2174519b48e75848227eb7f : SUCCESS in 12m 21s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/e749084455274c09b73d606d655b450d : SUCCESS in 24m 21s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c9c7a18112814111970e33c6763532f5 : SUCCESS in 15m 25s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6bdebae792924a4f8451267a2a80c437 : SUCCESS in 8m 59s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7e227384cbec44c08dbe82f3450c3ede : SUCCESS in 9m 17s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/ca9f46fe0f7c4f6a90acdc3bbc68a87b : SUCCESS in 10m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/55c554ec38d242c6b6c5df184d65490e : SUCCESS in 5m 53s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/922f299f0c20439b898defd89d9d27e4 : FAILURE in 29m 16s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/c8a1f0e0c8404c518cad859c73020480 : SUCCESS in 9m 27s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/14eb6056896e45e19330fd5d25b1c87b : SUCCESS in 1h 05m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/caa408c0d35f40858a39a6ba9b10d4ae : SUCCESS in 29m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b6e8aabdf78944308ed04080e8c0ee4d : SUCCESS in 58m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0c168b3392d745babb98ecf3e27467d1 : SUCCESS in 31m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c058b68c069149b0a59a0e82ed240da3 : SUCCESS in 3m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cfd533b614e24faa8a5e1a350b75e91b : SUCCESS in 20m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0e44b850441a4573bdbd33203aa0f1d1 : SUCCESS in 31m 51s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"3e9458f71163b3b1bba417ddd385cd5cd8cb2515","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 00:03:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec5af733a02c4c9f9f05ee617f4574c2\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/314a1c53268842009c16b92d068b4fdb : SUCCESS in 45m 55s","accounts_in_message":[],"_revision_number":2},{"id":"955e23218d8112696cebe7e0d75939250a680f05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-01 21:43:05.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"750b2c7f502030694373bcf32d598c9b9851839a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 21:45:46.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e00d3f3cae64b81ac1a65473e9a5c99\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b34c379f5f9d47b2bf4837e8e4301202 : NODE_FAILURE Node(set) request bbc6d7296d16424e8238f1a1800770cc failed in 0s","accounts_in_message":[],"_revision_number":3},{"id":"f3a5d8931ec35ce6177c02275af344a2dbe9b2ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 23:12:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/360b69adb55745f7b5132845b1d3cee5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/369e114b39b84e4c9a6d7db86b49e7c6 : SUCCESS in 10m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f8472ae240da48e7b6c36b4c36382db6 : SUCCESS in 8m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/185b360f0f8f4d70b83485b85b864056 : SUCCESS in 33m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0cfabb21c7e44412882377aadaf034a3 : SUCCESS in 1h 01m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9918b5fd54e046d6919c1fdee2115f8b : SUCCESS in 53m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4267daebe5db4cc8bdff4fe05d473916 : SUCCESS in 57m 23s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/170aed1156524b72a8ab81171c3eaff6 : SUCCESS in 16m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a13f26fc75d9469ba1e78d8c724de749 : SUCCESS in 17m 54s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/25b0b8e9c18a42f2bd7860bb8ea4b0f6 : SUCCESS in 18m 47s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6ce868b9974e4853a48f31c2b7be822a : SUCCESS in 4m 10s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f4859831ff2e43b2b2328d787b07b3ae : SUCCESS in 23m 28s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/1579d71e31c74f0c86a556b92a9ef4f2 : SUCCESS in 10m 15s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/db68f14d8bf246dc888f5c15a194cd2d : SUCCESS in 33m 52s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/df98e82d12a143438df583e0ed366573 : SUCCESS in 10m 25s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/009de60a10c7409e929cb134de63d6d4 : SUCCESS in 15m 15s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/89c5186ef82f446789b925e6e97c9031 : SUCCESS in 14m 17s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e1deb34809054b4a98c0bccacced3098 : SUCCESS in 9m 10s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/8683265523ee41d48bcc8cf0273668a5 : SUCCESS in 18m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7274bd700ce145539afb106730118f2d : SUCCESS in 3m 42s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b5e38738be3d4bce82867c694932f080 : FAILURE in 21m 13s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/ebe8473e47934c7cafa9ac917bb5c223 : SUCCESS in 5m 39s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b8545b4303a84d0b8b49724c2366402e : SUCCESS in 53m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ea6e470e2b5b4f7492ade488f1de7bb2 : SUCCESS in 25m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/131942c45ea447e0acaa1af042cc21fb : SUCCESS in 49m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/47519b59381341c8849022bbf9ae9c26 : SUCCESS in 30m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6699f489fd534d1b9d590b7f66f1a09f : SUCCESS in 3m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/448275825185426bbd579d144141e356 : SUCCESS in 25m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a580b28c82e949f6b6210b2cc15acfe6 : SUCCESS in 28m 50s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":3},{"id":"be82f7c6e3b2838ddcdeaf8499e2408549c8a965","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-09-02 13:51:32.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cadeeae00861e19404768e467f25b97317172a9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 13:52:07.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6b3c879333b1753a715cab78d69ccf19e076a748","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 01:04:51.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cabbae82f1054fcd849fe2c4d78957f4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0684bcdd9d3d46bfbe1789b52cff541b : SUCCESS in 5m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd4b26490bbc411ba943689b3df8db04 : SUCCESS in 5m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/c725e234fbc14c8b89bc25cb71038556 : SUCCESS in 41m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/04f200471bea49c5a08166ddb47d5a3d : SUCCESS in 32m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/689fdea79990483bbb83ec1a0acf7ec8 : SUCCESS in 38m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b52df272994f43d3993fb932634e8635 : SUCCESS in 59m 45s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/221c7405d0f14210945b10cf3563f3b7 : SUCCESS in 8m 46s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/41e12c3605d344e680cf8df35d8d091f : SUCCESS in 27m 12s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/140dd46aec544da79548db52ac3f1e38 : SUCCESS in 23m 01s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/54a6c5a7606a4d41b029c0a82f68af94 : SUCCESS in 32m 38s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/6dbf18b57ecb41e0ad4cee02e1d61ed3 : SUCCESS in 17m 53s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9e9382f5c16346c3a8e614b77bf547a2 : SUCCESS in 15m 25s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f1fe43ff3d644f1c9feb1d4da9501ad2 : SUCCESS in 9m 24s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/61b02569c6194755be4ded50024b2b74 : SUCCESS in 17m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b29ac56f29194491a513ae731e510d7e : SUCCESS in 9m 15s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/c530b588501e4818b19ae97268538b28 : SUCCESS in 12m 30s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4147908f34164b898a9243d4fa0ce3d1 : SUCCESS in 45m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/125c2e5bc1424b65ba1087447bc1d948 : SUCCESS in 56m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8265284fbbba4162ac06f494336f66fa : SUCCESS in 24m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/de62c77ce5db46d8bb08b5e855026d87 : SUCCESS in 22m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26095624665d488ca02bcc67fd683cf2 : SUCCESS in 3m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1fa023a4452e4e9487118c9160dbc126 : SUCCESS in 18m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/72fdd913cf244f3ebd60fa84e8c9f217 : SUCCESS in 35m 02s","accounts_in_message":[],"_revision_number":3},{"id":"cdb1790e20b5c5a449b400fb4fd791b69734ca9e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 01:04:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"46f6b973540d1aadd9c2a8c1131bf288a6d23f26","revisions":{"8e60e1643f7be53cc7b9fea22398c98a5991cbd7":{"kind":"REWORK","_number":1,"created":"2026-08-28 22:13:35.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/34/1002834/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/34/1002834/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/1002834/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/1002834/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/1002834/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/34/1002834/1"}}},"commit":{"parents":[{"commit":"544abf2865d3b95c491b9e58602c9ed1d9bfa9a9","subject":"Improve comments about multi-region ssync","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/544abf2865d3b95c491b9e58602c9ed1d9bfa9a9"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-08-28 22:11:23.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-08-28 22:11:23.000000000","tz":0},"subject":"Refactor handoff-deletion logic","message":"Refactor handoff-deletion logic\n\nChange-Id: Ia31d262dad9bd074a764900cb15b2ba81de268f8\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8e60e1643f7be53cc7b9fea22398c98a5991cbd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8e60e1643f7be53cc7b9fea22398c98a5991cbd7"}]},"branch":"refs/heads/master"},"9c2deadabd9018cfcdac3fc00a660b8985aaac67":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-08-31 22:39:44.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/34/1002834/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/34/1002834/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/1002834/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/1002834/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/1002834/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/34/1002834/2"}}},"commit":{"parents":[{"commit":"bfa302318bca97b5f676636db23814bb2095fa32","subject":"Improve comments about multi-region ssync","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bfa302318bca97b5f676636db23814bb2095fa32"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-08-28 22:11:23.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-08-31 22:38:52.000000000","tz":-420},"subject":"Refactor handoff-deletion logic","message":"Refactor handoff-deletion logic\n\nChange-Id: Ia31d262dad9bd074a764900cb15b2ba81de268f8\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9c2deadabd9018cfcdac3fc00a660b8985aaac67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9c2deadabd9018cfcdac3fc00a660b8985aaac67"}]},"branch":"refs/heads/master"},"46f6b973540d1aadd9c2a8c1131bf288a6d23f26":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-09-01 21:43:05.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/34/1002834/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/34/1002834/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/1002834/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/1002834/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/1002834/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/34/1002834/3"}}},"commit":{"parents":[{"commit":"4ff57666f016d4054a46dc2dcaa3ca2ad7ed4766","subject":"s3api: add normalize_checksum_algorithm","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ff57666f016d4054a46dc2dcaa3ca2ad7ed4766"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-08-28 22:11:23.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-09-01 21:42:12.000000000","tz":-420},"subject":"Refactor handoff-deletion logic","message":"Refactor handoff-deletion logic\n\nChange-Id: Ia31d262dad9bd074a764900cb15b2ba81de268f8\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/46f6b973540d1aadd9c2a8c1131bf288a6d23f26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/46f6b973540d1aadd9c2a8c1131bf288a6d23f26"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
