)]}'
{"id":"openstack%2Fswift~1003049","triplet_id":"openstack%2Fswift~master~I7c4e332c3d394789969d52422c7a03f627f8a98d","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"14250":{"account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"last_update":"2026-09-03 15:02:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7c4e332c3d394789969d52422c7a03f627f8a98d","subject":"staticweb: escape web-index in the listings-disabled 404","status":"MERGED","created":"2026-08-31 12:49:58.000000000","updated":"2026-09-03 15:06:11.000000000","submitted":"2026-09-03 15:02:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1003049","meta_rev_id":"9eb67826def290828cb9a3c1f86c03e28b25547c","_number":1003049,"virtual_id_number":1003049,"owner":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-03 15:02:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"all":[{"value":2,"date":"2026-09-03 13:04:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"all":[{"value":1,"date":"2026-09-03 13:04:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"REVIEWER":[{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-31 14:06:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-31 14:48:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-31 16:48:04.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2026-09-03 13:04:22.000000000","updated_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"reviewer":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"state":"REVIEWER"}],"messages":[{"id":"c27f5e493df15b65cec1f37de9036317002232ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2026-08-31 12:49:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cddadd77bccf32f42a869a7ca34180c6cb1697db","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 14:06:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d78a410725642d4862bbefff7ad33fa\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/42e300ff0161420e857f7f305ba3831b : SUCCESS in 1h 11m 42s","accounts_in_message":[],"_revision_number":1},{"id":"b400d7bcb0ff8ad6e80c71e98c4111cf4b722c23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 14:48:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4e7f491e190f4a4e843fd279e4313983\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b452b1cd5374598b51b0054e9d7f420 : SUCCESS in 9m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/13637108d49b4534a82c2e7d85706899 : SUCCESS in 10m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/1c050d751a0c462d90b5cb748dec71b7 : POST_FAILURE in 48m 51s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/023527a48ea9478c91140b72d8b0b072 : FAILURE in 26m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a5e78e9e4775448b989f817ac25e03f7 : RETRY_LIMIT in 9m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8bffc03b7bbd40beb0e74826c6233e0f : POST_FAILURE in 21m 31s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ed01941cd2b34641b738b80249700054 : SUCCESS in 16m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6e84c54870c9480bb129f628944d2fef : SUCCESS in 24m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8b27c2d3c6c94b83a4e0581f0ef77948 : SUCCESS in 16m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/01eb6e1bca104c33a3d82d4483684e27 : SUCCESS in 6m 40s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2d184b40ba0c4091ae32fb0b653e01a2 : SUCCESS in 25m 40s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/24bff18a437e421a8e0b482ee5202730 : SUCCESS in 23m 09s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/848db9b7ea2a46388420543e88b14637 : SUCCESS in 33m 38s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/fb164813fa9d4076ac5696229a726dbb : SUCCESS in 21m 06s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/2bd70631a0ec4814aeda9252794b871b : SUCCESS in 15m 01s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/991ecc4064ac49df9e4c09b7b8e8aaaf : SUCCESS in 15m 03s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/30806b628c1a4588a68db8e9912831bb : SUCCESS in 10m 32s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/e34cf5ed88c34f2dbce1aa6d7dd3b906 : SUCCESS in 20m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bd998449457e42c68f328a05f18773ce : SUCCESS in 7m 36s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6ce05e77d9744b6e84b078970bb11d2f : FAILURE in 32m 27s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/2b9b11e4f141429b951a8c1661616fa3 : SUCCESS in 10m 38s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/169d1500a9ae4ace8b44358c80c4a10f : SUCCESS in 1h 03m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/019c0f40e83643d1adadd72e2d07a160 : SUCCESS in 28m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cf7bb158bac84a7db4b3c0b2371dba2c : SUCCESS in 37m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d7c7665c416e4772b6486780f693849a : SUCCESS in 29m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2034ed2d60164f90bfd354b7067c44ee : SUCCESS in 4m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3d12d4c22faa41feac391314318587a1 : SUCCESS in 22m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/178ea0aacdd244008b5ddd827de4ae5e : SUCCESS in 42m 14s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"eb42fc2bf25430a5ec32391ff46dc3f7ad350d10","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-08-31 16:48:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6717a0193409bd0846f6c31fc41113e73e31a28b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 18:19:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f106561208dc4d9aa649bfff19620b2d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/55aa72282438470a9b8cb1a52ae9293a : SUCCESS in 53m 44s","accounts_in_message":[],"_revision_number":1},{"id":"ca74baa7ca70f0f419f8ec8f1ef1989000c0a94d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 18:44:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/776d31546ba643f183517cedd701e0f8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f82bcce2e149406dbf034f2fbde3f220 : SUCCESS in 10m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a3372f49ef841aca0d8f6a5aa687fc5 : SUCCESS in 8m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/5df26fee77a74f76b6ff90547079cf77 : SUCCESS in 1h 13m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4c16b56c1a4440428f3929def088ea3a : SUCCESS in 1h 06m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a04d50e030d0470db5694e38a6093aa9 : SUCCESS in 51m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ee22220ee2ad4a8ea24f00f71f624498 : SUCCESS in 1h 08m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9c0306e9a080454295517debc2e02105 : SUCCESS in 23m 03s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/302ac76bce5146899c82f987b1c68ce7 : SUCCESS in 11m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d318cfda781c441b842a70cea2ac7ae1 : SUCCESS in 21m 50s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/312f01cf38ea45458f0071b7f01cc442 : SUCCESS in 5m 50s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/951b5cd73a654c3aa3ec57b756e008f2 : SUCCESS in 14m 42s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/3786beeb334747d6b2326534c85958a4 : SUCCESS in 23m 47s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d2465794a24f410c8dd691b6638e2543 : SUCCESS in 21m 21s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/ca128db40e9f451cb1f37c719227ae8a : SUCCESS in 20m 58s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9504739d22224c2c839adde093337ad7 : SUCCESS in 17m 21s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/17d65ff7dc524ffeb586cd6caf4b1b17 : SUCCESS in 14m 54s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3ef62e8b8b5c45c888528765e545e6dc : SUCCESS in 17m 05s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/09babc7f37344ae5a5c5ee59316a909d : SUCCESS in 17m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0ffff1ed18ca48008e5150c969b852bc : SUCCESS in 6m 45s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/aedcb5276ac543d2bd0c9d6880d98196 : FAILURE in 17m 15s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/03fc1c9d30254f48aec2fd410027df91 : SUCCESS in 9m 55s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/267885bfbca64dc9a65f167dcef15369 : SUCCESS in 52m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e9789a51a45c4b9ca64f8c0b4aa20d6f : SUCCESS in 56m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e09de604ca63484aaafa3f06ba7b1f79 : SUCCESS in 36m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/750f0e2e28a04c1883e951a712d93f37 : SUCCESS in 12m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7addccf79a1e46ebad6e322bc51e9da2 : SUCCESS in 4m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c28d98084e05461190e790d1c87ade7c : SUCCESS in 20m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b06fd0ef34f74572ab919e727e14ec6e : SUCCESS in 1h 11m 21s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"e1353d153b7cb16b85c563b695587281aa311eb8","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-09-03 13:04:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"33c3da62aa5acaa2634fb5c5132b7551fb842778","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 13:05:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"40f212dd9112d57bb5f14884ba76f883430da920","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 15:02:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d81843ba00fd4facbd90706f623eedba\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9af93655d75848a4ac6a0b410074d563 : SUCCESS in 16m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ad4a790d181740c192fce06173e91a1c : SUCCESS in 11m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/84b3aecc774f4e988cf14f0b47cd4b3a : SUCCESS in 1h 10m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f809cf20584a47538c26f8de9b9e8179 : SUCCESS in 36m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4185e9cfa6a84ff1a03b456dee48fe5d : SUCCESS in 32m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/24d54d9f0a7c4ed18ea1eab9d942b222 : SUCCESS in 31m 48s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/e9a657959b634e30a17e8191e0a30648 : SUCCESS in 7m 51s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f9ca5e0edf744eaaa33b0b3afee854f8 : SUCCESS in 28m 41s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/dd4be0ae8ef041cf860632e64e08adc2 : SUCCESS in 23m 35s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/04a034eb191245dda61f2cf868194d65 : SUCCESS in 36m 30s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/f48bb51cdad246e49125b8e9bbf17db0 : SUCCESS in 13m 32s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ae48817340cb4ab9b31f6bebe5aaea7a : SUCCESS in 15m 21s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7f5bf6f6751543ca8200f07c9690475b : SUCCESS in 18m 35s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a44845cf76c44f8198b8bcceba550525 : SUCCESS in 12m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c6483fbe83df4ffe84f2a9bd62a3bd40 : SUCCESS in 6m 50s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/67b68b9b84fe4bdbb3326989114f2c55 : SUCCESS in 7m 58s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7d5a4a2a65f44654a0c9bf11d7f47e98 : SUCCESS in 1h 06m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6b7b9c20f578479a849f16f320ac1e1e : SUCCESS in 26m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/60919345d1de457b8eb378df941ce658 : SUCCESS in 56m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e386147d73e54b85a2f1b522d0902a09 : SUCCESS in 25m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f2fa75893814eca970ce4348bb47706 : SUCCESS in 4m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/31720e0de45e42b190bcc40756cdf1e4 : SUCCESS in 21m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1a03556594434f9bbb075988de3b5c66 : SUCCESS in 1h 00m 37s","accounts_in_message":[],"_revision_number":1},{"id":"9b3b7723b0c26e4b1b10a88eeeb3ec5f4ef8cfff","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 15:02:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9eb67826def290828cb9a3c1f86c03e28b25547c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 15:06:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d91009bf41614b6ca7320e87cfb0d7e5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57ae32a9c3b54816b3cc0b3b1f7e8763 : SUCCESS in 1m 15s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/1b49b070470646afa0c9e07c3f2b79a3 : SUCCESS in 1m 02s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"da0c8320d5c88d4fe74e2b784de0367fd4420695","revisions":{"da0c8320d5c88d4fe74e2b784de0367fd4420695":{"kind":"REWORK","_number":1,"created":"2026-08-31 12:49:58.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/49/1003049/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/49/1003049/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/1003049/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/1003049/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/1003049/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/49/1003049/1"}}},"commit":{"parents":[{"commit":"4ff57666f016d4054a46dc2dcaa3ca2ad7ed4766","subject":"s3api: add normalize_checksum_algorithm","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ff57666f016d4054a46dc2dcaa3ca2ad7ed4766"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2026-08-31 12:49:53.000000000","tz":120},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2026-08-31 12:49:53.000000000","tz":120},"subject":"staticweb: escape web-index in the listings-disabled 404","message":"staticweb: escape web-index in the listings-disabled 404\n\nThe \"Index File Not Found\" page echoes X-Container-Meta-Web-Index so\nowners can see what they configured. The path in the same response is\nalready escaped; escape the index name too so markup in that value is\nshown as text.\n\nAssisted-By: Cursor:grok-4.6\nChange-Id: I7c4e332c3d394789969d52422c7a03f627f8a98d\nSigned-off-by: Grzegorz Grasza \u003cxek@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/da0c8320d5c88d4fe74e2b784de0367fd4420695"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/da0c8320d5c88d4fe74e2b784de0367fd4420695"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
