)]}'
{"id":"openstack%2Fswift~1003144","triplet_id":"openstack%2Fswift~master~I4d63c9fab73a35af615e31eec6aa8745a77dbc59","project":"openstack/swift","branch":"master","attention_set":{"36606":{"account":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"last_update":"2026-09-17 15:59:21.000000000","reason":"\u003cGERRIT_ACCOUNT_35790\u003e replied on the change","reason_account":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"}}},"removed_from_attention_set":{"38368":{"account":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"last_update":"2026-09-03 21:30:53.000000000","reason":"\u003cGERRIT_ACCOUNT_38368\u003e replied on the change","reason_account":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"}},"35790":{"account":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"last_update":"2026-09-17 15:59:21.000000000","reason":"\u003cGERRIT_ACCOUNT_35790\u003e replied on the change","reason_account":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"}}},"hashtags":[],"change_id":"I4d63c9fab73a35af615e31eec6aa8745a77dbc59","subject":"update user label character restrictions","status":"NEW","created":"2026-08-31 20:24:28.000000000","updated":"2026-09-17 15:59:21.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c2d4848cc2b8bda1e6a03eb9283e08de894a80f3","_number":1003144,"virtual_id_number":1003144,"owner":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-03 15:45:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"all":[{"value":1,"date":"2026-09-03 21:30:53.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-31 21:37:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-31 22:23:44.000000000","updated_by":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"reviewer":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"state":"REVIEWER"},{"updated":"2026-09-16 21:29:56.000000000","updated_by":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"reviewer":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"state":"CC"}],"messages":[{"id":"7dd725a0c4c9dfd217653e276c37318b45cf9096","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-08-31 20:24:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b31f7730293190728bad0ade0cbaf7960b4d882a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 21:37:16.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0b9c86761fb34fb1b7d1a0dd6256946f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fe9308e52074d17b93112ac53f580d8 : SUCCESS in 11m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1719f80f342847c7a646646f6159a6a1 : SUCCESS in 9m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/aba27784cb3141eba25809db8c2818af : SUCCESS in 1h 07m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4b7ac30a79224e1f9f28e016c723d22a : SUCCESS in 33m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/90e79eba0d704b539368fb1aa96879f6 : SUCCESS in 53m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/81881ed4939e4c31b72c8301b424ff9e : SUCCESS in 1h 03m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0bd81ae2b8054644aaeddec6998ff93f : SUCCESS in 19m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6caf8b92557f48879abe419df308d13e : SUCCESS in 23m 13s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dc2724ba025b418ebdfaa98f940dbbdf : SUCCESS in 21m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9ba3007201fd4f32aa30d8ecfa99e584 : SUCCESS in 6m 57s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/675d1a50b2824e948b45552277ecda58 : SUCCESS in 24m 43s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/26eacaea1574416c94ca110aa53b5200 : SUCCESS in 23m 33s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/267f850912804f0ea1aa01f681950279 : SUCCESS in 32m 25s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/02274ac3ebda4b71b6ee1747bacd3ef2 : SUCCESS in 21m 04s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f1e19f42a6984ba98419bf32c9c4fcd1 : SUCCESS in 9m 02s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/cff3600dec73479f9e5e1f8865f68b80 : SUCCESS in 17m 23s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/62acb7bd517e408e8ca3b262287e1a6c : SUCCESS in 9m 23s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/2198bd170bf243f7b8fac81d355f4fda : SUCCESS in 17m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/46154b35263848c08ee46f6dc3cff22c : SUCCESS in 6m 01s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9e1e02a2a3f540bf81507a4e80c9db52 : FAILURE in 25m 50s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/d35b0cf7f33e46ba811c9e8b1a69259a : SUCCESS in 9m 42s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a0bd512ad1dc4a11a127dfddeb2860df : SUCCESS in 55m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f8766df872964b399b6111365ef8b633 : SUCCESS in 1h 00m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4d3b1009fe65421f98ca8735c30fe0c0 : SUCCESS in 56m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bb738232f9fb4767aa5735b6978e717a : SUCCESS in 11m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57b52881d22b42e48362f8fcdb8d6ecf : FAILURE in 4m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0f2828cacce44e29bbddfcab72de7459 : SUCCESS in 9m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/279beb1792614b24b7ceee2f4b637623 : SUCCESS in 26m 58s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"f98944632e9bdd3276e7c60ab4e3fe5bdac99964","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 21:53:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bec6b837db3c41dba30db72642acb2ad\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2528814ced404f0a8b94ca76970e4fb9 : SUCCESS in 1h 26m 00s","accounts_in_message":[],"_revision_number":1},{"id":"e461fc0f4f432bb32c87da230c9fe6c6755c94f7","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2026-08-31 22:23:44.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"986c0bbb22b7d4034e7e3422560a47c20c8e0170","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-09-03 14:33:03.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"fd5a9ad4bee17021ea9ef55760dc615f31e2067c","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-09-03 14:33:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3361f21f7b905172c7a4de0ae7fee6816268e9e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-09-03 14:34:11.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"eae30ae13e7c258e9d8658003c537f777d91e675","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 15:35:55.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4c58a22d3474e7892917afba1185b06\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f98bc699105b42b6a4502b6f13047a99 : SUCCESS in 57m 15s","accounts_in_message":[],"_revision_number":3},{"id":"d2d056e00fb651a7ef84865e0be4367528b0e63e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 15:45:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e92de8d64fa4b6fa8cf90e08dfa63ac\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e4afe2e372b455999ba59043b983fbc : SUCCESS in 12m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0712db294bbe451a832879953d99319d : SUCCESS in 5m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/b10dfd634ce44e28b284a372ca754284 : SUCCESS in 54m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f174be2c4d8b413b930cdc50092796db : SUCCESS in 1h 02m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0aa53c2b6c5248bebf7351a0d13b0342 : SUCCESS in 1h 02m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c76aa3f20d1f480f97252fec96ef547f : SUCCESS in 56m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/11cd7adcdd67440c9e6250474125ccfd : SUCCESS in 22m 21s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fd8a045c1aae45cb8be14b306d9d7062 : SUCCESS in 19m 38s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fb5484cf39cf425aaa30ef78410e9189 : SUCCESS in 24m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/951ca8fa1a914fd791d40e1bc3cbbb9d : SUCCESS in 7m 28s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/24d89467a8ec40609552e62daa7b6904 : SUCCESS in 31m 25s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b37976feb05045858b024e4885174bfe : SUCCESS in 19m 04s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/69a8dc31a604428091f43efe68be1e55 : SUCCESS in 38m 40s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/64fa63aa2c354a1585f0bccac1724b75 : SUCCESS in 25m 03s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/bd7bdcbc40d248799254246b8852535c : SUCCESS in 14m 18s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/305eb81c1b974394bef0bb452c43fbff : SUCCESS in 14m 59s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/eff92929c02545da8653b8b72bb4130a : SUCCESS in 14m 51s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/f10ebc487a0f47618e8fddb158df8cf1 : SUCCESS in 18m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ed162ea4cb7247338138b1778cf263a9 : SUCCESS in 6m 35s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ec171f06e948406eaecc99660fe42c51 : FAILURE in 33m 53s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/4b1cdffb3fb74a6e820a4a1a8a3f2126 : SUCCESS in 8m 37s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/25237241074f4a649d579816c86bceeb : SUCCESS in 38m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c6d99843f27047dfbc0eae22570ead4a : SUCCESS in 53m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d2863d94c17c4fbb905ba4dacb233046 : SUCCESS in 55m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/023116ae6a2d4ea7983893446616ec6f : SUCCESS in 14m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3fbb80400ef4dbea1e0dfe30951c239 : SUCCESS in 5m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d4122497261e4bbf9e18cb887fd4e175 : SUCCESS in 21m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cfbf307f95e846daa091edf27c0fd5a4 : SUCCESS in 48m 19s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":3},{"id":"944c5cfdd53651dfa1aa5da9cc4e1dfffca90582","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2026-09-03 21:30:53.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5c0b5325bf73a988893e36292199bf2caee1fb14","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2026-09-16 21:29:56.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c2d4848cc2b8bda1e6a03eb9283e08de894a80f3","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-09-17 15:59:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1949b12f9f5702f27c89b703816bb6792a3e3bb9","revisions":{"2ec7023b00fa4ec9d062578caf93f2b62ffac5ef":{"kind":"REWORK","_number":1,"created":"2026-08-31 20:24:28.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/44/1003144/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/44/1003144/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/1003144/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/1003144/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/1003144/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/44/1003144/1"}}},"commit":{"parents":[{"commit":"4ff57666f016d4054a46dc2dcaa3ca2ad7ed4766","subject":"s3api: add normalize_checksum_algorithm","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ff57666f016d4054a46dc2dcaa3ca2ad7ed4766"}]}],"author":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-08-31 19:46:29.000000000","tz":-300},"committer":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-08-31 20:22:06.000000000","tz":-300},"subject":"update user label character restrictions","message":"update user label character restrictions\n\ntags must be case inssensetive\ntags must start with a letter\na trailing underscore is removed\ncontiguous underscores are reduced to a single underscore\n\nChange-Id: I4d63c9fab73a35af615e31eec6aa8745a77dbc59\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2ec7023b00fa4ec9d062578caf93f2b62ffac5ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2ec7023b00fa4ec9d062578caf93f2b62ffac5ef"}]},"branch":"refs/heads/master"},"41007aec08816dcdbdcaa9637b478a00d3663533":{"kind":"REWORK","_number":2,"created":"2026-09-03 14:33:03.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/44/1003144/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/44/1003144/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/1003144/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/1003144/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/1003144/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/44/1003144/2"}}},"commit":{"parents":[{"commit":"4ff57666f016d4054a46dc2dcaa3ca2ad7ed4766","subject":"s3api: add normalize_checksum_algorithm","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ff57666f016d4054a46dc2dcaa3ca2ad7ed4766"}]}],"author":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-08-31 19:46:29.000000000","tz":-300},"committer":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-09-03 14:32:31.000000000","tz":-300},"subject":"update user label character restrictions","message":"update user label character restrictions\n\ntags must be case inssensetive\ntags must start with a letter\na trailing underscore is removed\ncontiguous underscores are reduced to a single underscore\n\nChange-Id: I4d63c9fab73a35af615e31eec6aa8745a77dbc59\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/41007aec08816dcdbdcaa9637b478a00d3663533"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/41007aec08816dcdbdcaa9637b478a00d3663533"}]},"branch":"refs/heads/master"},"1949b12f9f5702f27c89b703816bb6792a3e3bb9":{"kind":"REWORK","_number":3,"created":"2026-09-03 14:34:11.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/44/1003144/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/44/1003144/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/1003144/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/1003144/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/1003144/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/44/1003144/3"}}},"commit":{"parents":[{"commit":"4ff57666f016d4054a46dc2dcaa3ca2ad7ed4766","subject":"s3api: add normalize_checksum_algorithm","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ff57666f016d4054a46dc2dcaa3ca2ad7ed4766"}]}],"author":{"name":"shreeyad","email":"shreeyad@nvidia.com","date":"2026-08-31 19:46:29.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2026-09-03 14:34:11.000000000","tz":0},"subject":"update user label character restrictions","message":"update user label character restrictions\n\ntags must be case inssensetive\ntags must start with a letter\na trailing underscore is removed\ncontiguous underscores are reduced to a single underscore\n\nChange-Id: I4d63c9fab73a35af615e31eec6aa8745a77dbc59\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1949b12f9f5702f27c89b703816bb6792a3e3bb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1949b12f9f5702f27c89b703816bb6792a3e3bb9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
