)]}'
{"id":"openstack%2Fswift~1003172","triplet_id":"openstack%2Fswift~master~Ie16ac263ef79a1d751c36f70d560923146ac75b9","project":"openstack/swift","branch":"master","attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-08-31 23:23:43.000000000","reason":"\u003cGERRIT_ACCOUNT_38368\u003e replied on the change","reason_account":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ie16ac263ef79a1d751c36f70d560923146ac75b9","subject":"s3api: Respond to ListMultipartUploads even when MPUs disabled","status":"NEW","created":"2026-08-31 22:18:32.000000000","updated":"2026-08-31 23:45:04.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"762d45a2ad6400ba1543e55663202736bdadcef7","_number":1003172,"virtual_id_number":1003172,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-31 23:45:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-31 23:17:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-31 23:23:43.000000000","updated_by":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"reviewer":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"state":"CC"},{"updated":"2026-08-31 23:45:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"79d1186cf798db4522a32a3dabe91e528f7e661b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-08-31 22:18:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c56b20f8a9ba196c70ed4c04667f27dd5431dd2d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 23:17:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef118a872d194d07a375332eeca4da24\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3d2968588326423896894ce9bcb85040 : SUCCESS in 47m 43s","accounts_in_message":[],"_revision_number":1},{"id":"9fcf39c7c5081909f8620e324fe86ef2aafb83b1","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2026-08-31 23:23:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"762d45a2ad6400ba1543e55663202736bdadcef7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 23:45:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e84cab4eccd54238add3218e9283da73\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e1a0bf4f7d94d7c8a8dd456fe4222db : SUCCESS in 10m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca2e74589e9d41018d0fe14c74d95d31 : SUCCESS in 8m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/95d00e67dbb9469c9f2a83e06cc355a0 : SUCCESS in 1h 11m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fbae7487660640d6899655b3f34f173c : SUCCESS in 1h 21m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b1ba9e6f1fb347719c9242735ac24ab4 : SUCCESS in 51m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/684120a21b0e48aeaf30bcc089617e72 : SUCCESS in 51m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3ac4326838954ad8a24505ae6ffc95e0 : SUCCESS in 20m 33s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/71df74d961e14201b7d61bf5383b5067 : SUCCESS in 20m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/abd60c9431644bfc912ee3b1aa8e8ad1 : SUCCESS in 22m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/617ff00fb8ba43148130c1a4b1ae1095 : SUCCESS in 6m 04s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7489bd1ef525417299c967bc65e7a6df : SUCCESS in 17m 47s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/a0dcaafaa4144f318268c49d2cfba800 : SUCCESS in 17m 42s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/194944a97db7406b9f7199e7ae9f4ea3 : SUCCESS in 25m 30s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/b19f02febd594c019f940efb491c8486 : SUCCESS in 9m 31s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/2b23af11b4f3472296e2995e858a093e : SUCCESS in 13m 17s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c4e2b88afff04200bce2035a162b349e : SUCCESS in 14m 54s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/85a0f2f5a4464571b12b6a7c706bad6e : SUCCESS in 16m 57s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/14fd0960e867412c990d605bd749888b : SUCCESS in 17m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/665203d4bec64da792fa2449ccb00933 : SUCCESS in 6m 46s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0dafe0662c6848499cd5e24e0dcab2af : FAILURE in 17m 07s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/ec52281d5dda4f07884e99df26b1abd3 : SUCCESS in 8m 45s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/123fe88a372d48dba19621a87009a50b : SUCCESS in 39m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/66c9bb38384244a7a3dd20d54038f803 : SUCCESS in 26m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8f0588896b7d46ceae35d12f8dc3ce84 : SUCCESS in 26m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af3cbeffdd0a482486eb3a68a2090ed8 : SUCCESS in 26m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bb9d8222b0145dc879813ee980b80e0 : SUCCESS in 2m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6d50078fd03c4fe7a565e57f32085e3e : SUCCESS in 17m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/da64352e11ad406c92d662816ccd73bb : SUCCESS in 47m 31s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0f60d51fcc2958ecddf779875237538b52335118","revisions":{"0f60d51fcc2958ecddf779875237538b52335118":{"kind":"REWORK","_number":1,"created":"2026-08-31 22:18:32.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/72/1003172/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/1003172/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/1003172/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/1003172/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/1003172/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/1003172/1"}}},"commit":{"parents":[{"commit":"4ff57666f016d4054a46dc2dcaa3ca2ad7ed4766","subject":"s3api: add normalize_checksum_algorithm","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ff57666f016d4054a46dc2dcaa3ca2ad7ed4766"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-08-31 22:12:36.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-08-31 22:12:40.000000000","tz":-420},"subject":"s3api: Respond to ListMultipartUploads even when MPUs disabled","message":"s3api: Respond to ListMultipartUploads even when MPUs disabled\n\nPreviously, we would 501 on any MPU-related parameter being present, but\nthat makes it harder for users to write generic clean-up methods.\n\nNow, move the 501 logic to the MPU controller, like we have for tagging\nand object-lock. This allows us to respond (with an empty list of uploads)\nto ListMultipartUploads when allow_multipart_uploads is disabled.\n\nChange-Id: Ie16ac263ef79a1d751c36f70d560923146ac75b9\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0f60d51fcc2958ecddf779875237538b52335118"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0f60d51fcc2958ecddf779875237538b52335118"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
