)]}'
{"id":"openstack%2Fswift~1003305","triplet_id":"openstack%2Fswift~master~I0f06270537013d658cf31bec5859bed2d1794f8f","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-09-03 01:05:27.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-09-01 20:40:02.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I0f06270537013d658cf31bec5859bed2d1794f8f","subject":"Fix typo","status":"MERGED","created":"2026-09-01 16:23:23.000000000","updated":"2026-09-03 01:09:41.000000000","submitted":"2026-09-03 01:05:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1003305","meta_rev_id":"7bd18b9316fa0dce458ae503ff2141fa9c1518a7","_number":1003305,"virtual_id_number":1003305,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-03 01:05:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-01 20:40:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-09-01 20:40:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-01 16:25:09.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2026-09-01 17:01:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-01 17:32:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5ac925754a0aaa93ab1cbca6c53f2882f58b1e4d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-01 16:23:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b497fd936c315d61ddd24d83c70829cd5370c757","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-01 16:25:02.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"85be151ec3934a78e673113c639344fee5d2b073","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-01 16:25:09.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"6aa513b2ec13325717e89e93310b1a17890520f0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 17:01:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/414e38ff27fc49188f2f973113b68f6a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/dc7d7153c1e449b1bad8ebba5a277d4f : SUCCESS in 33m 15s","accounts_in_message":[],"_revision_number":2},{"id":"94d0e3bdbb91722d54f0bbd3537c1d2c97960d05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 17:32:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0b2379c5c3943b7aa8283bb630cbaef\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/264e506be17a406caa485dfda1a4f3f7 : SUCCESS in 5m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2fbe48809e75462795dcf0eb5bf59e55 : SUCCESS in 5m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e0016b6f3ecc414a964bf516dfc92a0a : SUCCESS in 20m 00s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f7640c8e0508435e9900b844880e3b2c : SUCCESS in 11m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7661d31af4d74717b3c055cb524d3aca : SUCCESS in 23m 11s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/df641a5867e54fcc870fb5f0c1adeae8 : SUCCESS in 5m 43s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8ef8fd0866a844669c8c14744e45de10 : SUCCESS in 24m 49s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/1d954ee06be14bedb9199e54d8096ca1 : SUCCESS in 17m 51s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/2e46fbfb3d0d4645a2c20f0aaa2cfae5 : SUCCESS in 22m 46s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/2e4b9f94291e40e9ad117f60da2e6657 : SUCCESS in 19m 54s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/fb5340b71ca040f1aac503f3ef06b513 : SUCCESS in 15m 58s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4fa7ba88f30e40139b980877e19fc85c : SUCCESS in 13m 54s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f74e59e1ec4b4c329c065161269fb60f : SUCCESS in 15m 47s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/b0a77ddfe0514995a3a7cb33f9e39790 : SUCCESS in 9m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6b3aa731ffda40d99507efa295a2adf6 : SUCCESS in 44m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5bfa6c8018d64e389fcd5b24b84fcfcd : SUCCESS in 28m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9050c7f7d7844cfd9c6232a5aa43ffeb : SUCCESS in 32m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee8011a0cefa4a9fb5a4ca533d0aa916 : SUCCESS in 4m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b1ff170e1041417cb9f8429ead63a094 : SUCCESS in 9m 15s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"173795f9b8aa33e0c7217604224e2e2f1b122826","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-01 19:17:54.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"ce8454112668026ea637b4f79f1ed4d29704ab81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-01 19:57:18.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"acb63023d4dfdf3ba6149d3458ba30f001e8df00","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-01 20:40:02.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6576fa5d8284287494833fa40835b1b27c3b1f40","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 20:40:07.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15d6c71b31ca446dbacad110caed9a43\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2d68420ff6d947e18cab2efd569f1eb9 : SUCCESS in 36m 34s","accounts_in_message":[],"_revision_number":4},{"id":"e2df96d63baf5d8d32fe7ab4e24635f949c39478","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 21:10:45.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c031a70251c8445cbc54ca9bff5aab5e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/058d6423de6e4d60a51b70887ab640e6 : SUCCESS in 10m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b97957adeefa47728bf39cc490d64e94 : SUCCESS in 10m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/6e2df354423e423dae29e4c60506683a : SUCCESS in 32m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0d4f981de18742dd87c6bf1245c5f74c : SUCCESS in 53m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3edd72785594472ead4b51788b10117f : SUCCESS in 57m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/51605d59b1b14f319e0faef3666b3149 : SUCCESS in 1h 07m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bbe65617df864fd8a1e7e883b1b94c82 : SUCCESS in 12m 21s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/64d607f2981d4da1842d05fde69f07e0 : SUCCESS in 18m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f307e13fca3a4ac2914fcd266169cdfe : SUCCESS in 23m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/271eb7e25f31455582c374a0a16a9ca3 : SUCCESS in 7m 10s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1cb3a78b2a3e45a786a7a17ba97c697a : SUCCESS in 26m 21s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/915d18b0dbab49f0a2f7742f83482b1d : SUCCESS in 20m 59s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/02420f24be0247099da3bfa4a1b497b8 : SUCCESS in 32m 01s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/d69cae5ab2644c5b977fecec5c0eb0ca : SUCCESS in 19m 04s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/bb9d7be0bfbc4100a61895942cf8d217 : SUCCESS in 14m 43s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1804218f6fa442889ad5933403158771 : SUCCESS in 9m 07s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1eebe496865a421481db17343f964d74 : SUCCESS in 17m 58s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/4f4d415a85f04faa8992015f3ca58f04 : SUCCESS in 18m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/50239207016443c0a46ccb8ca0a1416d : SUCCESS in 8m 27s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0dc0afd383134392bf0a98c0b7fc79bd : FAILURE in 28m 24s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/82cdab37fa8f4b03be2b09f9a4517563 : SUCCESS in 5m 46s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0deedc4fe2984ab5b9a7c16328a8c3ed : SUCCESS in 44m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e68b11de0ef04c8cb50eaafc81264938 : SUCCESS in 31m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/de620d101fae49b09d0a6d1c168fbf93 : SUCCESS in 31m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/43eb925dc26a4a8a9dbb13ba24490adb : SUCCESS in 12m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35c1bffd6e5140d8afb5557ca0b23a2e : SUCCESS in 4m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/826be6f213dc46819e26ac1a6db6a547 : SUCCESS in 18m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/704f4858e1b740dc9c6bc9a2d8a6e3d0 : SUCCESS in 44m 55s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":4},{"id":"93bd37bcc5a243ecc6d2372985fccf709aab690d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 21:11:17.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"257bfeb376c0ee1000f64fc2bba60934542a2f77","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 16:55:12.000000000","message":"Patch Set 4:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"8e94945557c0d4003f4960305baf776b9198c966","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 01:05:27.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a573eb180c2441ebb3083387d8d6ce08\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/416cff598c7d479dabbbdf5b2dfea046 : SUCCESS in 10m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/16f13bfb3c614bd4b75e68a38795e450 : SUCCESS in 10m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/58e34c7c650844eab23ddf2c4d09a625 : SUCCESS in 1h 07m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/57cd9049e9d545d2b82176f09e6dbfc6 : SUCCESS in 1h 10m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/716281c20ff1483484106ac5ec16eb8c : SUCCESS in 41m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/712015a6f3704a298ea5b1d29f7c73f3 : SUCCESS in 1h 00m 25s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/c484db2ae8874ccf8328443dc5d99385 : SUCCESS in 7m 48s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c1c46dc457b241b09be56af67b5a9270 : SUCCESS in 19m 26s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/7024d49269e34ea19fd5fda5f9901851 : SUCCESS in 19m 47s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/2aa0edbc10364639892f52ed354d33f3 : SUCCESS in 13m 00s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/8a3d4882a99046b5926c42a4ce5f353a : SUCCESS in 10m 28s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b418ee5fe8454a448ac3d627cb84fafb : SUCCESS in 15m 04s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c179044be5c84290b21756c28b78d0da : SUCCESS in 15m 26s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/57c125d035524b4998b506ee47057574 : SUCCESS in 16m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/29cf65e3a1eb4f59926f4a6b89c0a0b3 : SUCCESS in 4m 18s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/4fa18bdd832b4d748b6c7c417c65c4b6 : SUCCESS in 7m 30s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f4c6880bee01416988ef1edd9679c186 : SUCCESS in 1h 03m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7bb585aa021643feb1627aa616e72990 : SUCCESS in 41m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c5d081e90fe149c48737e29c91d89150 : SUCCESS in 37m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca9006abe9984895952830fcc12a8037 : SUCCESS in 29m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48dd52f266ec4546b4027ae68cfa304f : SUCCESS in 3m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1e55bc871d33417b817717df48baa558 : SUCCESS in 14m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/671ac35d7aa942b58d625c5963e3077a : SUCCESS in 45m 30s","accounts_in_message":[],"_revision_number":4},{"id":"19cdc7cfaeb98a301ad28f794a205c2a6922ba1d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 01:05:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"7bd18b9316fa0dce458ae503ff2141fa9c1518a7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 01:09:41.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d94e1070606f405a8035c2e8a2a5551e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8b2e2b230aa4bcf9a0f8476ccae5dd1 : SUCCESS in 48s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/100fff9dce9745f1a0667ae98bd404c9 : SUCCESS in 43s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"fc162b9edba512bfef5495a6de0ff0b6da0666c1","revisions":{"26c576e8e45872d46025501ddd276d47b4d01d49":{"kind":"REWORK","_number":1,"created":"2026-09-01 16:23:23.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/05/1003305/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/1003305/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/1003305/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/1003305/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/1003305/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/1003305/1"}}},"commit":{"parents":[{"commit":"4ff57666f016d4054a46dc2dcaa3ca2ad7ed4766","subject":"s3api: add normalize_checksum_algorithm","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ff57666f016d4054a46dc2dcaa3ca2ad7ed4766"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 16:23:23.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 16:23:23.000000000","tz":0},"subject":"Fix typo","message":"Fix typo\n\nChange-Id: I0f06270537013d658cf31bec5859bed2d1794f8f\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/26c576e8e45872d46025501ddd276d47b4d01d49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/26c576e8e45872d46025501ddd276d47b4d01d49"}]},"branch":"refs/heads/master","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"e544acc11348eafcd1e9181e7218addf54089018":{"kind":"REWORK","_number":2,"created":"2026-09-01 16:25:02.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/05/1003305/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/1003305/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/1003305/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/1003305/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/1003305/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/1003305/2"}}},"commit":{"parents":[{"commit":"4ff57666f016d4054a46dc2dcaa3ca2ad7ed4766","subject":"s3api: add normalize_checksum_algorithm","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ff57666f016d4054a46dc2dcaa3ca2ad7ed4766"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 16:23:23.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 16:25:02.000000000","tz":0},"subject":"Fix typo","message":"Fix typo\n\nChange-Id: I0f06270537013d658cf31bec5859bed2d1794f8f\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e544acc11348eafcd1e9181e7218addf54089018"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e544acc11348eafcd1e9181e7218addf54089018"}]},"branch":"refs/heads/master"},"55cc7e1ca72713677f5026e6a063a12cdd61dbef":{"kind":"REWORK","_number":3,"created":"2026-09-01 19:17:54.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/05/1003305/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/1003305/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/1003305/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/1003305/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/1003305/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/1003305/3"}}},"commit":{"parents":[{"commit":"4ff57666f016d4054a46dc2dcaa3ca2ad7ed4766","subject":"s3api: add normalize_checksum_algorithm","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ff57666f016d4054a46dc2dcaa3ca2ad7ed4766"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 16:23:23.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 19:17:54.000000000","tz":0},"subject":"Fix typo","message":"Fix typo\n\nChange-Id: I0f06270537013d658cf31bec5859bed2d1794f8f\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/55cc7e1ca72713677f5026e6a063a12cdd61dbef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/55cc7e1ca72713677f5026e6a063a12cdd61dbef"}]},"branch":"refs/heads/master"},"fc162b9edba512bfef5495a6de0ff0b6da0666c1":{"kind":"REWORK","_number":4,"created":"2026-09-01 19:57:18.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/05/1003305/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/1003305/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/1003305/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/1003305/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/1003305/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/1003305/4"}}},"commit":{"parents":[{"commit":"4ff57666f016d4054a46dc2dcaa3ca2ad7ed4766","subject":"s3api: add normalize_checksum_algorithm","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ff57666f016d4054a46dc2dcaa3ca2ad7ed4766"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 16:23:23.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 19:57:18.000000000","tz":0},"subject":"Fix typo","message":"Fix typo\n\nChange-Id: I0f06270537013d658cf31bec5859bed2d1794f8f\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fc162b9edba512bfef5495a6de0ff0b6da0666c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fc162b9edba512bfef5495a6de0ff0b6da0666c1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
