)]}'
{"id":"openstack%2Fswift~1003387","triplet_id":"openstack%2Fswift~master~I455acae764327af7aae24170dad7a8445fcee2d4","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"6968":{"account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"last_update":"2026-09-11 07:28:51.000000000","reason":"\u003cGERRIT_ACCOUNT_6968\u003e replied on the change","reason_account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2026-09-10 16:13:12.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-09-11 08:33:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I455acae764327af7aae24170dad7a8445fcee2d4","subject":"replicator: Quarantine suffixes we cannot delete","status":"MERGED","created":"2026-09-01 21:41:42.000000000","updated":"2026-09-11 08:35:12.000000000","submitted":"2026-09-11 08:33:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":15,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"1003387","meta_rev_id":"e06588c375a3c32306479ce1fc767b27752c0b42","_number":1003387,"virtual_id_number":1003387,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-11 08:33:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"all":[{"value":2,"date":"2026-09-11 07:28:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":2,"date":"2026-09-10 16:13:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-09-10 23:28:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"all":[{"value":1,"date":"2026-09-11 07:28:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-01 22:41:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-01 22:51:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-02 14:50:21.000000000","updated_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"reviewer":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"state":"REVIEWER"},{"updated":"2026-09-02 21:15:28.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2026-09-10 23:28:51.000000000","updated_by":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"reviewer":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"state":"REVIEWER"}],"messages":[{"id":"4f0090b1e08a4bf241aaebd94fbc5e3e4d9f0d91","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-01 21:41:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e0a358691e76912069db2f916cd1f0bf5213e2a8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 22:41:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abca8f1139e9415984726b46fc77952d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/07ccdeaf75b541179cd8dbc8e8b19c5b : SUCCESS in 57m 18s","accounts_in_message":[],"_revision_number":1},{"id":"b2e282217de9ee874ec121c2f459de1d0593108c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 22:51:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55f7bed3d3d344fc8118edf45592bd8e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2066760e5dc41aba16b37c2b5293c02 : SUCCESS in 10m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f2aaf46a9c7f467fa12f82cfa99e2163 : SUCCESS in 9m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/a0250be1fe524997adef594aab6b4a98 : SUCCESS in 1h 04m 29s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ed5aea4fa0b34e5293f54fab65e7fc15 : SUCCESS in 1h 04m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/71fa961475be4f72b49f61227779cd61 : SUCCESS in 53m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/867c07eacfd74c2fac1209a7eb312159 : SUCCESS in 1h 01m 28s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b9a90433dacc4338917554b08abb57c3 : SUCCESS in 16m 41s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1f8388d522bd488b9834b90b7a1769c5 : SUCCESS in 21m 42s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4b625d191fbb4482ad3cc2b1977103ef : SUCCESS in 20m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a2da3978828d40b99eaf934e9b8cdad2 : SUCCESS in 6m 37s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f0393977619c4887a9b60efb57ee2009 : SUCCESS in 11m 16s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/e24ee1ad468d418a97d1528a11bb2066 : SUCCESS in 21m 05s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/8c62211176bf4e12beb78d58451bce28 : SUCCESS in 27m 52s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/971b68e89f824c35b500b2d507835639 : SUCCESS in 16m 20s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b8ce04648cdb4b92a84e3cb9dc374e81 : SUCCESS in 14m 09s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/572d02302fce415ba619a56263d1dd5f : SUCCESS in 12m 35s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1bbfda48603041a894ed2b4099f138bf : SUCCESS in 13m 43s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/957f841f8cbe442d9afad99f7ea0871f : SUCCESS in 18m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1abbe1d3e2904c4da3aa9ceff424f1b5 : SUCCESS in 8m 21s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/10330ee8b18148548cee9cd11a8209cc : FAILURE in 30m 33s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/cc1ceea6fca54c7587ec9de6d048ab66 : SUCCESS in 8m 55s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/55a3b029ebcd456a9c775823e1f6e64a : SUCCESS in 58m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/012d1e7f51be4aa5bbda890493706ddb : SUCCESS in 38m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a39b10a0cfa14349a4320159875c4fe3 : SUCCESS in 53m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4e70d81b059f40b8b143cdf6b10ffc0c : SUCCESS in 31m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/301605db064a4f86af56d9f817635b27 : SUCCESS in 4m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fb38b1cc6afe4d8b8fc6d8dc91c403bb : SUCCESS in 25m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/17ed95694c7e4553a6e2c985040a3151 : SUCCESS in 1h 00m 24s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"f6236a1c51222a936fe938447b1e2d1b66632e7a","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-09-02 14:50:21.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b2beb2ffc3d34704e0bdb080e0c44308d48ad301","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-02 15:41:06.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"452ee2c52372b87c56763ea1bea3a4e062eba309","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 19:12:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/875f5df6c9e1447db4d1d9ea779a754a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/600dc51198384c7bb4af6c7abdc920cb : SUCCESS in 1h 11m 11s","accounts_in_message":[],"_revision_number":2},{"id":"c5cbd92052d9ea195d55e456ec1eb8e1e0ddae73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 19:15:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4bc8c35bd2e44f5d8083aca6e10202b4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56c33f977a2843daba8d57aba5ce1844 : SUCCESS in 7m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/40fecd3d447d409094347dba3bdeac1e : SUCCESS in 10m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/7db49ac1a6cb4bd0afe4d47f52a1555e : SUCCESS in 51m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/57516ce59b424cce8b4cdf82ed92273c : ERROR Failed to update project openstack/glance in 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/70ccf3c110404c6285fb8d517252ee08 : SUCCESS in 27m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4f4c4ea1a6174faca3be80fcc705e254 : ERROR Failed to update project openstack/glance in 1s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8037d10648da4546b683a124acfcdba3 : SUCCESS in 16m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c4343ea603ae4165b6e9050a583bc236 : SUCCESS in 11m 42s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f7f1b55663a045de8c2eccf66c4ed5ae : SUCCESS in 18m 33s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d93387ad5bb647529f1cd8357cf09e8c : SUCCESS in 7m 19s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/cf16db69e2054552954af860d444f536 : SUCCESS in 25m 19s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/2e227d8b10d24cb88bb54d72ffc9b63f : SUCCESS in 22m 53s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/42a27159bad6474987185d1288625e36 : SUCCESS in 32m 43s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/f23d170b7f6c466aa49d7ab89157fa84 : SUCCESS in 22m 22s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3cf7996aca79462abfeb327dec05a27a : SUCCESS in 14m 34s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0a2d63b384ae4c99a06cfb21652cc082 : SUCCESS in 12m 46s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/30a6b32b9c2840f59ded385569f7ae5f : SUCCESS in 16m 02s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/d285c96beda444ba87c290bd01ae0f54 : SUCCESS in 17m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/de6e6a7266da43a890719a4af4ffd882 : SUCCESS in 7m 35s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e8c09c445fee4c05b1ce4958d854f7ae : FAILURE in 28m 22s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/9eb8680db6f14456ad9bff261295c2b1 : SUCCESS in 8m 44s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8af321ac333d4fc689497c411003af20 : SUCCESS in 1h 01m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4af2a5896995480fae63c1836ec8ec10 : SUCCESS in 52m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0f11cf9934b84389b5387168b9747001 : SUCCESS in 23m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e194aa6a13c549f2a4bfd07bb4c99f5f : SUCCESS in 27m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b0358b206124c68ada628c149b8b710 : SUCCESS in 4m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e712d207edb54318845a89d8c412f3f5 : SUCCESS in 10m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1d910aa7e73746d3abf42458e3e4ec7e : SUCCESS in 53m 49s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"89407bf95636def737ef4d4b075145107d9258ba","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-02 19:41:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d393cf78bc0f04f5d6e1081f4453076fa1684dbb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 20:19:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9975a5dfc6641ea834bfb25a598efb9\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7a3c476ee94b4c06a80ec381067ae07f : SUCCESS in 34m 55s","accounts_in_message":[],"_revision_number":2},{"id":"8f021977d7429d58fd530478077bb0601521c97a","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-09-02 21:15:28.000000000","message":"Patch Set 2: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"9eb67ec388318e3755e1f2245ad455db91af259c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 00:54:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a792e70782904ba98de41cde0f839048\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a5d7ba94dc642eb925f65162f1ec7e8 : SUCCESS in 10m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0198bd0bbdfd43f0b5f8dadac5afa76c : SUCCESS in 9m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/abeadc18757442e2a97ecc6ac71d8973 : SUCCESS in 1h 03m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/39a8518a4fb94937bfc23eb3273760fb : SUCCESS in 1h 01m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/eb9b686e8f404ca7978bc2b63216528e : SUCCESS in 54m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8ad0b302cefe495285067a8a00eee8a5 : SUCCESS in 1h 00m 54s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/11b07af80ec640b29666d6a50c0041d9 : SUCCESS in 11m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9468065a816a4540b457a2bfff7f75cd : SUCCESS in 23m 08s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/eada63a34e6348e08eaeb549f952803f : SUCCESS in 21m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9aad78dd720f44bca18e0736fd03def3 : SUCCESS in 5m 29s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/46ee9191fcc44bdc9cf4f6bb8387bf05 : SUCCESS in 20m 55s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/eb731f9f573640ee8221159638b24dde : SUCCESS in 25m 03s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/30bb15010f554bf4970a0a5d1bf0f7f0 : SUCCESS in 36m 15s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/2cb95017cd194922aa04e743b0f5efa6 : SUCCESS in 9m 40s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/31c119ac04404d45ab77d1dbb1161a4c : SUCCESS in 17m 18s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/54f0b45661054df0bb21e1376f98aa0d : SUCCESS in 15m 56s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1f6023999b82481fa0472685bb9eed10 : SUCCESS in 9m 04s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/20ca0dd7965e4e3c9fa8f9d590792993 : SUCCESS in 24m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/09f949a3b49647f986047ec483200e56 : SUCCESS in 6m 32s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7959f8668608479aa04e2d3883d8ea3d : FAILURE in 37m 29s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/8fa849044e5242d787a3f9a659dabc3c : SUCCESS in 8m 43s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2d461f5fab064c8ab89225e53425c1eb : SUCCESS in 1h 05m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bdf94bafb8dd490ca087248001babf06 : SUCCESS in 52m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7b064a94ae124724bf951f029bffe5d9 : SUCCESS in 36m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/450974c1deb0402cb425e8c7f7ec4208 : SUCCESS in 29m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c4baf8977bc47f19cefccde15364b43 : SUCCESS in 4m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6fe0a893c4be4e539ba0f1ac0a9f54ce : SUCCESS in 20m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f5b73e4f2482453cb480c4b25fecffb8 : SUCCESS in 59m 37s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"116aeef4cc53637480ed7822f389619a145ecd18","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-03 17:29:43.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"d7fb60f11d21d492ecef6f479553d773ff670d74","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-03 17:31:19.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"d3aefa3b4efc6fb96a0ad49818e2212978c7f447","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 18:11:28.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2bd1bfb2bbb548288ab696a44fe35814\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cb6b334e3a2a4ece8c5939d144d14b01 : SUCCESS in 39m 30s","accounts_in_message":[],"_revision_number":3},{"id":"af297885a572d170381a558281041c7b2b5d8744","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 18:40:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2dc92d47da746a9b7df59f779dc6479\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2219546f37e94478b154cb37a9b3ae2d : SUCCESS in 10m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/97685850bc964d16a87fd18791284b9b : SUCCESS in 9m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/63ecfce44b194ab5aa1a253415c14277 : SUCCESS in 1h 02m 58s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ad62c0bbd0714e078a4dbd5d119c72fe : SUCCESS in 1h 02m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/823d3dfdec554277aba3fa254c810f37 : SUCCESS in 37m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d275a152a85f4155a4c1d065f47bf961 : SUCCESS in 32m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/be95cca2d83a4e8f874158b1bd485295 : SUCCESS in 20m 04s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/144630f2e4b44f85b276da65a3114028 : SUCCESS in 21m 55s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5f26825f7e754228988c3d9736488a58 : SUCCESS in 21m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/17903636b81440269fe281de37aa3bca : SUCCESS in 7m 27s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5c6aac3f709d401396f6256c3803dd1d : SUCCESS in 27m 58s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/20502a863dd94dc4951c34b5be2fcc55 : SUCCESS in 22m 07s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/6d13e3a72f284ad19335c562c36c16dd : SUCCESS in 13m 03s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/8b8f2664a8124005b9519cb962eb22e7 : SUCCESS in 11m 10s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9e434e04e37c4d06a61c53b044954c71 : SUCCESS in 13m 14s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/95058d5d49f94df997fe17af5058f4f3 : SUCCESS in 16m 30s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/bbc9f9be9fcc4e91b48a607b928a7e08 : SUCCESS in 16m 14s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/e769103b2e5145798a48b53733d0073b : SUCCESS in 17m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/56527750b56a43b5b6e503a317f2a786 : SUCCESS in 9m 17s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/15f54ffc65564d2db8fad46a6a56767c : FAILURE in 26m 00s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/bce4e1148c0b4d48a49efb1015350cc0 : SUCCESS in 8m 59s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3b61cf7e8b4b452ba03f914f910ec9ad : SUCCESS in 1h 00m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0f9a081f8faf4d95af1d6117db898eda : SUCCESS in 52m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9cc38c7e97c442fc925459fea632491a : SUCCESS in 57m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dc37b53d245344569d9a55b72551f4e9 : SUCCESS in 24m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2372c01eb0b148c2a3bd86db7266bd75 : SUCCESS in 4m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e113fefefe184de1b4ca356a5a6b8a51 : SUCCESS in 18m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c2e99d59461a4017b85183af59667ba7 : SUCCESS in 52m 03s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":3},{"id":"25de595c47084d9eb3d3863a57438536392c8ffb","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-09-10 16:13:12.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0465f256dacdca07d1df49a368be921eddc4ac94","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-09-10 23:28:51.000000000","message":"Patch Set 3: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"77df01a67471df4f6a4464458856552f7c0e8f13","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-09-11 07:28:51.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"1d5a75f6e526b579220482586a9e41a976a859ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 07:29:45.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"579fa830e5b7c1477c04dd8c71fa9ddec057bb31","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 08:33:57.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53d7a076e2f34902ade40bcb20dda1f8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/361228a834a2427b8f3cf9318e578b09 : SUCCESS in 5m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d582e7547f148abb74561e3e1e869ff : SUCCESS in 10m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/9c96cafbc2084c9da72094106f61c0f6 : SUCCESS in 33m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/027971e03add462299f068d469cd6b1c : SUCCESS in 35m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e99c0e51ca564f2e97974b44efcf23d9 : SUCCESS in 32m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/67bf0fe078da4581ade20dd60aeb2aa8 : SUCCESS in 59m 07s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/91d9fac67a434ee183159616c7bcfb89 : SUCCESS in 5m 11s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/21a9d1dadda243d992b92942f1dab03d : SUCCESS in 20m 25s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/3e124d916497426982ba08865da88710 : SUCCESS in 21m 49s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/62af245135f14e59810e818120ccb2fd : SUCCESS in 32m 27s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/e18595de09a34f70be3a1f43a493bdc2 : SUCCESS in 16m 00s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c0efdf3696e641f89969cb772b7781a4 : SUCCESS in 8m 39s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/2e999e126e0e4b15abab2f52c2a33aae : SUCCESS in 19m 43s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/050d7582369f4b94a144f23a9d0fdf27 : SUCCESS in 17m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bb62dc0a84c04203983b6045f0ff9254 : SUCCESS in 6m 35s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/7dbb43a1213a4713809d7510134afc18 : SUCCESS in 8m 50s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a3fdcae3e80a4eefaff89483bf31a65a : SUCCESS in 40m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/91ef892ed785496397c13b07bcc83813 : SUCCESS in 27m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/58c6500b60b24372996bb2c2ae995f40 : SUCCESS in 56m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aa61f6880db34193a616ec49f82348c0 : SUCCESS in 28m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a98a30ef25b45cb9a835fc3900ab9a6 : SUCCESS in 5m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e4a4ace9d73245df96b85569e24197d0 : SUCCESS in 20m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2fee41a990f94c85a6559a8770169a8d : SUCCESS in 37m 38s","accounts_in_message":[],"_revision_number":3},{"id":"fd9e1697674ba910774b19ab6f1809a3ffd57cfb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 08:33:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"e06588c375a3c32306479ce1fc767b27752c0b42","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 08:35:12.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/155498f9216f430f9816f10743cea09d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a0a6f4fe1fb4e9ebcadd6f3339bba4b : SUCCESS in 51s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/6617b6e9e62b4069a67008ff8fb21651 : SUCCESS in 54s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"be191cf7d452c852995b21b1e6584727bc729388","revisions":{"c3d5f7a7c4681da53e85c9624fbce6f3325414af":{"kind":"REWORK","_number":1,"created":"2026-09-01 21:41:42.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/87/1003387/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/87/1003387/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/1003387/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/1003387/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/1003387/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/87/1003387/1"}}},"commit":{"parents":[{"commit":"1b8abda1752a1dd42c99f692a915e37f36f18b4d","subject":"Improve comments about multi-region ssync","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1b8abda1752a1dd42c99f692a915e37f36f18b4d"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-09-01 21:39:28.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-09-01 21:39:31.000000000","tz":-420},"subject":"replicator: Quarantine suffixes we cannot delete","message":"replicator: Quarantine suffixes we cannot delete\n\n...rather than (bizarrely) flagging the remotes as failed devices and\nleaving the suffix to fail again on the next pass.\n\nChange-Id: I455acae764327af7aae24170dad7a8445fcee2d4\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c3d5f7a7c4681da53e85c9624fbce6f3325414af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c3d5f7a7c4681da53e85c9624fbce6f3325414af"}]},"branch":"refs/heads/master"},"e47db34d78afb28171b3fed214168ad1f09358af":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-09-02 15:41:06.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/87/1003387/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/87/1003387/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/1003387/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/1003387/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/1003387/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/87/1003387/2"}}},"commit":{"parents":[{"commit":"422a842dd39d0d91f4b38285776b677cc99e1c6b","subject":"Improve comments about multi-region ssync","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/422a842dd39d0d91f4b38285776b677cc99e1c6b"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-09-01 21:39:28.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-09-02 15:38:40.000000000","tz":-420},"subject":"replicator: Quarantine suffixes we cannot delete","message":"replicator: Quarantine suffixes we cannot delete\n\n...rather than (bizarrely) flagging the remotes as failed devices and\nleaving the suffix to fail again on the next pass.\n\nChange-Id: I455acae764327af7aae24170dad7a8445fcee2d4\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e47db34d78afb28171b3fed214168ad1f09358af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e47db34d78afb28171b3fed214168ad1f09358af"}]},"branch":"refs/heads/master"},"be191cf7d452c852995b21b1e6584727bc729388":{"kind":"REWORK","_number":3,"created":"2026-09-03 17:29:43.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/87/1003387/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/87/1003387/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/1003387/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/1003387/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/1003387/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/87/1003387/3"}}},"commit":{"parents":[{"commit":"422a842dd39d0d91f4b38285776b677cc99e1c6b","subject":"Improve comments about multi-region ssync","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/422a842dd39d0d91f4b38285776b677cc99e1c6b"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-09-01 21:39:28.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-09-03 17:27:46.000000000","tz":-420},"subject":"replicator: Quarantine suffixes we cannot delete","message":"replicator: Quarantine suffixes we cannot delete\n\n...at least, when it looks like filesystem corruption. For unknown\nerrors, continue logging (and leaving the suffix in place to fail again\non the next pass).\n\nIn all cases, stop (bizarrely) flagging the remotes as failed devices.\n\nChange-Id: I455acae764327af7aae24170dad7a8445fcee2d4\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/be191cf7d452c852995b21b1e6584727bc729388"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/be191cf7d452c852995b21b1e6584727bc729388"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
