)]}'
{"id":"openstack%2Fswift~1003790","triplet_id":"openstack%2Fswift~master~I7a21852c9a2c0e43b398abe743ec75eddf424cfd","project":"openstack/swift","branch":"master","topic":"s3-checksum-persistence","attention_set":{},"removed_from_attention_set":{"6968":{"account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"last_update":"2026-09-20 05:42:27.000000000","reason":"Change was submitted"},"39146":{"account":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"last_update":"2026-09-20 05:42:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7a21852c9a2c0e43b398abe743ec75eddf424cfd","subject":"doc: add AWS link for well known checksum types","status":"MERGED","created":"2026-09-03 14:15:18.000000000","updated":"2026-09-20 05:43:31.000000000","submitted":"2026-09-20 05:42:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1003790-s3-checksum-persistence","meta_rev_id":"fea38b1680dfac77c9beefd953e424a218c773ac","_number":1003790,"virtual_id_number":1003790,"owner":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-20 05:42:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"all":[{"value":2,"date":"2026-09-19 12:53:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-20 04:35:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-09-20 04:35:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-03 15:16:17.000000000","updated_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"reviewer":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"state":"REVIEWER"},{"updated":"2026-09-03 15:21:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-03 15:40:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-20 04:35:36.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"59f02946af431a48ff7aa6a009d09f1f38054fe7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-09-03 14:15:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a5e372dd67b8ba0b36cd1f46061825a1f40b4a95","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-09-03 14:17:55.000000000","message":"Topic set to s3-checksum-persistence","accounts_in_message":[],"_revision_number":1},{"id":"1ae07330fce4c39609a38e2b69924b55fc915a62","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-09-03 15:16:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ca6d57ad219e4d4fb9d8dd07a2098ca5b2cd4bde","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 15:21:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d232301d5c74e1482008fe6a983213e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5f37155ec84446dfaacff7166aa00bfa : SUCCESS in 54m 00s","accounts_in_message":[],"_revision_number":1},{"id":"896253f3233ed06c51a0132d2d2b7ac5f065b701","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 15:40:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a362906cf95f44f19252707725467fda\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc8db13ffddc4633ad3317a01f48aa2e : SUCCESS in 11m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1df0991e48424a21a5354398b09e9d1b : SUCCESS in 9m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/83954e50a82d42ebaed4596284cbd63b : SUCCESS in 1h 06m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4c3d502fb9cd447cb82dad16c1dd524e : SUCCESS in 45m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/36d6702852214d4bb248721f450fe3b7 : SUCCESS in 34m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/54812a0d1f6c4204b47dfc62c2ced5bd : SUCCESS in 30m 03s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/32cf58f37d404a9697e4fff426f7826c : SUCCESS in 20m 21s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/63766b700a464203b72c6b889e6b3bc4 : SUCCESS in 21m 26s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0a9dfd1bb8b941808af22f6a2c0274b5 : SUCCESS in 14m 07s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6f0ee0bfa2334f6fbe020467556332c6 : SUCCESS in 10m 26s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/fe70dd3461dd40159bbbbc81b29b1054 : SUCCESS in 22m 42s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/dab3adfa202d41aa8f1807c8938dcfe7 : SUCCESS in 21m 06s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/4ac16fe3ba314723ae90e884551e047f : SUCCESS in 23m 44s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/4a80bceed4d5456cb01cabdf6e4a6746 : SUCCESS in 20m 58s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a167f817b11b425aab1227a62a90e91e : SUCCESS in 15m 27s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f1ab49965b8f43ac86b42d65ee796277 : SUCCESS in 9m 32s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/264cd77dc8474f1086af780c3a4449a8 : SUCCESS in 15m 38s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/efbbd6a7293f4aa88e344abc5e6c0a70 : SUCCESS in 11m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1d72cd5dcfdf422dac7392c984c7239c : SUCCESS in 5m 00s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/70bd90392b20459db77b32cacdea94f0 : FAILURE in 19m 35s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/4ad490828f364013a3cf7f619895919e : SUCCESS in 8m 51s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f5e28181e8504872952069d280257ea1 : SUCCESS in 57m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/afa290022e5b44afa5c081e0d4d02afe : SUCCESS in 59m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/85f694dcd2154a829e04e3ef0effe1a9 : SUCCESS in 43m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/706b16985d54471a8677a8f1ee17b96c : SUCCESS in 12m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10eea1de2a6c41008b2f74bb1ad708b1 : SUCCESS in 5m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a5377bbcd3fc42cbb82b5ebc618a7f57 : SUCCESS in 19m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2134a04d1c7541b39df5273bf0abdd71 : SUCCESS in 1h 00m 44s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"b7db6a51045ef93a7bcd095a2cf270abcff9bd91","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-09-19 12:53:50.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"2cfa22959d71ecde62474d5ac589537a55d621cb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-19 13:42:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e8e7e681df74e2a92d97c5d7754e0c2\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/895e0671917d43bbad48b6203ff61912 : SUCCESS in 47m 05s","accounts_in_message":[],"_revision_number":2},{"id":"2ac803048096e6dbfd6f1c966c43d8fc46d3ca77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-19 13:59:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7c13708673114425a50e24a91c459f48\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a10e1f269424384901772bfa8720233 : RETRY_LIMIT\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/51ae8fb004e040158e850c7962ed1eac : SUCCESS in 6m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/0b755cde651e4ce08a4b3026964d56c0 : SUCCESS in 1h 03m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0c8ea0e67c604b9486be26ba0adf2153 : SUCCESS in 1h 02m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fb46f2a855704f5580aa6c84d17a68c2 : SUCCESS in 56m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/19ec680599cf4435b4b04897e61459a7 : SUCCESS in 59m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8644ca4e64c94b4495e3054ea6622fb0 : SUCCESS in 18m 36s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/29d3e394a9174f05b79be4f01068d2d5 : RETRY_LIMIT (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/991472485b004fe5a2ed379839aa0913 : SUCCESS in 19m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/862fe41e2d0a4432b81665e2ef84611d : SUCCESS in 6m 39s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f93dac7289bc4a908c67e3bf7ba9ec45 : SUCCESS in 10m 26s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/cad9f762193c4323aa06b2b200be6dd7 : SUCCESS in 11m 21s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/2d68f9055d3c4f96a7c8bc62d39d48c4 : SUCCESS in 19m 42s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/a3547536b05943fbbd6d9259393cbed2 : RETRY_LIMIT\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/de152fef38944ed7b4e043a1cb08166b : SUCCESS in 11m 26s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/899d79ef4a804f889f6faa65d788e88d : SUCCESS in 13m 24s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b7acd62ce07e4088a8c21b00d5522be4 : RETRY_LIMIT\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/1ce150e9c34f409ca64767946622ff84 : RETRY_LIMIT\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8ad27f4351954b9c8542fa7f69fecf84 : SUCCESS in 8m 16s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/3ebcb9782f9a4dc29d2aa8e8d7084654 : FAILURE in 11m 22s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/32fa1d64ab7d44e9b7b47186e19e3549 : SUCCESS in 57m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/450da8c24e9848b58ae18e97043171be : SUCCESS in 49m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8ecd7a5985d6454aae46fb6541de956c : SUCCESS in 52m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6b2d22e6595d495bb2714d758d121695 : SUCCESS in 29m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfd5811611bc410497ff1d61b2bbed6b : SUCCESS in 3m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/06ab549dd75a4e4eae0ccb5bd68b117a : RETRY_LIMIT\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/25918051a8194f13a88dbabb66c62da7 : SUCCESS in 54m 43s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"f751d91d70c79f758c3520c1dd8a88f328d88edc","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-09-19 20:01:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3df2a1f82b2cd6bec46791bf66d356f162e423bd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-19 21:05:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d0e17178df04ca389547e7433fe5664\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c54bf01c12954d658ade04e23a9c501e : SUCCESS in 1h 01m 46s","accounts_in_message":[],"_revision_number":2},{"id":"9497ab21e5162d8bd1518041564dea8d6ecab1b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-19 21:14:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d878897bcd44d9cb5a294e231aaf4a5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2700f4909fe4a2ea5800c5913f2d381 : SUCCESS in 9m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2580d1fb2f9f4d619d26ad74b6f8056d : SUCCESS in 10m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/dbe0433861114f58b00aec161fc14760 : SUCCESS in 29m 58s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f3554e37bf2944049ae21b6272ebcf0a : SUCCESS in 1h 02m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ffb905d67b554260adb501c7b7e844fb : SUCCESS in 38m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eba0ed7fe66943b5b355c9c57f407ae3 : SUCCESS in 40m 53s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b12e865a757c4287b151e5e2db64154d : SUCCESS in 20m 50s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/694e31f9585b4752925a592b607e9510 : SUCCESS in 20m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/28e651af733b4472b8348a9cebb3cf8b : SUCCESS in 16m 10s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/21abeb73e5754470b15c7e72e05e7a41 : SUCCESS in 6m 50s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3dc8a62fe0fb4da88c6b87a17ab6a94f : SUCCESS in 25m 23s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/92aa331ebf6e4ebf9aa778168887a4f6 : SUCCESS in 16m 41s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/6e8a0eb51c524c028d81e4ccb9a9b5de : SUCCESS in 31m 22s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/f80f6a761463449ba10a2aa4deea1497 : SUCCESS in 20m 00s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4c5abe4cfc8a47e7ba8585d3e7ddec1a : SUCCESS in 16m 08s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d102cb249b184022bffb8911cedddb1d : SUCCESS in 14m 53s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/56c0cf8eff054ef994a35dedfcab43a4 : SUCCESS in 13m 53s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/f20fe8b8309f47bcbde26a55608b4a0d : SUCCESS in 18m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a0f9a582965e4ce3b72f3cb853c40460 : SUCCESS in 8m 51s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/a81cf90b23e047b2aa3d646c082c9ee2 : SUCCESS in 8m 22s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3b83908c1eac42f8a76d44ab51bedce7 : SUCCESS in 1h 06m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8c7a3200217f4a79a4d118d3fb89c486 : SUCCESS in 50m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/faff4076021d4d2c8cad651530304080 : SUCCESS in 28m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e45f92c5ace4f06a0d8cf27fcc8236f : SUCCESS in 28m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3a66803b6544000968c526cd0310207 : SUCCESS in 2m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3fa6ef007da54ef7a21f9b8f46d08fea : SUCCESS in 18m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/00c438bd975a4ab59c2f7d408b0c6c7e : SUCCESS in 25m 20s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"a6db531222ae34e319ab7f38bcd794f2a06aa5d1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-20 04:35:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7115d27d70731d95d244322ff29b8d795435a70a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-20 04:36:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"04615678853a8b746b320af03707f085be597d1b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-20 05:42:27.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ee95d966bca454881c59fa02f546a8c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/673fdec41f1d4d069bfd6a68d9e37bf8 : SUCCESS in 10m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c661262afc9b4ab09df66e015e62334b : SUCCESS in 10m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/311685baf9254f39a8f9674190c5873b : SUCCESS in 1h 02m 29s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3ef5142267304bbe99da28639f89f378 : SUCCESS in 1h 01m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/161eef0a4563478eb8bad73ece972746 : SUCCESS in 34m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1548adc6fc404d4f887edf21c9df0084 : SUCCESS in 33m 14s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/f6d73025823549e7869008e5e22cbb9e : SUCCESS in 7m 15s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/96d1301e0ce24102a34bc773d2587476 : SUCCESS in 26m 58s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/fe19e14b64a146a0b2399f03756acf36 : SUCCESS in 14m 57s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/4aaa352ed83f42e8b6b9ed4dd8a47ebb : SUCCESS in 25m 20s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/da21d7bcf2c9417d8a2c4e77049b073d : SUCCESS in 19m 42s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4236a9594aa84b278a23114f1e79cba2 : SUCCESS in 9m 44s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/af859660f70348ee9b205f49fc20f7d3 : SUCCESS in 9m 53s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1c56cbdea58342cba4d8355e3916adba : SUCCESS in 8m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3099d611d2244db8b8e0e1bb6bb78085 : SUCCESS in 5m 52s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/7798ae537f0b404e964aff67e4162968 : SUCCESS in 9m 09s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4761a0d1a3aa4f2e900f8f15d29940f2 : SUCCESS in 47m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/040c2ce4cea049498a98486e049e9d54 : SUCCESS in 28m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9a70430d1c444101bccb47e9d500ac4b : SUCCESS in 23m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8a8185d474cd413b99beff1fe99feeec : SUCCESS in 27m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1388ed39daff4118b1229bfdecda5d56 : SUCCESS in 4m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/324fdb28b4624d2f98b731df3b90c530 : SUCCESS in 18m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/10bcb7bcc0f44d4691230ec0507825cc : SUCCESS in 27m 02s","accounts_in_message":[],"_revision_number":2},{"id":"8b468a1b3092a5421e7a738d6586dcf11bd8c433","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-20 05:42:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"fea38b1680dfac77c9beefd953e424a218c773ac","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-20 05:43:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8921f1c1975846be90b7853f2f2b11f4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe78e81c3458433eb5e634d2c6c7dbda : SUCCESS in 49s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/f47d7ede1ad940528cd8783ccd649816 : SUCCESS in 47s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"610c5ea8fee44468d00f1a08ba990b50412a8628","revisions":{"9b67114a49ca7a025c4a850989792ec973e13752":{"kind":"REWORK","_number":1,"created":"2026-09-03 14:15:18.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/90/1003790/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/1003790/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/1003790/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/1003790/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/1003790/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/1003790/1"}}},"commit":{"parents":[{"commit":"7d673e3b7d4f21b1f9054cb233fb427cfc1e0aea","subject":"s3api: add x-amz-checksum-type label to checksum metric","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7d673e3b7d4f21b1f9054cb233fb427cfc1e0aea"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-09-03 14:14:02.000000000","tz":-240},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-09-03 14:14:38.000000000","tz":-240},"subject":"doc: add AWS link for well known checksum types","message":"doc: add AWS link for well known checksum types\n\nChange-Id: I7a21852c9a2c0e43b398abe743ec75eddf424cfd\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9b67114a49ca7a025c4a850989792ec973e13752"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9b67114a49ca7a025c4a850989792ec973e13752"}]},"branch":"refs/heads/master"},"610c5ea8fee44468d00f1a08ba990b50412a8628":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-09-19 12:53:50.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/90/1003790/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/1003790/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/1003790/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/1003790/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/1003790/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/1003790/2"}}},"commit":{"parents":[{"commit":"caf11b1e93047eb09b97dc025a33fe4d5f7f7a8d","subject":"Merge \"relinker: Add audit step to quarantine stale hash directories\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/caf11b1e93047eb09b97dc025a33fe4d5f7f7a8d"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-09-03 14:14:02.000000000","tz":-240},"committer":{"name":"Nathaniel Martes","email":"natemartes09@gmail.com","date":"2026-09-19 12:53:43.000000000","tz":-240},"subject":"doc: add AWS link for well known checksum types","message":"doc: add AWS link for well known checksum types\n\nChange-Id: I7a21852c9a2c0e43b398abe743ec75eddf424cfd\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/610c5ea8fee44468d00f1a08ba990b50412a8628"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/610c5ea8fee44468d00f1a08ba990b50412a8628"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
