)]}'
{"id":"openstack%2Fswift~1003875","triplet_id":"openstack%2Fswift~master~Ib82c197c22ef66164a4404acd998ca517125b962","project":"openstack/swift","branch":"master","topic":"s3-checksum-persistence","attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-09-15 17:48:07.000000000","reason":"\u003cGERRIT_ACCOUNT_34930\u003e replied on the change","reason_account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}}},"removed_from_attention_set":{"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2026-09-16 19:07:59.000000000","reason":"\u003cGERRIT_ACCOUNT_34930\u003e replied on the change","reason_account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}}},"hashtags":[],"change_id":"Ib82c197c22ef66164a4404acd998ca517125b962","subject":"checksum: Add Python fallback for combine methods","status":"NEW","created":"2026-09-03 19:48:39.000000000","updated":"2026-09-18 21:27:56.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":25,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"9fd966f88e8132514195cb0714a4089d4bfceb25","_number":1003875,"virtual_id_number":1003875,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-18 21:27:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-03 20:51:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-03 21:04:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-03 22:59:01.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"CC"},{"updated":"2026-09-04 00:14:51.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"REVIEWER"}],"messages":[{"id":"4fbc7fdee185843a98777c5196bfba60c4aea838","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-03 19:48:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5acc236cea68b5ed7c5295605286ab778c3947fc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 20:51:54.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9ee703bf3d8405bb99968375db9c236\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/04280be0c1254f9794efdae50eee0c46 : FAILURE in 1h 00m 57s","accounts_in_message":[],"_revision_number":1},{"id":"608d8d05920dcc982cc07809f8d67fc2185a4631","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-03 21:03:07.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"cd9a03df7fa6a6f794b7856bdfcaa0c5b6063c2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 21:04:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/475c4fbffc0b4160ac993cd0c1e16f3d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8267f85be9a14f058e32bcc00dd82a09 : SUCCESS in 6m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/560f2fe775694f29a57132011f17f595 : SUCCESS in 7m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/fd20fa8a47a64b219d72329fbbea283b : SUCCESS in 1h 03m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f69368eb0d244001bc80b293b7180c5f : SUCCESS in 35m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/30561eecf6534963b5c55f1414e0608e : SUCCESS in 30m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fc8d18bd6e0543cfa96aee0645e4f636 : SUCCESS in 1h 00m 01s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4d82b23f08da4bdeb05f23fcdad5bdda : SUCCESS in 20m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/02f6a75979af487c815ea95109d67e36 : SUCCESS in 20m 20s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c9983a752cc04fc48934c35ff5e02430 : SUCCESS in 19m 50s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2bdc77d194e1485fa932f874426afe7d : SUCCESS in 6m 40s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/baf6dcb2cdbd4b5c9141ddb7f009c84d : FAILURE in 11m 57s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/2988c28bbe784bfd8548dc1dde4c2864 : FAILURE in 15m 59s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/3a055da5574b49ec860fd49b44fa7007 : FAILURE in 14m 42s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/2846677d920e44109ebedc4d4253d235 : FAILURE in 21m 28s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/1c23eee9fdbe4fbebf589822658ba172 : SUCCESS in 14m 48s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/145415c220e749b788af869fcbf2fa01 : SUCCESS in 15m 38s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d8f290549830420baa2289a7e31eb50c : SUCCESS in 14m 56s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/00408491cfcf4332950a51e580941762 : SUCCESS in 17m 18s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/89d3a34892fe464ab21c1ce11bb062cd : SUCCESS in 4m 10s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/991b7567129a437f919072c41c3450b3 : FAILURE in 17m 30s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/8a80a5f75f604e49bd315df76069caeb : SUCCESS in 11m 02s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/27c02c26852e44828c79127e08cd3d91 : SUCCESS in 38m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2f3203a5e3ea4d708b36c7e650065780 : SUCCESS in 1h 02m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b6ad2a870f3741d683266f14652f1330 : SUCCESS in 1h 01m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/76673498212a4f62be9ae82cba085252 : FAILURE in 12m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0d925d4bb204490aaf6f2ded0896ed5 : SUCCESS in 7m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/329fb928083643d480d951cb06d2a250 : SUCCESS in 20m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/afa90e773cfb4a15aca5fd7b60595018 : SUCCESS in 43m 19s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"241f6135e6003c1ff4dedc979bd8ebf068b0bc5e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-03 21:12:40.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"70dfb101f77feea0d2f3de3c30013497e99f5939","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 22:22:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/aaa8a4b0b930424fa8abdb07bf916118\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07ec1aba5e184629a23ac82f50a200f5 : SUCCESS in 5m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2889acc322f04786a667c093f180e0fe : SUCCESS in 3m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/e7f117dad92c4edc9aae4efb6ce91c96 : SUCCESS in 30m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2b5766a6b1284517be297fa4f2fcf3f2 : SUCCESS in 30m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/30e848d7a40d4e8ca29fef45be97a523 : SUCCESS in 52m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/294d7c1d8a674a46b6ebf0f8e527dccc : SUCCESS in 30m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/aa42cb59b1f54dc9b218afc5d92211db : SUCCESS in 17m 17s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/841c6db89725428aa77facb5d2dc1a5a : SUCCESS in 20m 02s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cdf18bc03a054c2e848acf1ba2ad7f41 : SUCCESS in 18m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5207a3912b1844c585cefd1eba2632a3 : SUCCESS in 6m 01s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/88e0d3999c984da8a16273901a8866a7 : FAILURE in 8m 10s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/0395934e90f542e0ab8858cb1fbb2d2c : SUCCESS in 17m 36s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/ab39fbaee5d54fb58d8b3458925c9d46 : SUCCESS in 33m 27s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/9ebae217e5e84448b0f4d41b53a2bae1 : SUCCESS in 16m 14s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/1b3afdcdfbfe4ca6894afaed6484a6c8 : SUCCESS in 18m 03s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9c3b4d806937465491d60ee753ba946c : SUCCESS in 12m 19s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/13635102f348497ca28584c5b1aa87a2 : SUCCESS in 16m 19s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/6073ca3d52574b7fbc1ae412303f6d51 : SUCCESS in 10m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/689ce469ad374bf29177ea9162a3da89 : SUCCESS in 4m 09s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/92cc110604584a9bb9e77cb9f1a35df0 : FAILURE in 30m 07s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/326bf11fc0a7429d98953e633e375d14 : SUCCESS in 14m 15s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/970fbfd36d00464faace14744b7bbbce : SUCCESS in 1h 04m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/723dcb98a65e495bb3bfbb6aa5065406 : SUCCESS in 1h 02m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/75ecd372644541c2bae2a4149dc29af6 : SUCCESS in 28m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3523724660114058821a147c6340993a : FAILURE in 10m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8af230af57a64d66962edf3be9f234be : SUCCESS in 5m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ddc2ec4e1c384d58af0b2877e1e252fc : SUCCESS in 21m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a0e8fbd3f7fd40e08d20ab89767f59dc : SUCCESS in 45m 11s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"a1a58a2ad5f5ae834c750eb33dce666cc457c673","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 22:24:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/774d02aa4b3740c3b77aea3c5b852c69\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7c1ec47925584b0f9d864c4f4708f4ef : SUCCESS in 1h 09m 27s","accounts_in_message":[],"_revision_number":2},{"id":"17bba1d0f70250aa2cb1c2093cfe5431832fc332","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-09-03 22:59:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fd54a164f3bc8a194be752f9fb506907d764990b","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-09-04 00:14:51.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"220520d639a3911ab87999b93052f5756eb480bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-04 18:57:50.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"12798d1465c37a96b7c2e8654bc9a39319f00818","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-04 18:59:06.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"b89f823ce82c1433d6e8d1eaf124125855b3de49","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 19:53:43.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e99c956bef65440796c2b2b7fbd43e96\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e0f8db5058244f9bbdeb325528c62184 : SUCCESS in 53m 19s","accounts_in_message":[],"_revision_number":3},{"id":"1f245c9b5661134f8b49d863f0a1589cec492421","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 20:15:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3eb53f9f7ea4490fa9403a0e0984f45c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59a30cba13f948369989dbc7198e8aba : SUCCESS in 9m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e11b6e9117784da6ac650a9e7f8b722f : SUCCESS in 7m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/717b5bf6ffe24c03a792f19acab4b104 : SUCCESS in 46m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0b0e9e30694447158bdc1de1edb07ac6 : SUCCESS in 47m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0598f987bade4999a571a17261e19f70 : SUCCESS in 42m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d658650edfc34f1a81c207842c69babb : SUCCESS in 1h 08m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/85722d9c2c2644f486238c696a60d6d2 : SUCCESS in 20m 19s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7d10a96d2e584e069ff7fba9488c6195 : SUCCESS in 20m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4ca3cda48def4e89bd8d4eab0f8b0c19 : SUCCESS in 21m 56s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/523856aea05640d78748d7ca341e0796 : SUCCESS in 8m 46s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a47d2dacdf2a4413bde6bd8f3e84d4a7 : SUCCESS in 23m 04s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/996d54aa87f84d799747d2724a1f6c93 : SUCCESS in 22m 43s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/dd3b8aaa0bad4d4fbcc2a46f1304f353 : SUCCESS in 34m 28s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/9397651e29e141ec9652a3e981b96789 : SUCCESS in 19m 58s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ba5b7d6c99f048ad913e593dbb1329a2 : SUCCESS in 15m 24s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e02edbc612f84e16a8ab78336cd2e3a6 : SUCCESS in 15m 55s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c979e013bee84c629f9441769943a62c : SUCCESS in 16m 38s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/54c6a462db74450991d3479c717b085f : SUCCESS in 17m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fc8fc5a11b3e4419a3ed533f11753a1c : SUCCESS in 8m 32s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/df2f360780b1473a916a6499f65e0c04 : FAILURE in 33m 41s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e34026460f474ad6a18151c961afabd7 : SUCCESS in 13m 13s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d4b6001761f34f38a74759e25a4f9832 : SUCCESS in 1h 01m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d53ac39365114b9a80f390994511885b : SUCCESS in 58m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/160d3cd7d36e44739a3b6661aaed649c : SUCCESS in 54m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4c92b4ccb99d405093c3a3813c36ae35 : SUCCESS in 26m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ed4393748b844b78dc6c4e1e22fc70c : SUCCESS in 4m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d9cab73624594ee69076633237516ada : SUCCESS in 19m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/40d337519fcd43d9a458e0fe722255a4 : SUCCESS in 1h 01m 22s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":3},{"id":"1a9e09771262c8c4ddb2e02f1618bd258552d11b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-04 21:13:44.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ae1ae8843c3361855630c460941cef958d82ae22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-11 16:12:02.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"a018179813bf487d2917c71e59acb1b64030f185","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 16:43:52.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0852aab620e44a55959b3998ff30b98a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8671d7896ab343bb9c016ee1183d6f75 : SUCCESS in 29m 59s","accounts_in_message":[],"_revision_number":4},{"id":"b167aa09503dbd281319b6eaccc1f8c7334d1cd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 17:21:18.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f69980b1f80545c2ad61607bed365d00\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f79ba47c4674a97973c2e14e115a03f : SUCCESS in 5m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/93a75dd8fc4748bca5864cfd9c1b1ee3 : SUCCESS in 4m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/4923eeb462e64b478de7bc07ce2ebc43 : SUCCESS in 32m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/47b6b61c67ba4ac1bdead7aa10857fbf : SUCCESS in 1h 01m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/92cf1d303d764ce58a4ce4e6c9490c62 : SUCCESS in 53m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/af16b380bde84ec983cd6bda9e6c8990 : SUCCESS in 1h 02m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/43886e00db414d028f448a14143384dd : SUCCESS in 20m 59s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1a92ad188300400abdf5214788945739 : SUCCESS in 21m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/706ae68a534b4a069908cdc4c8f91f28 : SUCCESS in 22m 31s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/39444a13a4ca496694b22c21530d5a79 : SUCCESS in 9m 11s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4f0e9417286448b880101a5f844ee436 : SUCCESS in 14m 25s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/817a0a5bc1ba40f0a7207345d3e9b15b : SUCCESS in 10m 47s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/b6fa3a37511446989220c00f5823815a : SUCCESS in 36m 20s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/abb784c214054d10b1e2c243fe118ec6 : SUCCESS in 21m 01s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/63f0de450e624389884181b62b03f165 : SUCCESS in 13m 24s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0f4703bc20ca414dba218bbf3d6bd427 : SUCCESS in 13m 17s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b7e1e5dad861450ca21bad512a9de2f0 : SUCCESS in 13m 49s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/58fb8703a34749398efc90343a65c118 : SUCCESS in 18m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0829ea8dbf4f4f1ab0959a9d9e0c8ee2 : SUCCESS in 5m 57s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2cf5a99a4f0542dfb441c2fdfda3618e : FAILURE in 28m 39s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/bd737a6233354bad9b8f1fcb6a6a2010 : SUCCESS in 10m 40s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7921c0b6ee93473d9ee0a387406df972 : SUCCESS in 1h 00m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7ddfd48a844c4545acc8d73cc94ee7e1 : SUCCESS in 45m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1669b15cfd6847be9498c6377a6fa34a : SUCCESS in 28m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/54cefbe18fa3497f947f4c32e1f3390e : SUCCESS in 12m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bcceef8919640f0b10a81b74f085b85 : SUCCESS in 4m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ea34ea6048044ab3b4680fd8b391191e : SUCCESS in 29m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9e57faa9caa741eb8c6878d706c72e3b : SUCCESS in 28m 43s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":4},{"id":"baceadf290cb87a57f7725098d5830fee9311c77","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-09-13 18:50:23.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"b9b1c92495a8b7000b28b4b60441f97acd14656e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-14 20:03:12.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"4d2142655c9d3fc7ac994bef2b4429f9cd4d0fcd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-14 20:03:23.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"416865eb8fe2ea56404ddab9c148102029e46f8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 21:10:10.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d13f4a0121440678e157b9a5d1fd289\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/479bffc92a7e4a8f8d7fb05f32c2eba8 : SUCCESS in 6m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3fb5575de37b4bcd94ff6bede389e49a : SUCCESS in 9m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/e3980c311601419ca6e309e7b28fa396 : SUCCESS in 32m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/687df613b9964d738bcd33d07ea13bcb : SUCCESS in 36m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/352dc434d9604f4eaff7fa07d6e6127d : SUCCESS in 32m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dbc71c6c372a4cbaa0edbec8a3e12c0e : SUCCESS in 34m 23s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cd12764ce925444b84ac1d174dd74c61 : SUCCESS in 20m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7966c2bed7e24f94bb6611e95958cd88 : SUCCESS in 11m 56s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3eceb5c60a0e4226a8d34cee82a45261 : SUCCESS in 13m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7a8e085cdf934a3b8dae2f39b6408223 : SUCCESS in 8m 53s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/716e658926b84d14ba627899f24f30f4 : SUCCESS in 26m 04s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/81df9390723e46dfb1538a6bbdcd0b44 : SUCCESS in 25m 01s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/bcdfa743b758400a873160a049c2d9e5 : SUCCESS in 12m 56s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/56c9278762bc47c3acd77cb63d85ff1a : SUCCESS in 16m 01s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/535d143ac57f4eec933b2e407f5b35c1 : SUCCESS in 15m 04s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/946c4503f2954d068a17e9921181ea31 : SUCCESS in 8m 54s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c5471931da27471cb734ac2c93a1ba6b : SUCCESS in 14m 16s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/523dfc08bf1046508a08a26a00715312 : SUCCESS in 13m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c105492e7e51476792f53403d646e63a : SUCCESS in 7m 48s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/db0670dfebdd4c95b2b6f3438f53ad97 : SUCCESS in 10m 40s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/773f7f9b449943b6a47b981cd9550a00 : SUCCESS in 1h 00m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/64157e69138141aa9be04814cd4dbfd0 : SUCCESS in 37m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/71786d832cb04db69fb88044c0d73473 : SUCCESS in 57m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ed9711192f14878b690ef767505bd82 : SUCCESS in 24m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70bbb985495040dcad5e6fb7e401e630 : SUCCESS in 3m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/906d41d9057641cc95f6235c7e47a6b4 : SUCCESS in 10m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e4ffbbaeebeb457eb4f65be5e95825a9 : SUCCESS in 28m 16s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":5},{"id":"d72ecd67d6da1e114c951050841b85a6900aa950","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 21:10:37.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c4d016ad836495a9bd10d7d0a194306\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/66a3ccd9ce5b4f4b814aa787a30c8147 : SUCCESS in 54m 28s","accounts_in_message":[],"_revision_number":5},{"id":"b434c65f685ad2af9f9c5e926ac39a3ca70af43a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-14 23:32:54.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"b06f79326e6cc5e300d42ab6d974e746f48d153d","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-09-15 17:48:07.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"55cc0fcb92732f86da6df5651d551ce1a02dcee1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-09-16 18:28:00.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"5c380624b8e85f45d28060f1836392f5bff64434","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-09-16 18:42:36.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"1c8360173c4db0c8e1e53dcb341d520458ace6e2","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-09-16 19:07:59.000000000","message":"Patch Set 7:\n\n(4 comments)","accounts_in_message":[],"_revision_number":7},{"id":"5fa1b166009169b61a28ba40992e42446e7e4a93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-09-16 19:24:19.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"914b9ac34de6859886eddc63de699980d37de59b","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-09-16 19:24:50.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"2eedbcaeb90832e3f49a300f83bb42df249c25bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 20:36:05.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/172b7b7621954357936e77c2fa67bd0a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2237bd1a5ac45b9a2815465b2aab82f : SUCCESS in 11m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/11ed5cce0fc14d98853df825a9c9a82b : SUCCESS in 8m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/f6eed20063e74bd786fa438916179c9d : SUCCESS in 1h 05m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2bef006fc1824c239d0cca163b09f656 : SUCCESS in 1h 02m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4e73d08d21e8461abf24174d3d8db678 : SUCCESS in 49m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/402a39dd13b24471947206ffb3063453 : SUCCESS in 38m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7c8e7ba8ba81448f9264e89d1bd01db1 : SUCCESS in 15m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0aee84cdf4104f16aeda816159bbace1 : SUCCESS in 18m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e354da5ebb1749399227d40561e426d2 : SUCCESS in 16m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ac9ab14d55a54951980f6d420262b5d2 : SUCCESS in 5m 22s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/380bcfde7e9b4063b89c0869102e538c : SUCCESS in 28m 05s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/f830d0da143843bbb4b06aa14e1acac1 : SUCCESS in 22m 31s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/5709fc79d9154cc9802890a525614aed : SUCCESS in 34m 16s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/08112e6c6cff4361953f0b3c273213a7 : SUCCESS in 19m 43s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c21a03078fbd456381327938f9037442 : SUCCESS in 14m 48s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/72a153b5f4614dcb8ea65ac632e70a69 : SUCCESS in 14m 28s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e8c8c145a8f34f619669bd86e3b238dc : SUCCESS in 11m 03s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/0ff4248bd33045e3be872a9d9bf21a9b : SUCCESS in 16m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1af3fbb1e59f4155bcc4cecb302a5f7e : SUCCESS in 6m 19s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/41de01fa20f7405d877f7b71455ef3c8 : SUCCESS in 9m 56s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5edfe4d1b8ab45dcaa48cfe0f3de8ad4 : SUCCESS in 55m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a04ecc00be16481d83f93157495d935f : SUCCESS in 37m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f09951cdebd74490ac05f17b143434b9 : SUCCESS in 1h 02m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af8b6c2f1eca43708b725ae64ed7bb64 : SUCCESS in 29m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1c59bddb7f34fd280d77391a23e99f1 : SUCCESS in 3m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1d32a8b77aa74a85877aef9a2346999d : SUCCESS in 27m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/feb569b8746c4cda93a8a39faaa26e55 : SUCCESS in 56m 18s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":8},{"id":"3afac6aabb1a7eda99c68e7df9ddbd854f7b0859","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-09-16 21:38:41.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"00da7609bb33fe67921dc4a05da04c74a283cce9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 22:54:30.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1457eb11733440138091ec5502ebc2bb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a87a4f7b7134997a3dd1477e28eee70 : SUCCESS in 8m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce81b4020f4643c98fc01071c28c230b : SUCCESS in 10m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/9389e9af801247818f87c9bf9b422ffc : SUCCESS in 47m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fbfdc8406f8541a8925d964169acc6a2 : SUCCESS in 1h 09m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2fc6bf13d0bc4f2082efc28efa74a4aa : SUCCESS in 39m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/54a1bd84adac4e79af406cdc02151e35 : SUCCESS in 1h 02m 08s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2fd21624d78c4e169724ed4bb2fb3767 : SUCCESS in 22m 01s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/86926017f2624ad2b241a096b41a8b4b : SUCCESS in 23m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c091421daa1f4e80b0327e9132246846 : SUCCESS in 22m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b0a88cabe41a4e859068f2335a0c6a82 : SUCCESS in 5m 30s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9eeb83fe34904c81b3ef74619aef2c09 : SUCCESS in 26m 06s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/3fcb85601dee472a84fa884e32fdb0e2 : SUCCESS in 23m 08s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/da5d3dc89785449dab7120ae1e65cd73 : SUCCESS in 33m 59s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/bb1d3de5235a4f1cb5d465255e06bbf1 : SUCCESS in 15m 53s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b0dc4a858d1b484aa62b24d1ffc229c0 : SUCCESS in 13m 39s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/cda0825a115b4a3e96e5f77ba980fff9 : SUCCESS in 15m 10s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/2d3bd6190cf7433daef9b8d87c500cb6 : SUCCESS in 15m 56s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/6d3b68c0d20244a7afba83c2e8bc99a1 : SUCCESS in 17m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/497d8ff7fea4434ab462f4659dacd70b : SUCCESS in 3m 59s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/4124ea5cf7354142993a811d85824625 : SUCCESS in 11m 37s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/de1faa9195e14b5f8495b1664704bb72 : SUCCESS in 1h 00m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/86f71d733573415aaaa13e4e319b57aa : SUCCESS in 59m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/77b42b063428401a8ca5ac79f75a34d7 : SUCCESS in 58m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a09005eda0bc44a7816a3f550df72841 : SUCCESS in 23m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9db2bef821e143548935f14da8f29a4d : SUCCESS in 4m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8cf0cca577ee43f68803643a5264f45e : SUCCESS in 18m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/969d1af37822441198b76306ac7be6be : SUCCESS in 1h 03m 10s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":9},{"id":"dbfe661fba2ccd74b94b80e780e27ee486fe5644","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 01:18:36.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98872e22b7bb44819ecbc899548c8257\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/48b5570dfa0d4bff9aa6fa8d74c41cf9 : SUCCESS in 1h 22m 30s","accounts_in_message":[],"_revision_number":9},{"id":"e7641fcaeaa611c2a8c39317f37fff55142aa113","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-09-17 22:19:06.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"1d4cc529a72f34eeae3fd006ffe3f6b17cd6c70b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 23:12:58.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b8cddbca6fb402eb6c3b70659538fc8\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ff4f7ba264334b909f7b0afcbb88bca2 : SUCCESS in 51m 25s","accounts_in_message":[],"_revision_number":10},{"id":"c743531b53d173aa00fd01445545fac72eb90800","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 23:29:57.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e88491befac40cf91844c3b9fae88ef\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07a2d587f77a4d319652b28b93f7273e : SUCCESS in 6m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/59e45d475cbd43d2bb2af2ee6491dfe0 : SUCCESS in 8m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/38e7405e9392491295928eef5d00d9f6 : SUCCESS in 31m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9ddea193d253423ab4f675d1b9cbc7ba : SUCCESS in 46m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d99be503e88d4d619f5cc58f34f2d362 : SUCCESS in 37m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fcc03f3b3b3e4832b8e6d09e3d617b4e : SUCCESS in 54m 33s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7f481d4f3fa743128345ec7d586ef19e : SUCCESS in 19m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/df7b9d80792c4c6a92ec653b25cdd0cf : SUCCESS in 18m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8a8b1339d13647aab33b4df9483e1a87 : SUCCESS in 20m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/288fed2aa3c248a09a4cef6ada6a0304 : SUCCESS in 7m 08s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/fa2c08f50f8a4bc39cc4d0107e2c35cb : SUCCESS in 18m 58s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/011269b655de401e8b9e146ce7e66d4d : SUCCESS in 18m 09s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/8aae5b2e4f134115a32c5b4e99d13007 : SUCCESS in 15m 23s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/f625737bec88413b97aaa595990fc02a : SUCCESS in 21m 21s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4289b4160e7140e1a1275351c7e42679 : SUCCESS in 14m 56s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9ff381bd55d74cec92aaad66bc14d5e1 : SUCCESS in 16m 39s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b15e63148bbd4140a166ced13d9d91e6 : SUCCESS in 13m 35s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/72b025790846460682816aad960b33f1 : SUCCESS in 18m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/932af18f48224ecb87ee203191d8fb5a : SUCCESS in 4m 17s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/5bfa7405ffc34bf1823f14b5f144c320 : SUCCESS in 10m 31s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c0ce9970ede7400ba4952867d792d94c : SUCCESS in 1h 04m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a135c05341d94cf58c1d699f01814669 : SUCCESS in 34m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4220e60b26434fe8a32c15f23a7fd2f9 : SUCCESS in 37m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5cffee5da1c746cf9c23b999acad4716 : SUCCESS in 28m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbe74f671d7b4414b9d0737f276663e0 : SUCCESS in 4m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ba97be3765e44bd1a2ef286bd4877a58 : SUCCESS in 19m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a0c471c49ccd43ccb5c010c678a2ab9f : SUCCESS in 45m 37s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":10},{"id":"f0b436ca0c7ef44e7e9974ea945241451c8782d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-09-18 05:19:28.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"995f3159a8aba68785470c6af8960468a5d5d5e6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 06:10:00.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c8cbb73fa7948068bd89df6cf2a283d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f34b0bb170274c118ec12625d6136158 : SUCCESS in 48m 27s","accounts_in_message":[],"_revision_number":11},{"id":"84db4c279b54131eca1845fec9e5f55e7c24f3a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 06:29:48.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7746b20a57014fc0aafbb3bfbc516b89\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9a497b37a8d4009be964caf04a7b42f : SUCCESS in 10m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8ea275101754409a576d5eb83be6354 : SUCCESS in 6m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/f14f911e2d074d13945846158e434f3c : SUCCESS in 35m 14s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/75fe80e2a0b443ae9bc1144cbc8fd4ec : SUCCESS in 1h 00m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cbe5dd5ea371480fbaaa811fae115329 : SUCCESS in 30m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/827ddea4581a4e42870cc5abb1ad989f : SUCCESS in 1h 03m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d5c0d1a578dc4df6a5e6508422c332e3 : SUCCESS in 19m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/15dd29ad494744bfbf6a077da2d17bf4 : SUCCESS in 20m 37s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2ba55f5e2f104f6aaeaad95aba210025 : SUCCESS in 11m 28s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/26749e02108944238e3f917dfe87b4eb : SUCCESS in 6m 45s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a5d46e9fac5a46f7aac5eaa1f62ba3db : SUCCESS in 26m 20s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/0959028def284b348b48a68ff0d98ba1 : SUCCESS in 11m 23s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d194380afc474d0bb30eb5d27cf075fb : SUCCESS in 31m 42s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/9dbd345129214e7393fb1115d23323a7 : SUCCESS in 19m 30s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ae5a57f27cae40a7b3c39eddd90b9e03 : SUCCESS in 11m 28s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a293e189c1514bc0b2cc830a66860f84 : SUCCESS in 14m 17s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/767b443ac1a74264a2b7ba9a0884b54d : SUCCESS in 15m 21s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/ba8fafb215bf4d83b764f9f35f642a67 : SUCCESS in 13m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/dd27dd4382b745119e7979d9fe1c0c2b : SUCCESS in 5m 55s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/c0ef868ed9514e83b6faaebd71645940 : SUCCESS in 8m 15s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/bd90d95861854305ba896bb599900576 : SUCCESS in 57m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/40153e6f6acd4fd6902e8bfcb1def7af : SUCCESS in 25m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1badf3eec1e54c5c887e79192cbf70fa : SUCCESS in 56m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/18a63c939cac481f8e23eff1b1ed0b1c : SUCCESS in 23m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be02482e7996421b86a09b0bbc149d55 : SUCCESS in 3m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/60fd2695cd7f4649a7044dbd8f45fc55 : SUCCESS in 16m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d2df7cbc66c245d8b3496392c99da567 : SUCCESS in 31m 26s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":11},{"id":"9fb2a0ffe2b4779a49f48d118ea2513aa9f06f5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-09-18 20:14:52.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"ba75bb7539c3ac38f057e24c756e71818126b843","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 21:17:06.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3b745bf98294e2fa4548941de1d7bbe\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ac1f04b8ceb9474a9742d46b760c62fd : SUCCESS in 53m 28s","accounts_in_message":[],"_revision_number":12},{"id":"9fd966f88e8132514195cb0714a4089d4bfceb25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 21:27:56.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c6fe2cf6c604351927fa60550e1fe09\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b68cd44b4dba49149b797758f3841317 : SUCCESS in 11m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f6c0f5920a1c491db85c44fec5628f9a : SUCCESS in 9m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/8cf172b2c8634636a9614fd1650cc7d8 : SUCCESS in 1h 05m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f680c4140314409fae4fef12b971e7d4 : SUCCESS in 32m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1d22bae9eae042a0ad01e4dac8c955ab : SUCCESS in 52m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cb45349a873e416a82281024e9450817 : SUCCESS in 53m 02s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8bc2909da615429fb4c73ba502711803 : SUCCESS in 16m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9f90071890ba4ec29b6970cf260f5ff3 : SUCCESS in 20m 58s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0d76585ae73d483eaaf963955c9d7625 : SUCCESS in 21m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9641cb07c1794adb96feb7d6e60a1867 : SUCCESS in 6m 57s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2e0c902409bc4006a7de533cf32078e0 : SUCCESS in 11m 33s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/2db5d62bc5724c6784cd89bbab8de044 : SUCCESS in 21m 21s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/ae7a22d89dd54b559b04e225e7fb606d : SUCCESS in 32m 59s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/d35fd9ea82d04da3801ec59f1be3d45d : SUCCESS in 20m 51s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9ff75e2ae16a4820b65b746f652d1705 : SUCCESS in 15m 00s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a3af6a2225cc484da27e1d856ed89fa3 : SUCCESS in 15m 52s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d14f6530819f4776964717b2a122e7dc : SUCCESS in 9m 47s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/a8157e7bf79a43be88eba649f61b5573 : SUCCESS in 10m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/57374c1f5c0f4f459d78b28491c27656 : SUCCESS in 7m 03s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/ecb6546170f24b02ac3a06452197f559 : SUCCESS in 10m 53s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f52fd829697a4f9dbce3f0be8606540f : SUCCESS in 59m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/023e2e3fc4bf4df98c792069f4261735 : SUCCESS in 40m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fa0359db6d3b439ab30deb656d34cd2e : SUCCESS in 41m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/718273cbd5ed4684ba327c40ed4048e0 : SUCCESS in 29m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa3d53b25ef84b45b845068d7c871ec9 : SUCCESS in 4m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1a54bce336854ab5ae5f2579c34ef6b5 : SUCCESS in 18m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/47545339ec294d60bb210cf6ec941eca : SUCCESS in 37m 27s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"b1ccbe5ac492c67bc0b4228467dcf92977a04b55","revisions":{"a345e4a02dc9d5b8d06a7fef9742766377ada33b":{"kind":"REWORK","_number":1,"created":"2026-09-03 19:48:39.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/75/1003875/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/75/1003875/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/1003875/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/1003875/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/1003875/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/75/1003875/1"}}},"commit":{"parents":[{"commit":"3eb67c54ef1b8b8c6ca539c7683e7d5d0cff9657","subject":"s3api: Checksum Persistence on CompleteMultipartUpload","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3eb67c54ef1b8b8c6ca539c7683e7d5d0cff9657"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-09-03 18:30:01.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-09-03 19:46:35.000000000","tz":-420},"subject":"checksum: Add Python fallback for combine methods","message":"checksum: Add Python fallback for combine methods\n\nIt\u0027ll be slower, but better than nothing. With the matrix caching,\nit\u0027s pretty tolerable on decent hardware, even if it gets a little\nCPU-intensive on wimpier machines.\n\nAdd a cooperative yield every 250 combine()s for the sake of slower\nhardware.\n\nCo-Authored-By: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\nChange-Id: Ib82c197c22ef66164a4404acd998ca517125b962\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a345e4a02dc9d5b8d06a7fef9742766377ada33b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a345e4a02dc9d5b8d06a7fef9742766377ada33b"}]},"branch":"refs/heads/master"},"928014eb6bcc49b95df619f406e35f4851fdac18":{"kind":"REWORK","_number":2,"created":"2026-09-03 21:12:40.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/75/1003875/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/75/1003875/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/1003875/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/1003875/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/1003875/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/75/1003875/2"}}},"commit":{"parents":[{"commit":"3eb67c54ef1b8b8c6ca539c7683e7d5d0cff9657","subject":"s3api: Checksum Persistence on CompleteMultipartUpload","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3eb67c54ef1b8b8c6ca539c7683e7d5d0cff9657"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-09-03 18:30:01.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-09-03 21:11:39.000000000","tz":-420},"subject":"checksum: Add Python fallback for combine methods","message":"checksum: Add Python fallback for combine methods\n\nIt\u0027ll be slower, but better than nothing. With the matrix caching,\nit\u0027s pretty tolerable on decent hardware, even if it gets a little\nCPU-intensive on wimpier machines.\n\nAdd a cooperative yield every 50 combine()s for the sake of slower\nhardware.\n\nCo-Authored-By: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\nChange-Id: Ib82c197c22ef66164a4404acd998ca517125b962\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/928014eb6bcc49b95df619f406e35f4851fdac18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/928014eb6bcc49b95df619f406e35f4851fdac18"}]},"branch":"refs/heads/master"},"f8610665c6221fad655de1a511b274a7caf8fb3e":{"kind":"REWORK","_number":3,"created":"2026-09-04 18:57:50.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/75/1003875/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/75/1003875/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/1003875/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/1003875/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/1003875/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/75/1003875/3"}}},"commit":{"parents":[{"commit":"3eb67c54ef1b8b8c6ca539c7683e7d5d0cff9657","subject":"s3api: Checksum Persistence on CompleteMultipartUpload","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3eb67c54ef1b8b8c6ca539c7683e7d5d0cff9657"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-09-03 18:30:01.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-09-04 18:56:57.000000000","tz":-420},"subject":"checksum: Add Python fallback for combine methods","message":"checksum: Add Python fallback for combine methods\n\nIt\u0027ll be slower, but better than nothing. With the matrix caching,\nit\u0027s pretty tolerable on decent hardware, even if it gets a little\nCPU-intensive on wimpier machines.\n\nAdd a cooperative yield every 50 combine()s for the sake of slower\nhardware.\n\nCo-Authored-By: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\nChange-Id: Ib82c197c22ef66164a4404acd998ca517125b962\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f8610665c6221fad655de1a511b274a7caf8fb3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f8610665c6221fad655de1a511b274a7caf8fb3e"}]},"branch":"refs/heads/master"},"fd43bf1c4512736aaf4a1923d4c39e4b1c69660a":{"kind":"REWORK","_number":4,"created":"2026-09-11 16:12:02.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/75/1003875/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/75/1003875/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/1003875/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/1003875/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/1003875/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/75/1003875/4"}}},"commit":{"parents":[{"commit":"4aedf6761dea26ea5f7c69ec1b75718f2f420e9e","subject":"s3api: Checksum Persistence on CompleteMultipartUpload","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4aedf6761dea26ea5f7c69ec1b75718f2f420e9e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-09-03 18:30:01.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-09-11 16:08:50.000000000","tz":-420},"subject":"checksum: Add Python fallback for combine methods","message":"checksum: Add Python fallback for combine methods\n\nIt\u0027ll be slower, but better than nothing. With the matrix caching,\nit\u0027s pretty tolerable on decent hardware, even if it gets a little\nCPU-intensive on wimpier machines.\n\nIn s3api, add a cooperative yield every 50 combine()s for the sake of\nslower hardware.\n\nCo-Authored-By: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\nChange-Id: Ib82c197c22ef66164a4404acd998ca517125b962\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fd43bf1c4512736aaf4a1923d4c39e4b1c69660a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fd43bf1c4512736aaf4a1923d4c39e4b1c69660a"}]},"branch":"refs/heads/master"},"c68912340f39e036c5479feba98e7545fce06890":{"kind":"REWORK","_number":5,"created":"2026-09-14 20:03:23.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/75/1003875/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/75/1003875/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/1003875/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/1003875/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/1003875/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/75/1003875/5"}}},"commit":{"parents":[{"commit":"4aedf6761dea26ea5f7c69ec1b75718f2f420e9e","subject":"s3api: Checksum Persistence on CompleteMultipartUpload","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4aedf6761dea26ea5f7c69ec1b75718f2f420e9e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-09-03 18:30:01.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-09-14 20:02:50.000000000","tz":-420},"subject":"checksum: Add Python fallback for combine methods","message":"checksum: Add Python fallback for combine methods\n\nIt\u0027ll be slower, but better than nothing. With the matrix caching,\nit\u0027s pretty tolerable on decent hardware, even if it gets a little\nCPU-intensive on wimpier machines.\n\nIn s3api, add a cooperative yield every 50 combine()s for the sake of\nslower hardware.\n\nCo-Authored-By: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\nChange-Id: Ib82c197c22ef66164a4404acd998ca517125b962\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c68912340f39e036c5479feba98e7545fce06890"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c68912340f39e036c5479feba98e7545fce06890"}]},"branch":"refs/heads/master"},"0adca7e690e886f79b7ee4af242b695beeff18a6":{"kind":"REWORK","_number":6,"created":"2026-09-16 18:28:00.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/75/1003875/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/75/1003875/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/1003875/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/1003875/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/1003875/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/75/1003875/6"}}},"commit":{"parents":[{"commit":"335cf46ea4416b70bc76b849385a05fa857f2f16","subject":"s3api: Checksum Persistence on CompleteMultipartUpload","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/335cf46ea4416b70bc76b849385a05fa857f2f16"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-09-03 18:30:01.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2026-09-16 18:27:29.000000000","tz":-420},"subject":"checksum: Add Python fallback for combine methods","message":"checksum: Add Python fallback for combine methods\n\nIt\u0027ll be slower, but better than nothing. With the matrix caching,\nit\u0027s pretty tolerable on decent hardware, even if it gets a little\nCPU-intensive on wimpier machines.\n\nIn s3api, add a cooperative yield every 50 combine()s for the sake of\nslower hardware.\n\nCo-Authored-By: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\nChange-Id: Ib82c197c22ef66164a4404acd998ca517125b962\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0adca7e690e886f79b7ee4af242b695beeff18a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0adca7e690e886f79b7ee4af242b695beeff18a6"}]},"branch":"refs/heads/master"},"25e0fd1fd8af0b9626f226b80be11acbf5dc4969":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-09-16 18:42:36.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/75/1003875/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/75/1003875/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/1003875/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/1003875/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/1003875/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/75/1003875/7"}}},"commit":{"parents":[{"commit":"d69d2b5dc82a9d0c40147ab67d93bb4cbc396ba4","subject":"s3api: Checksum Persistence on CompleteMultipartUpload","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d69d2b5dc82a9d0c40147ab67d93bb4cbc396ba4"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-09-03 18:30:01.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2026-09-16 18:42:19.000000000","tz":-420},"subject":"checksum: Add Python fallback for combine methods","message":"checksum: Add Python fallback for combine methods\n\nIt\u0027ll be slower, but better than nothing. With the matrix caching,\nit\u0027s pretty tolerable on decent hardware, even if it gets a little\nCPU-intensive on wimpier machines.\n\nIn s3api, add a cooperative yield every 50 combine()s for the sake of\nslower hardware.\n\nCo-Authored-By: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\nChange-Id: Ib82c197c22ef66164a4404acd998ca517125b962\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/25e0fd1fd8af0b9626f226b80be11acbf5dc4969"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/25e0fd1fd8af0b9626f226b80be11acbf5dc4969"}]},"branch":"refs/heads/master"},"7d0463a6d12c769aab98f82e2438ebd1e7a17ee5":{"kind":"REWORK","_number":8,"created":"2026-09-16 19:24:19.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/75/1003875/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/75/1003875/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/1003875/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/1003875/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/1003875/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/75/1003875/8"}}},"commit":{"parents":[{"commit":"d69d2b5dc82a9d0c40147ab67d93bb4cbc396ba4","subject":"s3api: Checksum Persistence on CompleteMultipartUpload","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d69d2b5dc82a9d0c40147ab67d93bb4cbc396ba4"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-09-03 18:30:01.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2026-09-16 19:06:04.000000000","tz":-420},"subject":"checksum: Add Python fallback for combine methods","message":"checksum: Add Python fallback for combine methods\n\nIt\u0027ll be slower, but better than nothing. With the matrix caching,\nit\u0027s pretty tolerable on decent hardware, even if it gets a little\nCPU-intensive on wimpier machines.\n\nIn s3api, add a cooperative yield every 50 combine()s for the sake of\nslower hardware.\n\nCo-Authored-By: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nChange-Id: Ib82c197c22ef66164a4404acd998ca517125b962\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7d0463a6d12c769aab98f82e2438ebd1e7a17ee5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7d0463a6d12c769aab98f82e2438ebd1e7a17ee5"}]},"branch":"refs/heads/master"},"c21622f2c38e1ac6d121ef2ab9fbed9e5f0364ff":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2026-09-16 21:38:41.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/75/1003875/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/75/1003875/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/1003875/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/1003875/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/1003875/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/75/1003875/9"}}},"commit":{"parents":[{"commit":"35cf83abf941253916b1b5a3a1729f041c44a42c","subject":"s3api: Checksum Persistence on CompleteMultipartUpload","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/35cf83abf941253916b1b5a3a1729f041c44a42c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-09-03 18:30:01.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2026-09-16 21:38:18.000000000","tz":-420},"subject":"checksum: Add Python fallback for combine methods","message":"checksum: Add Python fallback for combine methods\n\nIt\u0027ll be slower, but better than nothing. With the matrix caching,\nit\u0027s pretty tolerable on decent hardware, even if it gets a little\nCPU-intensive on wimpier machines.\n\nIn s3api, add a cooperative yield every 50 combine()s for the sake of\nslower hardware.\n\nCo-Authored-By: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nChange-Id: Ib82c197c22ef66164a4404acd998ca517125b962\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c21622f2c38e1ac6d121ef2ab9fbed9e5f0364ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c21622f2c38e1ac6d121ef2ab9fbed9e5f0364ff"}]},"branch":"refs/heads/master"},"e576b79daecff1e214460fd16277550b23d4460b":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2026-09-17 22:19:06.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/75/1003875/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/75/1003875/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/1003875/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/1003875/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/1003875/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/75/1003875/10"}}},"commit":{"parents":[{"commit":"baf68678435dd89186c8e14aa9af72e5e7e1f3f2","subject":"s3api: Checksum Persistence on CompleteMultipartUpload","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/baf68678435dd89186c8e14aa9af72e5e7e1f3f2"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-09-03 18:30:01.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2026-09-17 21:47:41.000000000","tz":-420},"subject":"checksum: Add Python fallback for combine methods","message":"checksum: Add Python fallback for combine methods\n\nIt\u0027ll be slower, but better than nothing. With the matrix caching,\nit\u0027s pretty tolerable on decent hardware, even if it gets a little\nCPU-intensive on wimpier machines.\n\nIn s3api, add a cooperative yield every 50 combine()s for the sake of\nslower hardware.\n\nCo-Authored-By: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nChange-Id: Ib82c197c22ef66164a4404acd998ca517125b962\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e576b79daecff1e214460fd16277550b23d4460b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e576b79daecff1e214460fd16277550b23d4460b"}]},"branch":"refs/heads/master"},"e36f2d74d923ee372681ec6c0ff5f5b63b03f7cb":{"kind":"REWORK","_number":11,"created":"2026-09-18 05:19:28.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/75/1003875/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/75/1003875/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/1003875/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/1003875/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/1003875/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/75/1003875/11"}}},"commit":{"parents":[{"commit":"20c59d9bd2d964045ef583798f5b5a931492ffe3","subject":"s3api: Checksum Persistence on CompleteMultipartUpload","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/20c59d9bd2d964045ef583798f5b5a931492ffe3"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-09-03 18:30:01.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2026-09-18 05:12:54.000000000","tz":-420},"subject":"checksum: Add Python fallback for combine methods","message":"checksum: Add Python fallback for combine methods\n\nIt\u0027ll be slower, but better than nothing. With the matrix caching,\nit\u0027s pretty tolerable on decent hardware, even if it gets a little\nCPU-intensive on wimpier machines.\n\nIn s3api, add a cooperative yield every 50 combine()s for the sake of\nslower hardware.\n\nCo-Authored-By: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nChange-Id: Ib82c197c22ef66164a4404acd998ca517125b962\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e36f2d74d923ee372681ec6c0ff5f5b63b03f7cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e36f2d74d923ee372681ec6c0ff5f5b63b03f7cb"}]},"branch":"refs/heads/master"},"b1ccbe5ac492c67bc0b4228467dcf92977a04b55":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2026-09-18 20:14:52.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/75/1003875/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/75/1003875/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/1003875/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/1003875/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/1003875/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/75/1003875/12"}}},"commit":{"parents":[{"commit":"4763b353e70f9684389ba5d7b48d5fb4b29232ca","subject":"s3api: Checksum Persistence on CompleteMultipartUpload","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4763b353e70f9684389ba5d7b48d5fb4b29232ca"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-09-03 18:30:01.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2026-09-18 20:14:38.000000000","tz":-420},"subject":"checksum: Add Python fallback for combine methods","message":"checksum: Add Python fallback for combine methods\n\nIt\u0027ll be slower, but better than nothing. With the matrix caching,\nit\u0027s pretty tolerable on decent hardware, even if it gets a little\nCPU-intensive on wimpier machines.\n\nIn s3api, add a cooperative yield every 50 combine()s for the sake of\nslower hardware.\n\nCo-Authored-By: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nChange-Id: Ib82c197c22ef66164a4404acd998ca517125b962\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b1ccbe5ac492c67bc0b4228467dcf92977a04b55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b1ccbe5ac492c67bc0b4228467dcf92977a04b55"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
