)]}'
{"id":"openstack%2Fswift~1003885","triplet_id":"openstack%2Fswift~master~I3e1ee9d752be1f86c19e5ea55a75a27528df63e3","project":"openstack/swift","branch":"master","topic":"tmp_label_validation","hashtags":[],"change_id":"I3e1ee9d752be1f86c19e5ea55a75a27528df63e3","subject":"Add per-call statsd label escaping","status":"NEW","created":"2026-09-03 21:25:58.000000000","updated":"2026-09-29 20:57:51.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"cff492af34c0ffcfabc04a876a86e2882aa719ac","_number":1003885,"virtual_id_number":1003885,"owner":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-29 20:57:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-03 22:28:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"984a9e238c7d63371dd3a49eea250cf3769c6df1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2026-09-03 21:25:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2775a9aaf34740beb97584be5eccac12028a690b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 22:28:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72c71496e032412ab48c1d8345950148\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ae679a604be4fb69c71ac70af78da49 : SUCCESS in 10m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/217aa2dffdfd4c3b8d377584b27dd603 : SUCCESS in 7m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/bbe2e4b5e6cb4dc88d533e334f90fdce : SUCCESS in 46m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7634acdec9d744a49c384c2bfcbebfd2 : SUCCESS in 34m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8817688f773a4a628a5dde5f47ef6e87 : SUCCESS in 32m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5451829dd6b9422ea2526b4cccc11471 : SUCCESS in 39m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/65b1622aba2e46dda6e3cb877200e218 : SUCCESS in 21m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/92d76d2f722340b1bab1ad29b10da864 : SUCCESS in 20m 16s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8f12f137381940f6abaee41b003dec71 : SUCCESS in 11m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/30c7a1c80443431c81a044c7e2951b64 : SUCCESS in 6m 27s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3caed2747ef64087bfd72b39fd29a40a : SUCCESS in 10m 44s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d712541946f747c9873d9f2c6117eb6c : SUCCESS in 21m 02s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/1e2bb29d87b14f23b6e4759b8db66775 : SUCCESS in 32m 51s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/3b84060cd25245a9907176f5dd78519e : SUCCESS in 20m 30s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b9681f8bd9f44fa4bf902123e89206fd : SUCCESS in 12m 43s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/773904ce98414e0ab3373e0bdc482891 : SUCCESS in 13m 12s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3fcc916631474a37a077ce2c835eddac : SUCCESS in 9m 21s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/82ab283d969a4f0e923afe242022a281 : SUCCESS in 10m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/92d4da3f4c1543c380eb8bb47d7a5200 : SUCCESS in 6m 55s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b75770fc103f4f96a5c7ec850514d665 : FAILURE in 36m 25s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/138a5a75cf124beaba4240cad1cd8221 : SUCCESS in 4m 58s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/42113da6b1954e4c93e8f5eea1c8fc7f : SUCCESS in 39m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8ff1d4155d2e49da9dc2e54bda1b8554 : SUCCESS in 23m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b1046f7e437c414883c72806d5d21d92 : SUCCESS in 56m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c007a190ac424b9e9ef42bbb249d29ca : SUCCESS in 28m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b36d07958cf4eb5ada08c5c942b5422 : SUCCESS in 4m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/50d22108012d4af9a6f9c4e2a6eef240 : SUCCESS in 9m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a994681f5c624673a72e03ebd6825886 : SUCCESS in 56m 46s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"d920f825ad963b2702276db1ad5cfaabf8c77b19","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 22:36:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/672674682223432a972eec20e8655ae1\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/134e3ef6c2ab417bbde6153c72961cc3 : SUCCESS in 1h 08m 16s","accounts_in_message":[],"_revision_number":1},{"id":"e0d5c37f9acb906bf38b6f43ac6c436e2a8f43cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2026-09-29 19:36:35.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"6aa33fc0db2b084e54521fc58c92d72999467508","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 20:09:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef3636960b39484da21b8fffcbdcbca9\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c5591b0bf86f4340ba7f2cdee0ec324b : SUCCESS in 31m 21s","accounts_in_message":[],"_revision_number":2},{"id":"cff492af34c0ffcfabc04a876a86e2882aa719ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 20:57:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/690b797111be40fcaded5f3a739e642d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0b6bbaae6f045da93b4f7703e0d79ee : SUCCESS in 10m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b87567c6995441d984a95e787a72313 : SUCCESS in 4m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/6d7b86334cea4a53856c40f183bcc28e : SUCCESS in 1h 08m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/34bcabe3fe4748b588a2bbf98a4c1ec6 : SUCCESS in 59m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/267d0f17349f4c039de99e69c9bf7692 : SUCCESS in 56m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9166073e200c4cccba8cbea3de939e7d : SUCCESS in 1h 15m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/020d67bdc9a643af8158d78f43a9c804 : SUCCESS in 20m 49s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bb2b0523cd2149ad95df6a4b13311c44 : SUCCESS in 19m 42s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1ff6daf98b5448fdba56d92aabfb447a : SUCCESS in 20m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ab4656018fd04bfb959977ee9354938d : SUCCESS in 7m 16s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2393e57ec93c4822ba05a60640a79269 : SUCCESS in 30m 03s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/e66d20931d3341bda6a2e08838a0314b : SUCCESS in 25m 33s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/b8e4c4ddbda34b0695963db9080306e9 : SUCCESS in 38m 13s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/f426086b18b84f84836aeb71adca7c42 : SUCCESS in 23m 16s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7860de58a30b4a208374ba04c241d5ce : SUCCESS in 14m 07s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5184c0aca523495aa8e171c2c47e6055 : SUCCESS in 16m 23s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c4f54ef023e04059a5ccbb28fc1e1aa7 : SUCCESS in 15m 40s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/27f0d9d9d5ab41aebd517f63ad99b389 : SUCCESS in 18m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a3d6e61e79274a0f8c9fde5015e01caa : SUCCESS in 6m 36s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/5edbba5868594da7bcdde489c8213a0c : SUCCESS in 12m 42s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a3672a7aa2f54889a7ead680024f1e04 : SUCCESS in 59m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/af4df71c81124fb7a11b85d91ea726aa : SUCCESS in 55m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/70346f719cc34e10841d326fed466feb : SUCCESS in 50m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bc861374d23c4606bcde7ff595494dfe : SUCCESS in 11m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de280d5874754649857bda6990a27087 : SUCCESS in 4m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3e8f338f7c1849cca8bf12d735ccf75d : SUCCESS in 20m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a930d8f45bf74977b3fb8c0ed7976d8c : SUCCESS in 27m 35s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5331e15b913f82c6391e1531a5b00d33e7fc3c62","revisions":{"c017e17eb893af0ab6598dbaecd869b3a89cdd6b":{"kind":"REWORK","_number":1,"created":"2026-09-03 21:25:58.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/85/1003885/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/1003885/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/1003885/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/1003885/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/1003885/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/1003885/1"}}},"commit":{"parents":[{"commit":"1c49c2d5465cc8b5e8cd46c187dbee9473ce46a0","subject":"add PrometheusClient for file-based metrics export","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1c49c2d5465cc8b5e8cd46c187dbee9473ce46a0"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2026-09-02 19:15:43.000000000","tz":-240},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2026-09-03 21:25:26.000000000","tz":-240},"subject":"Add per-call statsd label validation","message":"Add per-call statsd label validation\n\nFor LabeledStatsdClient, _build_line currently emits caller-supplied\nlabels without applying the same validation as configured statsd user\nlabel keys and values in default_labels. Consequently, callers could\nemit label keys and values containing invalid characters such as\nsyntax delimiters that produce malformed or semantically altered\nmetrics on the wire.\n\nChange-Id: I3e1ee9d752be1f86c19e5ea55a75a27528df63e3\nSigned-off-by: Yan Xiao \u003cyanxiao@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c017e17eb893af0ab6598dbaecd869b3a89cdd6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c017e17eb893af0ab6598dbaecd869b3a89cdd6b"}]},"branch":"refs/heads/master"},"5331e15b913f82c6391e1531a5b00d33e7fc3c62":{"kind":"REWORK","_number":2,"created":"2026-09-29 19:36:35.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/85/1003885/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/1003885/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/1003885/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/1003885/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/1003885/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/1003885/2"}}},"commit":{"parents":[{"commit":"e60791398bfefc0f2b3c4b0012c292d983248765","subject":"Merge \"tempurl: Disallow X-Copy-From on tempurl requests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e60791398bfefc0f2b3c4b0012c292d983248765"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2026-09-02 19:15:43.000000000","tz":-240},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2026-09-29 19:35:26.000000000","tz":-240},"subject":"Add per-call statsd label escaping","message":"Add per-call statsd label escaping\n\nFor LabeledStatsdClient, _build_line currently emits caller-supplied\nlabels without escaping the special characters in labels. Consequently,\ncallers could emit labels containing invalid characters such as syntax\ndelimiters that produce malformed or semantically altered metrics on\nthe wire.\n\nChange-Id: I3e1ee9d752be1f86c19e5ea55a75a27528df63e3\nSigned-off-by: Yan Xiao \u003cyanxiao@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5331e15b913f82c6391e1531a5b00d33e7fc3c62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5331e15b913f82c6391e1531a5b00d33e7fc3c62"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
