)]}'
{"id":"openstack%2Fswift~1003890","triplet_id":"openstack%2Fswift~master~Ic88dc8a0e9402df2fb7254c8b2b0825b547deb6d","project":"openstack/swift","branch":"master","topic":"s3-checksum-persistence","attention_set":{},"removed_from_attention_set":{"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2026-09-10 19:22:15.000000000","reason":"\u003cGERRIT_ACCOUNT_34930\u003e replied on the change","reason_account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-09-10 21:34:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic88dc8a0e9402df2fb7254c8b2b0825b547deb6d","subject":"checksum: Add ISA-L implementation for CRC32","status":"MERGED","created":"2026-09-03 22:12:44.000000000","updated":"2026-09-10 21:35:46.000000000","submitted":"2026-09-10 21:34:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1003890-s3-checksum-persistence","meta_rev_id":"d2ce3e67c759cb9451e41b87dfb79529d302b1c6","_number":1003890,"virtual_id_number":1003890,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-10 21:34:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"all":[{"value":2,"date":"2026-09-10 19:22:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"all":[{"value":1,"date":"2026-09-10 19:22:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-03 23:15:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-03 23:24:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-08 23:15:19.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"REVIEWER"}],"messages":[{"id":"e7cd3772656ed40780aa4c1d4503ce0c0ab95753","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-03 22:12:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"36b6549f2b42d01e95d23e5899e08e8e6833ce04","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-03 22:28:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"70b17c39e52343b5d104335e2788793c17f8fdd2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 23:15:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d93d493496e74134a7c281da527df387\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/024cb8dd07ef47bfb25a3cd9fb95a932 : SUCCESS in 59m 48s","accounts_in_message":[],"_revision_number":1},{"id":"39f86c3033aa15c621e27435ca73fe151ed5cb7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 23:24:29.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/718b4a209c3e476b84f68c93072b3ab6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93a44c36d49e4fcfa91bd0bb53544a2a : SUCCESS in 5m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/440bacd2bea649af83e58b8eae7e8436 : SUCCESS in 9m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/4d01670370c744bbbc3b2b45393b6fab : SUCCESS in 31m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e823ce045ba144c4a9926741d13b9c47 : SUCCESS in 46m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8fa486aacecb4586a0de9106e58c952e : SUCCESS in 29m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e9b8481f37ca4bad989fef399c4dd841 : SUCCESS in 28m 12s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7b43c4c69c5b42bcba16bef67153fff6 : SUCCESS in 15m 19s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/753a483cae364735829189a4fc17d95a : SUCCESS in 10m 56s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e038c56fa85049d6a8299567d67ab8be : SUCCESS in 11m 11s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4ea0c02debae4ac197b66886aad5dd7e : SUCCESS in 6m 26s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/840b53db53434e5bae437cbfe79b3bf4 : SUCCESS in 11m 26s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b8365ac29df741a1813c06ccb2a80725 : SUCCESS in 10m 10s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/08a45811e0e944358b45235d5493ac87 : SUCCESS in 24m 28s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/66f8305a1d7f4e32ada1ab23f48e819e : SUCCESS in 20m 04s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a805762c965941fdb854fec47ee0db69 : SUCCESS in 15m 47s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/18c507dd2c5a41489b02d2b75b932398 : SUCCESS in 14m 44s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/86e46d742a9a4bd1a0e2762c5711480d : SUCCESS in 16m 21s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/a83785ab21cc42a1b021c23fae8a820b : SUCCESS in 18m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b5a1cfb16f8c42e4b557152cce4f58c9 : SUCCESS in 9m 39s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4a21b5427cf244b98a7ac967553b70c3 : FAILURE in 17m 19s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/1868c74102594802a844e80860b8904a : SUCCESS in 6m 42s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/fe861944d8c14e8abb4fe1baa2d69141 : SUCCESS in 58m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/81f2f49a322042ad822bfff469c60f33 : SUCCESS in 57m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1b06d872f09044fbabf697b4c3848eb8 : SUCCESS in 28m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c16c60389c5b40eaa661ad2949320bbc : SUCCESS in 26m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecb0876551c54799875e03a7b72f0889 : FAILURE in 4m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/23cd7197261a48d28363e89d460e76e7 : SUCCESS in 9m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c1d1bd05a9434bd789dc3e0e3e8754c0 : SUCCESS in 58m 20s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"9823d3e22be00de4fb79cdad8728ce15f6c66fc7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-04 19:01:43.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"996f26f8e066e8cf5ab2892b49ea09292ab70ee4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 19:58:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbbbf4226d1b4a28bc2bfe1f9ef81968\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c70b4be4846842eea511bf3be37be7d8 : SUCCESS in 53m 09s","accounts_in_message":[],"_revision_number":2},{"id":"6a84351b3092a8ba0836964c78b4b43788dde3b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 20:16:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35d3021d795b458e9d9ab4f0422ed3e1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6aeeadd9965f462b9ec5cb7505d04e1a : SUCCESS in 9m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d114fe2116884099b79ce6e8711a82e7 : SUCCESS in 11m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/46764c69d21042efbdccc694e7ee978a : SUCCESS in 44m 40s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3d4d404b71924803964ee08544e65454 : SUCCESS in 1h 00m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c189d65336d840e3b8aba24aec048577 : SUCCESS in 55m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b73008d0881549259358e900f59b1f59 : SUCCESS in 57m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8c451852734040348d69c0e52eb6f5d7 : SUCCESS in 17m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a344431c26f64cb3803f22d95d8a6224 : SUCCESS in 20m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e7c15111c2284f62950af5fd33a9000c : SUCCESS in 17m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/407c77773f5e470c8acceed9c4826b60 : SUCCESS in 6m 34s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/87a05dfdeebd467096041c5d741bd6b1 : SUCCESS in 26m 04s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/58487520216149ca9a37d4dc5b3ee22d : SUCCESS in 17m 11s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/f7ae07eff39e47a29f7ff7cb6fb69f33 : SUCCESS in 29m 01s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/ae4ce2dba09d4bbabf5bbd74b68e8a5d : SUCCESS in 10m 25s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/8ce6e268a307452baae3f9ad542e8b54 : SUCCESS in 9m 17s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8c846910ab4741c2a0867a1860a911c9 : SUCCESS in 14m 26s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/84fb717132bf4e7fa47f8a69d9da58b1 : SUCCESS in 16m 33s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/bbabcfeae23e40719f2ff427b878cf39 : SUCCESS in 10m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/17414e044bd4486a8d6d94566183a6cc : SUCCESS in 6m 16s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b9338ee1e30d4aa9a5cd2a41d532fb64 : FAILURE in 26m 29s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/8a28ea8d668046489007085230654944 : SUCCESS in 12m 48s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1b60c684d14042e1b2d69f10009f47c2 : SUCCESS in 1h 01m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8dac8aa94b804f0881f58d0d83ef21f5 : SUCCESS in 46m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8ec1ee7f18cf45de9cc925c6a77aea86 : SUCCESS in 28m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ad55f52cb4e94d3d8cfa5e2e2a1541eb : SUCCESS in 22m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cda54409122241188f9d0b8e8e364e08 : SUCCESS in 5m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8ac1c4cde84a4736a35cd6dd87ae9770 : SUCCESS in 26m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/48d153acc36546bb90334893e669cf7f : SUCCESS in 1h 02m 33s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"7ede05493322057db1f1b55cae7eeff53e0c66c7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-04 21:02:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ffac6939fd9a5fb3c8f0c8d593796aba2c04e915","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-09-08 20:46:23.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"89222568114f7b0efb3f6e60c36a18be30970771","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 21:49:01.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6711688086384f9b9dab395c409029d9\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7131fc1d7ce04f5dbcfd4b966ec041a6 : FAILURE in 1h 00m 24s","accounts_in_message":[],"_revision_number":3},{"id":"ff782640dabaab1f910b562cfd230d5099c63f4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 21:58:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f2dc7a07029475d98013a3694ef531d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0124a05268fe4d68a2c761c9805ff9aa : SUCCESS in 11m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4504f8f5751c4a91bfb01a5f0ddcc7fe : SUCCESS in 7m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/7d4b2b3637db4df59f6754f15ea9dde9 : SUCCESS in 31m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/88023e8695934716a24a6f1ecbaffeb2 : SUCCESS in 1h 05m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/df67519e0a4e4dd7a09e24275aa15651 : SUCCESS in 46m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/61dabe3f60dc4bae8d7ec1202eb5a7c5 : SUCCESS in 43m 34s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5aca688bc5234e6a84e48aa1cb8e0937 : SUCCESS in 21m 23s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e633166d650c4c5a9b427e3ff24060f9 : SUCCESS in 15m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4da5f51f07bc40f8b127522bcd0647ba : SUCCESS in 23m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/911f3316cd92407296bfc9a7ce338243 : SUCCESS in 7m 18s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/46a2593abdbd416e95e76316c8bf1e2d : SUCCESS in 27m 37s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d6d284e746ac4468b7f804d66d891b7c : SUCCESS in 10m 05s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/a1f85650d15e4a7f914c0f7b2f7af593 : SUCCESS in 23m 16s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/43e80bbce9604000a0e86bfb536721be : SUCCESS in 25m 03s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/caff5fa918464525ac34cc08e41347d5 : SUCCESS in 16m 34s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/62818226a599444480cdecd375435f5e : SUCCESS in 17m 48s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b5b99ffe6543456eb52f81d3eb74a91a : SUCCESS in 14m 51s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/26f4fde3dc98474dac7dc063b3e69de4 : SUCCESS in 15m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8d8b241744c844ca9bfe53dd5d2d872c : SUCCESS in 7m 03s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6c3a95b185794209840a949aa5676a18 : FAILURE in 31m 59s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e007a82e7d864343bbc09eee421f0b2b : SUCCESS in 7m 44s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7b08875de1994b88ae9185a7930abd3a : SUCCESS in 37m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8eff80b1ee0246ee95b766247be66d56 : SUCCESS in 57m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/194fa0677b3644afae84dacaa7745308 : SUCCESS in 53m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/06d9efb50213415fb0bce4da88c99775 : SUCCESS in 32m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f37935f69974406995a520c2708c4800 : SUCCESS in 3m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/411617856bed4d309bdb8f2c83ad00df : SUCCESS in 21m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/98d2d07915b445048d8d4a5d0adecb47 : SUCCESS in 1h 00m 35s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":3},{"id":"471ec187664076a40d6cc70d345ba4291fc2736e","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-09-08 23:15:19.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"66607cc50e837101b272952acaf5fabd3ec07aeb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-10 16:40:15.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"88de5d725f3f97b4b5dd9f579515655b2a2ec0c1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-10 16:43:04.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7c4798d39acc2fcf5b300ee42d74c2fb517cf695","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 17:19:17.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b10875daf3bc4a6cb420a746fa7ef2a6\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/78a845f5d1304fb6b8ae222d8742e2f9 : SUCCESS in 36m 14s","accounts_in_message":[],"_revision_number":4},{"id":"945da7351147bc880575b6f95b6c8c1834a7c2e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 17:52:41.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9ca525beb6444e82b2543fc81d47edcd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e765ccec0b74233aaf82aa6c31f38cc : SUCCESS in 11m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5ca07d4c7be34e29acc1f4c686fccae4 : SUCCESS in 3m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/66f7f30343054dcf8b80a698659133fd : SUCCESS in 1h 05m 42s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e855f2bbdc544df6904344f08361a031 : SUCCESS in 59m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ed8022659c544e698ab28ca76f526cf4 : SUCCESS in 37m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/603d4925f544451f941b40d2c1bbc22d : SUCCESS in 1h 05m 03s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ae431373bfeb41fa91fa32c2b09231e5 : SUCCESS in 15m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/07ccefbf7c8a44e2a27b24753ee2fbd6 : SUCCESS in 15m 39s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/559784e6971e479ba525cce470067d79 : SUCCESS in 11m 03s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/59c3870fee0941de917096855546824a : SUCCESS in 4m 01s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d8568a99cbc14935a37059eefd0d4cb3 : SUCCESS in 27m 23s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/66634eee88c7443f81991bbe71484251 : SUCCESS in 14m 47s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/517e9488f3ef45fb83a2b0ed926dc177 : SUCCESS in 30m 11s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/b56cad6591b74a1d8d65c0d703bd5667 : SUCCESS in 21m 25s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/8575b125c4624e5282f422a0e6166a00 : SUCCESS in 15m 41s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/31a33d395e81466dad153400a68a0ffd : SUCCESS in 12m 14s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/63971b38d124483fb7c8566fa86a0bdc : SUCCESS in 18m 31s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/f2005188e3314a75957b23b517cb5a99 : SUCCESS in 9m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/645609557c6b44039317d37449f58a7f : SUCCESS in 3m 49s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8177d38f5d08478dbca04246aca34b56 : FAILURE in 31m 35s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/1cdb773fda7a4d00bfe72c409470f369 : SUCCESS in 8m 36s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/06b4a4c878d94a7eb3b85ceab32b9599 : SUCCESS in 1h 01m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/99d5341b14f2491a89aa4018924f0aaa : SUCCESS in 55m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0c74a3e4f77c44aab1e8aee3f02798f3 : SUCCESS in 54m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bef84f9e979e4c469d1010484e280a5f : SUCCESS in 30m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d060d810eb5d42eeb5ee74b7105fea44 : FAILURE in 3m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/01fc7795ae23487282188f1ee5ce1243 : SUCCESS in 19m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e57d4d87b52f4f6185a456503c2f87a2 : SUCCESS in 26m 52s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":4},{"id":"0908beee5413be9f8399d1a1406d4348822c5c40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-10 19:01:51.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"2d282a8188c5d69adfbc686654e4311b1852c3da","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-09-10 19:22:15.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"21976439d3f8f02f722fc08fc8e6506e78d57401","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 20:02:15.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e41780b3a8c04ba7b0a047ecf9e1e063\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/229d9aeee5cd48f7afb299bff20d6a51 : SUCCESS in 58m 05s","accounts_in_message":[],"_revision_number":5},{"id":"1e077adf443ef39822471e02a66238375ed1c827","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 20:21:58.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a673aa93d7849878ea23469df9c5a6f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8fec7dc59eb4ffc85598c70942ffc90 : SUCCESS in 9m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79fe51083c7144619111b249fdf5dc62 : SUCCESS in 9m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/4a44a9ae044f437081b39cae469fa67a : SUCCESS in 1h 05m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d1bdb83f3ee54ef89cfe887cf1b05eda : SUCCESS in 33m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c16e7f4dd9994cda9702d458cd8fe4a2 : SUCCESS in 44m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ae950584f7514cf8a954d8135bf5ee71 : SUCCESS in 28m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b778f54ccfe04009ab7f33529a4c1975 : SUCCESS in 18m 10s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/02809425f8c24f0f8f7a3e87b14cf615 : SUCCESS in 20m 46s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c4c3d9f819284a5ea263597228cfe7b6 : SUCCESS in 10m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ce58322ad3e6488a9de7354f2fd9fe0c : SUCCESS in 6m 34s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/84db7c082bf74f39893b8635bb8cb331 : SUCCESS in 10m 38s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/e9979dc1d4474ccda75ba8f5310018ed : SUCCESS in 21m 09s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/ae98a5d0b30b479394cae5775c5e0040 : SUCCESS in 26m 40s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/b4009b12a28b42f2b6f3e6cb6279a8d2 : SUCCESS in 22m 10s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6f79462ef0b045bbbdbab8d88b1cfdca : SUCCESS in 16m 08s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0bab6742c94e4c7ead55de10e86654ce : SUCCESS in 9m 03s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/bc7eb378d7cd4a69bbce7a673b03a72e : SUCCESS in 15m 18s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/46d62a1492c94cf39b170ef93cad0248 : SUCCESS in 10m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/238e9a07a7ce48b2b396107a492fe8b8 : SUCCESS in 3m 33s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cdf540d97feb4292848d608acededd69 : FAILURE in 26m 18s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/c2afeafb16dc416387d979e69fde7cbc : SUCCESS in 5m 10s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/20baa1986e1e4550a387927669064eaf : SUCCESS in 38m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6527aaec5bbc4c369ad2b22c90d16a8b : SUCCESS in 59m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7093e364970748039731d881b35799bf : SUCCESS in 41m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d814e92f031e43c78b9fe7bbcf66e5fe : SUCCESS in 11m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9d0d1133e114810956087eb3c33240f : SUCCESS in 3m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a97d54ff9bfd404b851177a2a42e1f23 : SUCCESS in 26m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d51f104791ec47689ea9b7a8c4d501aa : SUCCESS in 55m 54s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":5},{"id":"ccad5ddef95f2e1d23e4cb9943e40e3bc4c240cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 20:23:04.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"4678d1512a7129cb5cd5cac71e6581b2350de878","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 21:34:43.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28ba5a3549684e749f7b04b03f835e52\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e816be3023664fb6b483b5be40c57e74 : SUCCESS in 10m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dea01381ce55413c8f2be5e1b056e54d : SUCCESS in 10m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/f83bcf11cd8a42a18f298f515720cba3 : SUCCESS in 1h 02m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a37eb31a0a214ce08b86969629ecc8c6 : SUCCESS in 1h 06m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d79db0ebe32a42a388af9adc83461e06 : SUCCESS in 51m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a810378d87bf436680a638c8f631f8be : SUCCESS in 1h 00m 41s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/4ad759d3af974749844bac6c25fa7e66 : SUCCESS in 6m 28s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/96251875a1cd4bdcba4e7ad301a7d673 : SUCCESS in 26m 32s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/e8dbb227e57e4c7c8e7ad8e789f8b334 : SUCCESS in 20m 07s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/43eaefb1758840b49da0ee56b3db375e : SUCCESS in 13m 50s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/81b8cfd66ad648ab85e343daf9813b0d : SUCCESS in 21m 16s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/569098c1a96b464d875e40b026455b0b : SUCCESS in 15m 25s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/159ec7c0138c4b76a164dd57271b48d1 : SUCCESS in 15m 34s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/53c66f269dad4df1af6719eb9783de01 : SUCCESS in 18m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/68acdf08f429480582c38bf0c9a0e1cc : SUCCESS in 6m 15s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/58f440ce9de144d1ba5e70dece42c244 : SUCCESS in 8m 49s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/bfb689d908f5469d8f26bcfa1064fabd : SUCCESS in 59m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c6b6f5a2bfe64acc8c28d4b95f5db24a : SUCCESS in 26m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6e0af4f702874aa8964c23e8c31d1dff : SUCCESS in 27m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d3a64d0fd7774d37b87efcc1dd033358 : SUCCESS in 24m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/430ae55fc372451295878e2a37df7eeb : SUCCESS in 4m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b347d349abca4bc6bf0262b646edf06d : SUCCESS in 20m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5f4c3fc6c1684255bb0ae2b274834da5 : SUCCESS in 48m 05s","accounts_in_message":[],"_revision_number":5},{"id":"46f574364a066de7b51206cd95f8f2cbf07b9555","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 21:34:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"d2ce3e67c759cb9451e41b87dfb79529d302b1c6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 21:35:46.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7c2da9380324186aaf5b06df7176746\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6061aac4c94458ebb6788a530da90c1 : SUCCESS in 51s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/d3a35e06a6a046f59aeb2df411f1892f : SUCCESS in 51s (non-voting)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"fe1ae77c33170927fe5629ca10aa0d1e165b85c2","revisions":{"5822594d9eb89a183728fa2bf6346b9398e18695":{"kind":"REWORK","_number":1,"created":"2026-09-03 22:12:44.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/90/1003890/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/1003890/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/1003890/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/1003890/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/1003890/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/1003890/1"}}},"commit":{"parents":[{"commit":"568d312317d7ae8e679f9473b91cb466ec0c9057","subject":"utils/checksum: Prefer anycrc for CRC32","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/568d312317d7ae8e679f9473b91cb466ec0c9057"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-09-03 22:11:01.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-09-03 22:11:01.000000000","tz":-420},"subject":"checksum: Add ISA-L implementation for CRC32","message":"checksum: Add ISA-L implementation for CRC32\n\nChange-Id: Ic88dc8a0e9402df2fb7254c8b2b0825b547deb6d\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5822594d9eb89a183728fa2bf6346b9398e18695"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5822594d9eb89a183728fa2bf6346b9398e18695"}]},"branch":"refs/heads/master"},"8941d29367445835c0cd3ac9b7ee1e347fb75bbd":{"kind":"REWORK","_number":2,"created":"2026-09-04 19:01:43.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/90/1003890/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/1003890/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/1003890/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/1003890/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/1003890/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/1003890/2"}}},"commit":{"parents":[{"commit":"568d312317d7ae8e679f9473b91cb466ec0c9057","subject":"utils/checksum: Prefer anycrc for CRC32","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/568d312317d7ae8e679f9473b91cb466ec0c9057"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-09-03 22:11:01.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-09-04 19:01:34.000000000","tz":-420},"subject":"checksum: Add ISA-L implementation for CRC32","message":"checksum: Add ISA-L implementation for CRC32\n\nChange-Id: Ic88dc8a0e9402df2fb7254c8b2b0825b547deb6d\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8941d29367445835c0cd3ac9b7ee1e347fb75bbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8941d29367445835c0cd3ac9b7ee1e347fb75bbd"}]},"branch":"refs/heads/master"},"9434f3a244577bce91770115e83ddbd479417583":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-09-08 20:46:23.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/90/1003890/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/1003890/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/1003890/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/1003890/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/1003890/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/1003890/3"}}},"commit":{"parents":[{"commit":"f4aa2a7fcf53aa133cd73987506bada73343c6ae","subject":"utils/checksum: Prefer anycrc for CRC32","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f4aa2a7fcf53aa133cd73987506bada73343c6ae"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-09-03 22:11:01.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2026-09-08 20:46:10.000000000","tz":-420},"subject":"checksum: Add ISA-L implementation for CRC32","message":"checksum: Add ISA-L implementation for CRC32\n\nChange-Id: Ic88dc8a0e9402df2fb7254c8b2b0825b547deb6d\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9434f3a244577bce91770115e83ddbd479417583"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9434f3a244577bce91770115e83ddbd479417583"}]},"branch":"refs/heads/master"},"1eaaee85e3574af09df44397600b7ce76bdfaeba":{"kind":"REWORK","_number":4,"created":"2026-09-10 16:40:15.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/90/1003890/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/1003890/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/1003890/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/1003890/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/1003890/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/1003890/4"}}},"commit":{"parents":[{"commit":"396cf55464a006beb4b6c401f3f2e41f2323290c","subject":"utils/checksum: Prefer anycrc for CRC32","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/396cf55464a006beb4b6c401f3f2e41f2323290c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-09-03 22:11:01.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-09-10 16:34:08.000000000","tz":-420},"subject":"checksum: Add ISA-L implementation for CRC32","message":"checksum: Add ISA-L implementation for CRC32\n\nOn a production node, ISA-L delivered 15.3-34.1 GiB/s across\n4 MiB-4 GiB inputs for one-shot and 64 KiB incremental CRC32\ncalculations -- 7.6-16.9 times faster than anycrc and 12.5-27.9\ntimes faster than zlib -- while producing identical checksums.\n\nChange-Id: Ic88dc8a0e9402df2fb7254c8b2b0825b547deb6d\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1eaaee85e3574af09df44397600b7ce76bdfaeba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1eaaee85e3574af09df44397600b7ce76bdfaeba"}]},"branch":"refs/heads/master"},"fe1ae77c33170927fe5629ca10aa0d1e165b85c2":{"kind":"REWORK","_number":5,"created":"2026-09-10 19:01:51.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/90/1003890/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/1003890/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/1003890/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/1003890/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/1003890/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/1003890/5"}}},"commit":{"parents":[{"commit":"396cf55464a006beb4b6c401f3f2e41f2323290c","subject":"utils/checksum: Prefer anycrc for CRC32","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/396cf55464a006beb4b6c401f3f2e41f2323290c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-09-03 22:11:01.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-09-10 19:01:45.000000000","tz":-420},"subject":"checksum: Add ISA-L implementation for CRC32","message":"checksum: Add ISA-L implementation for CRC32\n\nOn a production node, ISA-L delivered 15.3-34.1 GiB/s across\n4 MiB-4 GiB inputs for one-shot and 64 KiB incremental CRC32\ncalculations -- 7.6-16.9 times faster than anycrc and 12.5-27.9\ntimes faster than zlib -- while producing identical checksums.\n\nChange-Id: Ic88dc8a0e9402df2fb7254c8b2b0825b547deb6d\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe1ae77c33170927fe5629ca10aa0d1e165b85c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe1ae77c33170927fe5629ca10aa0d1e165b85c2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
