)]}'
{"id":"openstack%2Fswift~1003892","triplet_id":"openstack%2Fswift~master~I5c6da351fdcb4cbefeff19f59cdde28180a7ca24","project":"openstack/swift","branch":"master","attention_set":{"39146":{"account":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"last_update":"2026-09-04 00:06:47.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I5c6da351fdcb4cbefeff19f59cdde28180a7ca24","subject":"s3api: Make checksum combine yielding configurable","status":"NEW","created":"2026-09-03 22:57:58.000000000","updated":"2026-09-04 00:06:47.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a96e91c9addaf891cf25c5effcdd5df9ea48c92b","_number":1003892,"virtual_id_number":1003892,"owner":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-04 00:06:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-04 00:01:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-04 00:06:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c3b427308b1c26f0da6ebb9d8a92d9f06b30f44a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-09-03 22:57:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8ee99c629cd526f2c92ac49cb3792154ab29ffad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 00:01:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20a43c1ea24c45a8a3fdbf2942a5a6dc\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/76dd66282fcb45639510c8f1524483af : SUCCESS in 1h 00m 37s","accounts_in_message":[],"_revision_number":1},{"id":"a96e91c9addaf891cf25c5effcdd5df9ea48c92b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 00:06:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/263253c90ac343a7afce4be211a515cb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc233fe810a241bb8213a303542116de : SUCCESS in 10m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/30f39e97eede43d0aab9c7a9f8ec6a09 : SUCCESS in 9m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/75885c4cab584b73ae93431c2465208c : SUCCESS in 1h 00m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/976a36e51dd3440ca3d367247bcac640 : SUCCESS in 1h 02m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a87c5fff0b5d4da487046a63327033eb : SUCCESS in 47m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6802a9e37f384a28884b38a20e63156c : SUCCESS in 1h 03m 32s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/223b5b3fced946aaa9b924c7dd29ae29 : SUCCESS in 19m 04s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1ed0af6e6b1c4cafa4d1f7e487e9ee8d : SUCCESS in 18m 54s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c894e7aa12f14c489086db04649d21da : SUCCESS in 21m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ec8a072a0c3b4447823a79a663d68336 : SUCCESS in 6m 44s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/64aa65086366422a80d82c4920335d23 : FAILURE in 11m 47s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d332a2842c4b4f939cf50b7bc9921cd1 : SUCCESS in 23m 42s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/02c06896b5174d07b969534d823ceac8 : SUCCESS in 32m 49s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/d94058c6e1d9485884d72a33f6111138 : SUCCESS in 19m 52s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0d7b14a49dcb4f3f8720e558811ec187 : SUCCESS in 14m 09s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/070e5b6bfc5241c9b2285e988cb0345f : SUCCESS in 14m 42s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/133b3679d96c450699c574f1e454692c : SUCCESS in 15m 15s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/f5717e8f1f3442899d26b2b50460de33 : SUCCESS in 10m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0f84956c54c84a6e830c30731f24d0b5 : SUCCESS in 8m 15s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0c38488fbaa14b7d8706b982f4e99eb3 : FAILURE in 30m 26s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/b69269c08c304dd7999e855a123c12b2 : SUCCESS in 7m 07s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/437bade95d9046fe930f9f28577648b3 : SUCCESS in 39m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/60f427e4f0294660adb6b16289b5a992 : SUCCESS in 53m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/18586384342f46a59604b8addec01c28 : SUCCESS in 59m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/423170e5c5f34c72bda704a6dde89233 : FAILURE in 10m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8aff93f7e6024466b5cdd953acc2ff05 : SUCCESS in 3m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/eaaea3f7c0924591b5090fe8098cb3ca : SUCCESS in 10m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7d08d159e7c34ff18b2d27c168d48b06 : SUCCESS in 30m 23s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4eeadef4cf593122f365c255bfebab9c70f82e24","revisions":{"4eeadef4cf593122f365c255bfebab9c70f82e24":{"kind":"REWORK","_number":1,"created":"2026-09-03 22:57:58.000000000","uploader":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"ref":"refs/changes/92/1003892/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/92/1003892/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/1003892/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/1003892/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/1003892/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/92/1003892/1"}}},"commit":{"parents":[{"commit":"928014eb6bcc49b95df619f406e35f4851fdac18","subject":"checksum: Add Python fallback for combine methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/928014eb6bcc49b95df619f406e35f4851fdac18"}]}],"author":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-09-03 22:55:01.000000000","tz":-240},"committer":{"name":"Nathaniel Martes","email":"nmartes@NVIDIA.com","date":"2026-09-03 22:56:11.000000000","tz":-240},"subject":"s3api: Make checksum combine yielding configurable","message":"s3api: Make checksum combine yielding configurable\n\nAdd checksum_combine_yield_frequency, defaulting to 250,\nto control how often FULL_OBJECT MPU checksum calculation yields.\n\nChange-Id: I5c6da351fdcb4cbefeff19f59cdde28180a7ca24\nAssisted-By: Codex:gpt-5.5\nSigned-off-by: Nathaniel Martes \u003cnmartes@NVIDIA.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4eeadef4cf593122f365c255bfebab9c70f82e24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4eeadef4cf593122f365c255bfebab9c70f82e24"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
