)]}'
{"id":"openstack%2Fswift~1003908","triplet_id":"openstack%2Fswift~master~I765ceb59c85429d1c8d4c8f2317a68e2c63308bf","project":"openstack/swift","branch":"master","topic":"s3-checksum-persistence","attention_set":{},"removed_from_attention_set":{"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2026-09-05 01:21:56.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I765ceb59c85429d1c8d4c8f2317a68e2c63308bf","subject":"sq? reuse existing tests in test_multi_upload.py","status":"ABANDONED","created":"2026-09-04 04:44:42.000000000","updated":"2026-09-05 01:21:56.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7c57d9316d7b7c6d296c34e5a6349cbfec5c2c0f","_number":1003908,"virtual_id_number":1003908,"owner":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-04 05:44:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"all":[{"value":1,"date":"2026-09-04 20:35:02.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-04 05:35:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-04 05:44:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-04 20:35:02.000000000","updated_by":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"reviewer":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"state":"REVIEWER"}],"messages":[{"id":"bccfc605d9f4ca7479a137764bb9df1e3c16c35f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-09-04 04:44:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1b9b2f42fd0cf1e2a85b24f864d992afd87c2d61","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 05:35:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5521eaaccc44465db6df3a02a5226746\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/10ec1ca2027c454dba478d07b1f75944 : SUCCESS in 48m 38s","accounts_in_message":[],"_revision_number":1},{"id":"639467625a54c6b4ec4b483bd8a250842516be2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 05:44:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10409d589b0346f4ae977079b8a0ab92\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8a540177cc444d59b76095dd9a37c2d : SUCCESS in 10m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/00a91fc6f0044b63ae65dcd45efabf88 : SUCCESS in 9m 31s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9f94d8fca5dd43609029b1ad32842f82 : SUCCESS in 18m 19s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a8f69d9317c34a0e9b699e30d0eb87b3 : SUCCESS in 20m 38s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b12605eb155341d5805072e4688ab38f : SUCCESS in 20m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6c4c264d866443bea842e8ba724653c2 : SUCCESS in 6m 46s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6c88f3fd5d1d405c93e2a182aac5d6c1 : SUCCESS in 24m 33s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/1e33298170774189807d61654ff22548 : SUCCESS in 17m 51s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/ab1bcb4e88044e6bbe7fe603e21331ac : SUCCESS in 33m 31s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/a85c88e764d54562813b48e056affd9f : SUCCESS in 10m 19s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4a5b8f2e097946b49ecd7deebcbffcbb : SUCCESS in 13m 21s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/2d5cc3cbfb1a401582d418ba4fac54e5 : SUCCESS in 15m 12s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/2d08309dee32499693524caef8a78255 : SUCCESS in 13m 10s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/7562a22393b3460ba9e5833de48242b4 : SUCCESS in 16m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/94d103684f1b4143a91988a005976f83 : SUCCESS in 53m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bc346f6bd13d4f858b7e4d679af8d0d4 : SUCCESS in 22m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1a9af299c49b49ea9200d053e7f21543 : SUCCESS in 28m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0fb7c8841ab4945849f4997df4e7f19 : SUCCESS in 3m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/251c6d2015e148eb819f7feb905b9a7b : SUCCESS in 19m 40s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"947cf3265007051384da385f66707a153b42e3e9","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-09-04 20:35:02.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2df30cae123ad4e88f25a98d4af1e082af55d849","author":{"_account_id":39146,"name":"Nathaniel Martes","display_name":"Nate Martes","email":"natemartes09@gmail.com","username":"nmartes"},"date":"2026-09-04 20:55:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7c57d9316d7b7c6d296c34e5a6349cbfec5c2c0f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-09-05 01:21:56.000000000","message":"Abandoned\n\nsquashed into https://review.opendev.org/c/openstack/swift/+/1003004","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"93f3b51b383a2cf69fb76b0ba2f1e65b378c7bc3","revisions":{"93f3b51b383a2cf69fb76b0ba2f1e65b378c7bc3":{"kind":"REWORK","_number":1,"created":"2026-09-04 04:44:42.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/08/1003908/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/08/1003908/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/1003908/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/1003908/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/1003908/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/08/1003908/1"}}},"commit":{"parents":[{"commit":"7d673e3b7d4f21b1f9054cb233fb427cfc1e0aea","subject":"s3api: add x-amz-checksum-type label to checksum metric","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7d673e3b7d4f21b1f9054cb233fb427cfc1e0aea"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2026-09-04 04:43:54.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2026-09-04 04:43:54.000000000","tz":-420},"subject":"sq? reuse existing tests in test_multi_upload.py","message":"sq? reuse existing tests in test_multi_upload.py\n\nChange-Id: I765ceb59c85429d1c8d4c8f2317a68e2c63308bf\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/93f3b51b383a2cf69fb76b0ba2f1e65b378c7bc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/93f3b51b383a2cf69fb76b0ba2f1e65b378c7bc3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
