)]}'
{"id":"openstack%2Fswift~1004324","triplet_id":"openstack%2Fswift~master~Ia577a4c2b609af52859235c83290ed2683bdc423","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-09-19 19:18:59.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-09-19 19:18:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia577a4c2b609af52859235c83290ed2683bdc423","subject":"Fix typos","status":"MERGED","created":"2026-09-06 19:39:55.000000000","updated":"2026-09-19 19:20:10.000000000","submitted":"2026-09-19 19:18:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":16,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004324","meta_rev_id":"496ac76eef74281e509a0a31562bd9bd4e4b5691","_number":1004324,"virtual_id_number":1004324,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-19 19:18:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-18 21:38:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-09-18 21:38:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-06 19:42:44.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2026-09-06 20:32:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-06 20:39:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-19 17:01:49.000000000","updated_by":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"CC"}],"messages":[{"id":"3396da6bc675cb9b20cdf13997ea65da6950b679","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-06 19:39:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cef75a118ff53a1c838d39e5cfc770b41dfdc796","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-06 19:42:39.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"1e014330a2174656175732a8ee73c3e8711a7c3e","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-06 19:42:44.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"6ad35f335d35f1626e046730d0742cd8b2071db6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-06 20:32:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76467251095d4cd481c1eb342c3ad22f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/967e743e45c24cf498e821f7ad0f013a : SUCCESS in 48m 33s","accounts_in_message":[],"_revision_number":2},{"id":"d4232b94601d3374dac496e4adf95e7c5b963883","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-06 20:39:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5356a981f640488791b7bdb067900858\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/563936b850b24eaca11d333fc078ecd0 : SUCCESS in 10m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/31fd208594234431a5f9471acf9c19be : SUCCESS in 9m 02s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9a7c566961fe40008e8472551d3181df : SUCCESS in 19m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f9e652d1a3bb44f4b11bc760f7e99ef8 : SUCCESS in 20m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e14c440930354df39aa86e26531729b8 : SUCCESS in 20m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ba56bd59b9e74552a432a59e2f029819 : SUCCESS in 6m 25s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/16ab7b18e3ea45829cbbd17bb85d3f56 : SUCCESS in 20m 36s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/8905efacafd041069d5e7038f5f6b111 : SUCCESS in 9m 22s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/fbae161d553d49a2b1684950810440c9 : SUCCESS in 33m 06s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/dd24f6d7f2b447c7bcb26ff3f7d6c8b0 : SUCCESS in 21m 09s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/401db4cca66e4b63821eccbc110987b9 : SUCCESS in 15m 10s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/58b9bcd9a239467bb9d966bb0c146838 : SUCCESS in 14m 48s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/fbac1bf8b9d54072a810589cd66aa82f : SUCCESS in 15m 23s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/1fa1a36bcb93420dbe774c2ace8e44cb : SUCCESS in 16m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3effab0edb7447c18a345bf08f913185 : SUCCESS in 53m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/faa1cccd2baf4fe1b1c42792b433ca4d : SUCCESS in 49m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eb42b930baf9465083171c7b9f1aa53a : SUCCESS in 19m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f539a36118b04e24b960534ccdc29c1a : SUCCESS in 4m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/907dfc49986b440092418df61b8282f3 : SUCCESS in 18m 48s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"88db9513d99ef596a96c6ba009b25c10c4d04c53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-06 22:08:22.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"ec98a83fe27ba2285e095b0b17c3afd42a56354f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-06 22:09:03.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"fc9f6bce9d085a62b05ee1866027e41a20bd725f","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-06 22:22:11.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":4},{"id":"952ca47fe182fbdabff923a39fb266c34f7cb9ca","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-06 22:24:49.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"c0a8ddf80088f121f46c19a65c83c52678320eb5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-06 22:51:24.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"7be696f9320377bc3ddd893af2b3ca10003ca506","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-06 23:02:45.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"dfaf2d9c62dbaa27f403ad8d2115d3727f1278c4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-06 23:07:22.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"e6cf5e75c93c45297d3d3770f37a1520c04d5015","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-06 23:13:51.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"2a2c6708556452fb91a4c4141ec7be96c8308453","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-06 23:16:54.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"04006686f930ee894dd6c39ab334605e14c8658e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-06 23:18:58.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"2352500d165fa129e6917c40b1bb5a2a791f6e92","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-06 23:42:21.000000000","message":"Patch Set 12: Published edit on patch set 11.","accounts_in_message":[],"_revision_number":12},{"id":"12d83668b220c048b6160806f82208fa1c7eccf8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-07 00:08:43.000000000","message":"Patch Set 13: Published edit on patch set 12.","accounts_in_message":[],"_revision_number":13},{"id":"cc2661842a2d752e5d9e7d46bc1c1f66da4fa5fa","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-07 00:12:47.000000000","message":"Patch Set 14: Published edit on patch set 13.","accounts_in_message":[],"_revision_number":14},{"id":"a52e5a18abedcce8425a21a6e8be217e034874fb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-07 00:15:13.000000000","message":"Patch Set 15: Published edit on patch set 14.","accounts_in_message":[],"_revision_number":15},{"id":"fc12a10221810c34e22586669aabfc9bb74eff26","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-07 00:16:08.000000000","message":"Patch Set 16: Published edit on patch set 15.","accounts_in_message":[],"_revision_number":16},{"id":"a753e13537dfc0c91ebc01ef998bf3be994e875f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-07 00:17:39.000000000","message":"Patch Set 17: Published edit on patch set 16.","accounts_in_message":[],"_revision_number":17},{"id":"0b0b66e956de8a0570c50860639d5f591a1235d4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-07 00:26:02.000000000","message":"Patch Set 18: Published edit on patch set 17.","accounts_in_message":[],"_revision_number":18},{"id":"00635b222d532c77cfe9fad8874d129a3fe8c8f6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-07 00:49:36.000000000","message":"Patch Set 19: Published edit on patch set 18.","accounts_in_message":[],"_revision_number":19},{"id":"6b636ec2dc5d037f7cf186e96ae7fe3520842ad9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-07 00:51:04.000000000","message":"Patch Set 20: Published edit on patch set 19.","accounts_in_message":[],"_revision_number":20},{"id":"a7e7ac73f4db798780cb110a9c857206a6ed6ade","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-07 00:52:58.000000000","message":"Patch Set 21: Published edit on patch set 20.","accounts_in_message":[],"_revision_number":21},{"id":"ec4040e88694ca435ac8f8396466d60595827dd5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-07 00:57:47.000000000","message":"Patch Set 22: Published edit on patch set 21.","accounts_in_message":[],"_revision_number":22},{"id":"8bec8f79e577b3913c9ed0bad657c5346cc78b06","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-07 00:59:00.000000000","message":"Patch Set 23: Published edit on patch set 22.","accounts_in_message":[],"_revision_number":23},{"id":"67812afb369d3680332d5ee7a4a43eeb04bd6484","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-07 01:04:31.000000000","message":"Patch Set 24: Published edit on patch set 23.","accounts_in_message":[],"_revision_number":24},{"id":"6a7e7a4cfe62c2d9617b25dd935c324626b046a3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-07 01:14:06.000000000","message":"Patch Set 25: Published edit on patch set 24.","accounts_in_message":[],"_revision_number":25},{"id":"56a6c433d425d08c4a4c11a189842a2c66c02e11","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-07 01:16:01.000000000","message":"Patch Set 26: Published edit on patch set 25.","accounts_in_message":[],"_revision_number":26},{"id":"2572514c206323d6b1421544823a8449c50b5d50","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-07 01:17:11.000000000","message":"Patch Set 27: Published edit on patch set 26.","accounts_in_message":[],"_revision_number":27},{"id":"4d054a3d71ad2862a13b86924479a7a883e98204","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-07 01:18:29.000000000","message":"Patch Set 28: Published edit on patch set 27.","accounts_in_message":[],"_revision_number":28},{"id":"1f4777c44947c46b0b6fa8e919dabbefad47afee","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-07 01:21:19.000000000","message":"Patch Set 29: Published edit on patch set 28.","accounts_in_message":[],"_revision_number":29},{"id":"a577e49b05afeabce190128dae0f4b376e58444a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-07 01:25:31.000000000","message":"Patch Set 30: Published edit on patch set 29.","accounts_in_message":[],"_revision_number":30},{"id":"5d8650e3f79723778784042f1f337dbc4f0a180f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-07 01:29:13.000000000","message":"Patch Set 31: Published edit on patch set 30.","accounts_in_message":[],"_revision_number":31},{"id":"ff2e096d24edb5da5a08f8b65a57ab5e058d42de","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-07 01:34:22.000000000","message":"Patch Set 32: Published edit on patch set 31.","accounts_in_message":[],"_revision_number":32},{"id":"4e20952fc25fce0a19f28aeff37501010a0a52ad","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-07 01:48:19.000000000","message":"Patch Set 33: Published edit on patch set 32.","accounts_in_message":[],"_revision_number":33},{"id":"eb2439feeedf037ff29567068a23defcfa10afc8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-07 01:53:19.000000000","message":"Patch Set 34: Published edit on patch set 33.","accounts_in_message":[],"_revision_number":34},{"id":"7d492ed8fea8513c41522a166d51ce52c8af62c0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-07 01:58:42.000000000","message":"Patch Set 35: Published edit on patch set 34.","accounts_in_message":[],"_revision_number":35},{"id":"759a2658811675a6fe5f724f579ea0a7ea99031c","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-07 01:58:47.000000000","message":"Patch Set 35:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":35},{"id":"834cb29fcc0f80707752fc29d354e5fc7763c2ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 02:58:33.000000000","message":"Patch Set 35:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2716eaba91614b23adc5e275f3714b3f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5f8ba3c6a47340d98ccd0258c925da29 : FAILURE in 58m 06s","accounts_in_message":[],"_revision_number":35},{"id":"430a16a921a6b35b8ea6e5a424ee1638592e05a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 03:03:46.000000000","message":"Patch Set 35: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/946e9cb0e09346919aa8eeccaa91ae60\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfc7db5b1e3d4f719ea552ef2bbab4d8 : SUCCESS in 10m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/08981998f96046279533c53e5b419708 : SUCCESS in 10m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/76ff1b8321324e16a53f811598b92ad8 : SUCCESS in 47m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/aeed49027f6547a78d48e13910280d82 : SUCCESS in 58m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/942ef1d9c434412da2be3a3a6e9ddbfd : SUCCESS in 36m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/de3f01018be043179321938df1f24a44 : SUCCESS in 56m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/129d44a35dca47588b07b4e5fc137abb : SUCCESS in 20m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/61d58155fd834a438420c859f2ec219a : SUCCESS in 21m 24s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/765ea3e1dc2d403ca2e3d3f75ece9430 : SUCCESS in 20m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8bea588fa84a4e758ec560b3c3e41b03 : SUCCESS in 6m 07s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/794516d99b7347589788e23eed9a5d08 : FAILURE in 22m 55s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/81212d3466ca45daacc10f4357141ad2 : FAILURE in 14m 32s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/6efca9e45a55412f806214a261485542 : FAILURE in 30m 19s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/8bddcf2b4e5a44918e605db7da7f974c : FAILURE in 15m 10s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5ca12ada9e5e4bfea159298499e4a306 : SUCCESS in 15m 42s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/98be2a21984d413683660e5c618e4d90 : SUCCESS in 16m 08s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/891c42fcbe4042daa3a1e24e38baaa5b : SUCCESS in 19m 33s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/2e2c538ff7f4480488bf8b91b4708c56 : SUCCESS in 18m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5bc38c4f3d77412f9a35cc8dd94378ca : SUCCESS in 7m 28s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9383aa98fb6b49a8948b356dc28c82b9 : FAILURE in 29m 05s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/c356c251208e444b8d487999cb77a784 : SUCCESS in 9m 33s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/276f1101d4f54c90a7899f1801160630 : SUCCESS in 1h 02m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/91a42bb3ec0e41db961ea71d9d348f0b : SUCCESS in 57m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4a43649335f24c43b18b8c6fa2df60b5 : SUCCESS in 40m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/720730364e16442999c959b5bba955e1 : FAILURE in 24m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76a336e4111c48a1ad1c6a2430f831f6 : FAILURE in 5m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a10d84a665244142b09fc94ccf413c65 : SUCCESS in 20m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0825d5a750794880b7bb82b66eb240a7 : SUCCESS in 41m 53s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":35},{"id":"e20cdc92868034b6cc8a6cca02bc1cb96dbf956e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-07 11:40:53.000000000","message":"Patch Set 36: Published edit on patch set 35.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":36},{"id":"8f630aed3ce36eaa23242e064ec5b7994ede75a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-07 11:41:23.000000000","message":"Patch Set 37: Published edit on patch set 36.","accounts_in_message":[],"_revision_number":37},{"id":"cad9bd7cc06ca3027fcde7e9474a880a5b35f47e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 12:32:49.000000000","message":"Patch Set 37:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f031e8a0f3b343e1a520e1935a748ca6\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8bd991329332437184350ac6b8408aca : FAILURE in 50m 17s","accounts_in_message":[],"_revision_number":37},{"id":"f9bba9ffba740860dac07384e27bfb2ee6553ae1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 12:49:53.000000000","message":"Patch Set 37: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/532be5fde6e34e5e95321191c5a815e9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36389283cd994837b843224f6b8bfa87 : SUCCESS in 10m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4917dda140354750b5dff3e456ad2272 : SUCCESS in 10m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/467f030dbfe840c69d0f552bb695b10f : SUCCESS in 1h 00m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0f2cccf568e44f5ea635dfa5211853af : SUCCESS in 50m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2f08c92ff1f14d2ba63dd9b5ae8e6ac0 : SUCCESS in 53m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9ac1dc46c1574dc4906cbf57e1263732 : SUCCESS in 1h 02m 31s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4ef05635751f424a8359b7dc3b39491b : SUCCESS in 20m 10s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8868e6d90a4947a7a8f8c2754bfd5cfc : SUCCESS in 17m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6108c9d1f1a440bda7aeca9f754e4881 : SUCCESS in 21m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1e422ecab64343a995a1a41c727cecb8 : SUCCESS in 7m 05s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/911e8000b7894f5fa3ce37257dc4b1b8 : FAILURE in 24m 53s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ef67f82265424e278624b382ff3100dd : FAILURE in 22m 34s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/9d9358d9a8c0429f864d7e0f72f09ee1 : FAILURE in 32m 22s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/8aaa83a1599c42f1b33c72179d1f2612 : FAILURE in 21m 39s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9a3e7f644b70440d970b026fcbda66ef : SUCCESS in 14m 40s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3bff0776122e4a35952a8f3eb0fa4286 : SUCCESS in 13m 10s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0abcaf0c8c10438f9f54dda985503458 : SUCCESS in 16m 24s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/9229e2fa37404942a9e5210963ac9ffb : SUCCESS in 17m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2db578ad72fd433f8a4eeda356f3dda6 : SUCCESS in 7m 13s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4353fd554cdf4e9a93c837b1f39dd1ac : FAILURE in 30m 02s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/99891764150b4cf9a7062f6acc628291 : SUCCESS in 9m 01s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/efc42fde8b0148be88f2bdc5e814e186 : SUCCESS in 1h 01m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1234cfcee46941b5b1f8d7d88fb58935 : SUCCESS in 50m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b598fb0615a74da69182517ba118edde : SUCCESS in 39m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5714243689da4d71ade8fb4582e73450 : FAILURE in 28m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5c6ea935c774df08bb887cabcc824a5 : SUCCESS in 5m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2033edd777ed42768edc644855805c12 : SUCCESS in 19m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ba0332cf1e294045a6fd09d27ad1ae07 : SUCCESS in 58m 57s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":37},{"id":"0b15748a6e39d48925d0612c36b51e366627a7f6","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-07 13:12:03.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":37},{"id":"25a1827ea2c991a03161e18d0377221935695b2f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-07 15:50:26.000000000","message":"Patch Set 38: Published edit on patch set 37.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":38},{"id":"6cc3856a951143c55f83146bb9edb8f70fb0cff7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 16:41:53.000000000","message":"Patch Set 38:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d9b4b250803490cba69b39d6cc248c9\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b751f67f73a947b28fc7074a1ec6d346 : SUCCESS in 49m 47s","accounts_in_message":[],"_revision_number":38},{"id":"6e96a363c626c576f67a76ea25a3ab5bdd43fffd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 17:09:17.000000000","message":"Patch Set 38: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6886da4cc8144a08dac8391110b500d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78494e475716431d83403bf566f14d6b : SUCCESS in 5m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b776e3d630d64e38be29c849368f133b : SUCCESS in 3m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/29f26453a8da470e8d8554119128f3cf : SUCCESS in 1h 12m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9d933f6ebd0444a99e928c6f364264a6 : SUCCESS in 1h 03m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ccf08d083ad14175b17b872a393a33ea : SUCCESS in 1h 05m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f9b621f576be4777a7b59e7a93948c74 : SUCCESS in 58m 04s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/17db7a1f90e147e098b02aa2e2f5df45 : SUCCESS in 20m 45s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6df1737bbd1b45c4bc43d1783b421165 : SUCCESS in 20m 33s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/17de87a8c789466aa28d126400c72aae : SUCCESS in 21m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c45fc44c33be4d328de8fb017bca9f18 : SUCCESS in 6m 23s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/24c592ad8e5c4c3e9dd0ffad23c20f78 : SUCCESS in 21m 07s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/6ae4b0e1821141828a5b14701456738a : SUCCESS in 21m 12s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/398eeb22331f4b7aa3e87354dd743739 : SUCCESS in 25m 26s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/5ff3939f14f044cb9b9849ba001a38b9 : SUCCESS in 11m 15s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3e51524e061249baa1a97e505822c9f2 : SUCCESS in 8m 12s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/91213d8ba24544d89e903c136c0252e2 : SUCCESS in 14m 23s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4e2d80dbad204adc99a803939e8e68ec : SUCCESS in 15m 56s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/c772ebaaf234459fb71756c7174f7b15 : SUCCESS in 14m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/364337a065d74bac8519be950aa931bb : SUCCESS in 8m 27s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/663ad974e62544e69fe4924971e8b043 : FAILURE in 25m 06s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/5b77baf2669343c0bc263ecc6c5ce2ed : SUCCESS in 13m 15s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2530a10ede154da9b573971b03c629fa : SUCCESS in 55m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/21938a7a5ee74f4885cfefb09144cefb : SUCCESS in 28m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bb5a29d51de84c06894b60fc7ed3fa83 : SUCCESS in 28m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e726092c09e4dc8b89ebe8c542f2727 : SUCCESS in 29m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3864cf34fbe84276b7d1a6c8b26df263 : SUCCESS in 3m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/36dd231f27334399bed808a64a90bbf5 : SUCCESS in 11m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2a7686339eac4709b136594b53a61feb : SUCCESS in 52m 15s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":38},{"id":"b859448709a2d850fe93fe9b1acd521d957fadf5","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-07 17:12:38.000000000","message":"Patch Set 38:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":38},{"id":"fecd1ff5157b7f77a7e719b7b4051e54c18838dc","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-09 18:30:46.000000000","message":"Patch Set 38: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":38},{"id":"69d50572bbd15fb3b5e22c9db5b0b1272671709b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-18 21:38:14.000000000","message":"Uploaded patch set 39.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":39},{"id":"72093a8b7ec62775b0ef3f60c2a3f9c6ce23fa39","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-18 21:38:29.000000000","message":"Patch Set 39: Code-Review+2 Workflow+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":39},{"id":"9509dbd6cc369e8eaab48f240eb87f2de2c940e4","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-18 21:57:30.000000000","message":"Patch Set 39:\n\n(1 comment)","accounts_in_message":[],"_revision_number":39},{"id":"57ec381838dfb6bd8e53376182376c08f870b514","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 22:46:26.000000000","message":"Patch Set 39:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6dee4ac4d621499d8efe4d5d711df32a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bc70ef52569449108f8a85a87dbbcdb7 : SUCCESS in 1h 06m 22s","accounts_in_message":[],"_revision_number":39},{"id":"7631a4e837d9f5c254286f2fca71a383389cfa70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 22:50:24.000000000","message":"Patch Set 39: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c84f3b85bb0d469189b80703621a0624\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60f9347fa6e646df8d5871823a4320eb : SUCCESS in 10m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1dfdd5ba247b47d68eba89228905911a : SUCCESS in 10m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/58709914a5034a73821b8416dccd6988 : SUCCESS in 1h 08m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/15396432289748d1baa6f890642e154f : SUCCESS in 1h 05m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0502873c2a174286b76835bbaaa28532 : SUCCESS in 54m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7cd172b560044055be1c6faadb70e564 : SUCCESS in 55m 26s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/44f9b2f480de4bffb3163c7eb6183a5d : SUCCESS in 10m 33s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a83989ac2edc48dab58c6a9a2b667019 : SUCCESS in 12m 07s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/de6a20d189ce44d8953e514fac3ffa64 : SUCCESS in 20m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/151da7a9d4a4490aa4678531a3bd98f2 : SUCCESS in 7m 04s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f25cd1e1514443c5aca32f38e7777821 : SUCCESS in 13m 47s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/a92deab6a3664637ae2727f78caecc4b : SUCCESS in 21m 11s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/6d14099193894c96b51a6d2b341ecefe : SUCCESS in 14m 20s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/eb157c9fdc8541ff82d5d03413f0e1ec : SUCCESS in 21m 32s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0ecf3d242da4494fbe10896b680c7886 : SUCCESS in 9m 23s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9e13063fe47449078ea777a407f007c1 : SUCCESS in 15m 03s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/872834d1893041aeb62d3874af1bdc2d : SUCCESS in 18m 22s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/35e718cd405a490f8c8687b7bc6235c0 : SUCCESS in 17m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e7cbac91a6694ab297b44011c4ff5393 : SUCCESS in 3m 31s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/54490643ba7c40d8874e05a75e44f351 : SUCCESS in 14m 20s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/513fec3089b24bdab7e30ba64b52ac84 : SUCCESS in 37m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/37a8883d6b5745d7ae5af6b825b7d28b : SUCCESS in 59m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cdf591015a4a4b1e9614b96af89a48cc : SUCCESS in 40m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b9c92253bb0d4754b2a4098c3fe9daf0 : SUCCESS in 29m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/576ed233ca154e8eb437da5df7c9973f : SUCCESS in 2m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1eb8b30d7bce4c769a5844c5572044c0 : SUCCESS in 26m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fca29a42a555448da37f1c728d898d6e : SUCCESS in 52m 07s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":39},{"id":"0c8e799e452fabaf7d3b32c032c54660cfde2d1c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 22:50:57.000000000","message":"Patch Set 39: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":39},{"id":"38e2cadcbcc373cf0aa85815fd95eccb240101c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-19 00:15:43.000000000","message":"Patch Set 39: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/786983f991844687ba5bc9de1289bca1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9909f640bb0489ab12caf0789ee729c : SUCCESS in 5m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e001b285374a40f1b2ddacf1a4b4c3d7 : SUCCESS in 7m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/56c9e78f838b471f8adb52eecff88308 : SUCCESS in 1h 02m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/db1abba8a3a94129a2b29d2416b66212 : SUCCESS in 1h 06m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8425addb700c4a9c8bf861eb500b49b4 : SUCCESS in 47m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ac53f6a9db964631aa216635b14030e2 : SUCCESS in 57m 22s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/06828eb47e924b2eae79c751192afda3 : SUCCESS in 7m 10s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b75c098259904f9ca87b36f17146c337 : SUCCESS in 18m 49s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/620af5f7005047f599af8bce042fac49 : RETRY_LIMIT\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/7aafbc301d7046ae9767cce94ef16124 : SUCCESS in 22m 20s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/839e5e9b286741a88199d42317463289 : SUCCESS in 16m 00s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3fbcf4dd4ac84061a5547509569c3b3b : SUCCESS in 15m 54s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/2af33b0a70d14a09b4fb74d3091324e6 : SUCCESS in 10m 27s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6b408ba1a5b3430d9e9345cd6e14e4ea : SUCCESS in 10m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/61d700515f1a4eae946605a3fbbf33ee : SUCCESS in 4m 16s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/7707a36214564301ad77a244a03c0d90 : SUCCESS in 7m 58s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7034f68d09594dafbe39a7f165e28dd5 : RETRY_LIMIT\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bd52353a814c40758e2d9c8264ae9447 : SUCCESS in 50m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/07205e0e075444dab9b51d9d65589bb7 : SUCCESS in 54m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cc0a3d24767346abadaf723ba72f8a1d : SUCCESS in 23m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f0e2f70fed847b3a14628ec25112341 : SUCCESS in 6m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5aacbd5b3e99485892bece5899bce401 : RETRY_LIMIT\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cbccb88e3ec3451aa4d753ee367b6533 : RETRY_LIMIT","accounts_in_message":[],"_revision_number":39},{"id":"c138643358f8dfd45c52a41fae451d277893cfb8","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-09-19 17:01:49.000000000","message":"Patch Set 39:\n\n(1 comment)","accounts_in_message":[],"_revision_number":39},{"id":"08d1b65d1287092c505f5db719a59ee1f738d510","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-19 17:02:24.000000000","message":"Patch Set 39:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c09855492a1540a3a4a80b28489e1fac\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/faff9ffd7294440c869c721d7f51e20e : RETRY_LIMIT","accounts_in_message":[],"_revision_number":39},{"id":"6a2a0ddc3a021cc08d0fb4b05fe380141b6ed06d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-19 18:12:39.000000000","message":"Patch Set 39: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1f18826a6e149028c810e6c2fbc268c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5db1cfb3163a4d8aadfd26d9cd0b2839 : SUCCESS in 8m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b971d3ac0d8546eeabc5a66021e4a290 : SUCCESS in 9m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/ed8a7dcb25dd4612bf486eae7f01a8e2 : SUCCESS in 1h 00m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d076ad494b4b4d1781cee7a2ab309b4d : SUCCESS in 1h 01m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3a08a16c094943a09ada7c970c3b34b9 : SUCCESS in 29m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a2be161bd37046a0b48cbe178d208217 : SUCCESS in 56m 15s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a33def42233748109af75232217effe5 : SUCCESS in 19m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/879d793e0a0f4013858afa555aa37d76 : SUCCESS in 10m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/695b456c1363410192c0c2140266531b : SUCCESS in 18m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5af0b09dac2d4ac18cebdeb43623e56b : SUCCESS in 5m 18s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b631a02854ab4de8a05f4fc7616b9495 : SUCCESS in 10m 40s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/1d822d47348a4f6195a504458dd70414 : SUCCESS in 16m 53s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/71554f027d564e41bbbcb1fb7385d954 : SUCCESS in 32m 44s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/d2258c8c96bf44f5a0f3875d926733ca : SUCCESS in 9m 11s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0d54524d176c43eea6a3d4ccfdd1254c : SUCCESS in 9m 17s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/85549f5332d74a18bb532462d5b224cf : SUCCESS in 16m 09s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c9bc1bc7e66a4daea88ebcffb16103a2 : SUCCESS in 21m 36s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/26a0347c3b124fc6b74ebfd2e1f07a8a : SUCCESS in 9m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/73ae5989f0bd449aa11cd50cc0394374 : SUCCESS in 5m 38s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e614067405cd497aa57aa0ba1aea7b51 : SUCCESS in 8m 19s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7cd13e4f8c964eab975bd80cf7e9de84 : SUCCESS in 1h 04m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b2ec9028d8af441ab607541e3cf5ef9b : SUCCESS in 32m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/15de75ca3d144985b3d5ae53d6cad3e8 : SUCCESS in 50m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1edbd788e0564059b311a9f99949ad66 : SUCCESS in 28m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/728bc3381eaa400b90bee4c6a2905917 : SUCCESS in 4m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c674008a3d194becb061fc30e6063f7c : SUCCESS in 27m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af1c39553e5d4982857f3fe6806c3595 : SUCCESS in 52m 46s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":39},{"id":"4dac00da4932dd2372f280f1b6a92fe90687258e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-19 18:13:23.000000000","message":"Patch Set 39: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":39},{"id":"4d74dd60b8d551c5a8930ce95580795c3b3c80e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-19 19:18:59.000000000","message":"Patch Set 39: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2aa241c6e9342d5876d12598d4c5078\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b0864ec99554cfa90a1d9e7ddd23355 : SUCCESS in 11m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b6dea8aadc414e70b83016ff5a12f215 : SUCCESS in 9m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/6b20ea9a6ec142428df390a90a9ca787 : SUCCESS in 1h 05m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/91c3ee7bc3cd4c95aebc556eabe7801e : SUCCESS in 46m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/84c10529c8e44c0cb513c2a7da4f9b22 : SUCCESS in 55m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/98baf3c600f94dbe82d56dcdfd5bca6d : SUCCESS in 1h 03m 20s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/12c685aa295940a79a9cc6b2ba7b19d0 : SUCCESS in 7m 24s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/29171b4818b246a29743f94bfc5854b5 : SUCCESS in 13m 03s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/f7d7378eea22490d960e7bd7cc9c1b83 : SUCCESS in 16m 57s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/2362d4b5c4b94852ac67e21826b86e92 : SUCCESS in 32m 53s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/28faa1de2eb64eb58c56015f221b34d7 : SUCCESS in 20m 02s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/cd70786a033d4d7aacb07240b67c9897 : SUCCESS in 14m 16s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/84dd602fd094406d925011562717885a : SUCCESS in 9m 00s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/afb0182a40f448ceb25b74c89c6fb092 : SUCCESS in 8m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/088672df06f64523946e97048686d120 : SUCCESS in 5m 40s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/31c815d24c9d41199ecd0e6a63cbaffb : SUCCESS in 11m 59s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/680b025579ec46ae88a72ceb46588208 : SUCCESS in 46m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/408292ece91047fcbd07f0a6a45cac7e : SUCCESS in 32m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c00046f156bf43d5bcdd05e16413c813 : SUCCESS in 51m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c971841491d340498ffa166c714b44bb : SUCCESS in 28m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76e785e042384def897cf844c179ddc6 : SUCCESS in 3m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e38cfca5662347a2bc90c05e43a07f04 : SUCCESS in 18m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ffcf69bc775e4ad983b06f3e33f229b4 : SUCCESS in 29m 52s","accounts_in_message":[],"_revision_number":39},{"id":"9cfa068c9e11a9382ca1dcd153af27fa52e1bcdd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-19 19:18:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":39},{"id":"496ac76eef74281e509a0a31562bd9bd4e4b5691","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-19 19:20:10.000000000","message":"Patch Set 39:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/210308780c6b417aae670dd0df0b78a5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bfc2603d2144ece82557d8ae02ffdf2 : SUCCESS in 46s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/550ec25c8eb14eaba6ab9e193be4bb0e : SUCCESS in 55s (non-voting)","accounts_in_message":[],"_revision_number":39}],"current_revision_number":39,"current_revision":"39ba31f469b2f51679ddf82dafd428944fd3fef0","revisions":{"61fd050b744a066ed3416b2f9cf3170496f799a7":{"kind":"REWORK","_number":1,"created":"2026-09-06 19:39:55.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/24/1004324/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/1004324/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/1004324/1"}}},"commit":{"parents":[{"commit":"87834c289e96a6dd0be4635e506233d87c12c2a2","subject":"s3api: add x-amz-checksum-type label to checksum metric","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87834c289e96a6dd0be4635e506233d87c12c2a2"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 19:39:55.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 19:39:55.000000000","tz":0},"subject":"Fix typo","message":"Fix typo\n\nChange-Id: Ia577a4c2b609af52859235c83290ed2683bdc423\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/61fd050b744a066ed3416b2f9cf3170496f799a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/61fd050b744a066ed3416b2f9cf3170496f799a7"}]},"branch":"refs/heads/master","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"2bf31f6d0d59484093b90d1df80857325b668e9f":{"kind":"REWORK","_number":2,"created":"2026-09-06 19:42:39.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/24/1004324/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/1004324/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/1004324/2"}}},"commit":{"parents":[{"commit":"87834c289e96a6dd0be4635e506233d87c12c2a2","subject":"s3api: add x-amz-checksum-type label to checksum metric","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87834c289e96a6dd0be4635e506233d87c12c2a2"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 19:39:55.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 19:42:39.000000000","tz":0},"subject":"Fix typo","message":"Fix typo\n\nChange-Id: Ia577a4c2b609af52859235c83290ed2683bdc423\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2bf31f6d0d59484093b90d1df80857325b668e9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2bf31f6d0d59484093b90d1df80857325b668e9f"}]},"branch":"refs/heads/master"},"c25a22dee23a7a03c7609fcfcfcac69f98668f57":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-09-06 22:08:22.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/24/1004324/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/1004324/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/1004324/3"}}},"commit":{"parents":[{"commit":"87834c289e96a6dd0be4635e506233d87c12c2a2","subject":"s3api: add x-amz-checksum-type label to checksum metric","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87834c289e96a6dd0be4635e506233d87c12c2a2"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 19:39:55.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 22:08:22.000000000","tz":0},"subject":"Fix typos","message":"Fix typos\n\nChange-Id: Ia577a4c2b609af52859235c83290ed2683bdc423\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c25a22dee23a7a03c7609fcfcfcac69f98668f57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c25a22dee23a7a03c7609fcfcfcac69f98668f57"}]},"branch":"refs/heads/master","description":"Edit commit message"},"f6dba4f9099a71136dd4c92da4c37ccd55be7ce6":{"kind":"REWORK","_number":4,"created":"2026-09-06 22:09:03.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/24/1004324/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/1004324/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/1004324/4"}}},"commit":{"parents":[{"commit":"87834c289e96a6dd0be4635e506233d87c12c2a2","subject":"s3api: add x-amz-checksum-type label to checksum metric","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87834c289e96a6dd0be4635e506233d87c12c2a2"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 19:39:55.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 22:09:03.000000000","tz":0},"subject":"Fix typos","message":"Fix typos\n\nChange-Id: Ia577a4c2b609af52859235c83290ed2683bdc423\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f6dba4f9099a71136dd4c92da4c37ccd55be7ce6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f6dba4f9099a71136dd4c92da4c37ccd55be7ce6"}]},"branch":"refs/heads/master"},"f45fd5cc577cb2a885b0ed33c3f03fa3ae552eca":{"kind":"REWORK","_number":5,"created":"2026-09-06 22:24:49.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/24/1004324/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/1004324/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/1004324/5"}}},"commit":{"parents":[{"commit":"87834c289e96a6dd0be4635e506233d87c12c2a2","subject":"s3api: add x-amz-checksum-type label to checksum metric","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87834c289e96a6dd0be4635e506233d87c12c2a2"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 19:39:55.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 22:24:49.000000000","tz":0},"subject":"Fix typos","message":"Fix typos\n\nChange-Id: Ia577a4c2b609af52859235c83290ed2683bdc423\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f45fd5cc577cb2a885b0ed33c3f03fa3ae552eca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f45fd5cc577cb2a885b0ed33c3f03fa3ae552eca"}]},"branch":"refs/heads/master"},"30e89ddb32bd01ae60b74af1f35068708c369753":{"kind":"REWORK","_number":6,"created":"2026-09-06 22:51:24.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/24/1004324/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/1004324/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/1004324/6"}}},"commit":{"parents":[{"commit":"87834c289e96a6dd0be4635e506233d87c12c2a2","subject":"s3api: add x-amz-checksum-type label to checksum metric","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87834c289e96a6dd0be4635e506233d87c12c2a2"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 19:39:55.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 22:51:23.000000000","tz":0},"subject":"Fix typos","message":"Fix typos\n\nChange-Id: Ia577a4c2b609af52859235c83290ed2683bdc423\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/30e89ddb32bd01ae60b74af1f35068708c369753"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/30e89ddb32bd01ae60b74af1f35068708c369753"}]},"branch":"refs/heads/master"},"daa1338c8fb68db0c3cc859675a8ef31dc44914d":{"kind":"REWORK","_number":7,"created":"2026-09-06 23:02:45.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/24/1004324/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/1004324/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/1004324/7"}}},"commit":{"parents":[{"commit":"87834c289e96a6dd0be4635e506233d87c12c2a2","subject":"s3api: add x-amz-checksum-type label to checksum metric","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87834c289e96a6dd0be4635e506233d87c12c2a2"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 19:39:55.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 23:02:45.000000000","tz":0},"subject":"Fix typos","message":"Fix typos\n\nChange-Id: Ia577a4c2b609af52859235c83290ed2683bdc423\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/daa1338c8fb68db0c3cc859675a8ef31dc44914d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/daa1338c8fb68db0c3cc859675a8ef31dc44914d"}]},"branch":"refs/heads/master"},"6c56c31436e03f162e51c327903ce79b718de37d":{"kind":"REWORK","_number":8,"created":"2026-09-06 23:07:22.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/24/1004324/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/1004324/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/1004324/8"}}},"commit":{"parents":[{"commit":"87834c289e96a6dd0be4635e506233d87c12c2a2","subject":"s3api: add x-amz-checksum-type label to checksum metric","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87834c289e96a6dd0be4635e506233d87c12c2a2"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 19:39:55.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 23:07:22.000000000","tz":0},"subject":"Fix typos","message":"Fix typos\n\nChange-Id: Ia577a4c2b609af52859235c83290ed2683bdc423\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6c56c31436e03f162e51c327903ce79b718de37d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6c56c31436e03f162e51c327903ce79b718de37d"}]},"branch":"refs/heads/master"},"3ec93dbd437722ce55010f38f7a1fc8ae1d007c1":{"kind":"REWORK","_number":9,"created":"2026-09-06 23:13:51.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/24/1004324/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/1004324/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/1004324/9"}}},"commit":{"parents":[{"commit":"87834c289e96a6dd0be4635e506233d87c12c2a2","subject":"s3api: add x-amz-checksum-type label to checksum metric","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87834c289e96a6dd0be4635e506233d87c12c2a2"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 19:39:55.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 23:11:36.000000000","tz":0},"subject":"Fix typos","message":"Fix typos\n\nChange-Id: Ia577a4c2b609af52859235c83290ed2683bdc423\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ec93dbd437722ce55010f38f7a1fc8ae1d007c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ec93dbd437722ce55010f38f7a1fc8ae1d007c1"}]},"branch":"refs/heads/master"},"e51c67d0eaeb6123ffd39b7313dcd893dd86020a":{"kind":"REWORK","_number":10,"created":"2026-09-06 23:16:54.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/24/1004324/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/1004324/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/1004324/10"}}},"commit":{"parents":[{"commit":"87834c289e96a6dd0be4635e506233d87c12c2a2","subject":"s3api: add x-amz-checksum-type label to checksum metric","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87834c289e96a6dd0be4635e506233d87c12c2a2"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 19:39:55.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 23:16:54.000000000","tz":0},"subject":"Fix typos","message":"Fix typos\n\nChange-Id: Ia577a4c2b609af52859235c83290ed2683bdc423\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e51c67d0eaeb6123ffd39b7313dcd893dd86020a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e51c67d0eaeb6123ffd39b7313dcd893dd86020a"}]},"branch":"refs/heads/master"},"292859b51f25c8f073cce97d812e7c16768c9290":{"kind":"REWORK","_number":11,"created":"2026-09-06 23:18:58.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/24/1004324/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/1004324/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/1004324/11"}}},"commit":{"parents":[{"commit":"87834c289e96a6dd0be4635e506233d87c12c2a2","subject":"s3api: add x-amz-checksum-type label to checksum metric","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87834c289e96a6dd0be4635e506233d87c12c2a2"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 19:39:55.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 23:18:57.000000000","tz":0},"subject":"Fix typos","message":"Fix typos\n\nChange-Id: Ia577a4c2b609af52859235c83290ed2683bdc423\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/292859b51f25c8f073cce97d812e7c16768c9290"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/292859b51f25c8f073cce97d812e7c16768c9290"}]},"branch":"refs/heads/master"},"791970798f9c198e5b0371d54832a80cb6737a19":{"kind":"REWORK","_number":12,"created":"2026-09-06 23:42:21.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/24/1004324/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/1004324/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/1004324/12"}}},"commit":{"parents":[{"commit":"87834c289e96a6dd0be4635e506233d87c12c2a2","subject":"s3api: add x-amz-checksum-type label to checksum metric","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87834c289e96a6dd0be4635e506233d87c12c2a2"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 19:39:55.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 23:42:21.000000000","tz":0},"subject":"Fix typos","message":"Fix typos\n\nChange-Id: Ia577a4c2b609af52859235c83290ed2683bdc423\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/791970798f9c198e5b0371d54832a80cb6737a19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/791970798f9c198e5b0371d54832a80cb6737a19"}]},"branch":"refs/heads/master"},"8732fa44eead5f47aead030c16807ae73b88370c":{"kind":"REWORK","_number":13,"created":"2026-09-07 00:08:43.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/24/1004324/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/1004324/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/1004324/13"}}},"commit":{"parents":[{"commit":"87834c289e96a6dd0be4635e506233d87c12c2a2","subject":"s3api: add x-amz-checksum-type label to checksum metric","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87834c289e96a6dd0be4635e506233d87c12c2a2"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 19:39:55.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-07 00:07:03.000000000","tz":0},"subject":"Fix typos","message":"Fix typos\n\nChange-Id: Ia577a4c2b609af52859235c83290ed2683bdc423\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8732fa44eead5f47aead030c16807ae73b88370c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8732fa44eead5f47aead030c16807ae73b88370c"}]},"branch":"refs/heads/master"},"ef33329193b12408b90910aaa14d23f7a6c9c19d":{"kind":"REWORK","_number":14,"created":"2026-09-07 00:12:47.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/24/1004324/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/1004324/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/1004324/14"}}},"commit":{"parents":[{"commit":"87834c289e96a6dd0be4635e506233d87c12c2a2","subject":"s3api: add x-amz-checksum-type label to checksum metric","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87834c289e96a6dd0be4635e506233d87c12c2a2"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 19:39:55.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-07 00:11:49.000000000","tz":0},"subject":"Fix typos","message":"Fix typos\n\nChange-Id: Ia577a4c2b609af52859235c83290ed2683bdc423\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ef33329193b12408b90910aaa14d23f7a6c9c19d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ef33329193b12408b90910aaa14d23f7a6c9c19d"}]},"branch":"refs/heads/master"},"ce85cbc2cfe21b0dfb56e326d94909ab9bfe9391":{"kind":"REWORK","_number":15,"created":"2026-09-07 00:15:13.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/24/1004324/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/1004324/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/1004324/15"}}},"commit":{"parents":[{"commit":"87834c289e96a6dd0be4635e506233d87c12c2a2","subject":"s3api: add x-amz-checksum-type label to checksum metric","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87834c289e96a6dd0be4635e506233d87c12c2a2"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 19:39:55.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-07 00:15:13.000000000","tz":0},"subject":"Fix typos","message":"Fix typos\n\nChange-Id: Ia577a4c2b609af52859235c83290ed2683bdc423\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ce85cbc2cfe21b0dfb56e326d94909ab9bfe9391"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ce85cbc2cfe21b0dfb56e326d94909ab9bfe9391"}]},"branch":"refs/heads/master"},"33a4bad3a5ceba14d75484e0ad21f17d44cdeb7b":{"kind":"REWORK","_number":16,"created":"2026-09-07 00:16:08.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/24/1004324/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/1004324/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/1004324/16"}}},"commit":{"parents":[{"commit":"87834c289e96a6dd0be4635e506233d87c12c2a2","subject":"s3api: add x-amz-checksum-type label to checksum metric","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87834c289e96a6dd0be4635e506233d87c12c2a2"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 19:39:55.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-07 00:16:07.000000000","tz":0},"subject":"Fix typos","message":"Fix typos\n\nChange-Id: Ia577a4c2b609af52859235c83290ed2683bdc423\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/33a4bad3a5ceba14d75484e0ad21f17d44cdeb7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/33a4bad3a5ceba14d75484e0ad21f17d44cdeb7b"}]},"branch":"refs/heads/master"},"d98b390ff98e4f1398169298334aa6453cf4b4ec":{"kind":"REWORK","_number":17,"created":"2026-09-07 00:17:39.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/24/1004324/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/1004324/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/1004324/17"}}},"commit":{"parents":[{"commit":"87834c289e96a6dd0be4635e506233d87c12c2a2","subject":"s3api: add x-amz-checksum-type label to checksum metric","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87834c289e96a6dd0be4635e506233d87c12c2a2"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 19:39:55.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-07 00:17:38.000000000","tz":0},"subject":"Fix typos","message":"Fix typos\n\nChange-Id: Ia577a4c2b609af52859235c83290ed2683bdc423\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d98b390ff98e4f1398169298334aa6453cf4b4ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d98b390ff98e4f1398169298334aa6453cf4b4ec"}]},"branch":"refs/heads/master"},"866b5766beaa5efc72f45d09740606aa51698584":{"kind":"REWORK","_number":18,"created":"2026-09-07 00:26:02.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/24/1004324/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/1004324/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/1004324/18"}}},"commit":{"parents":[{"commit":"87834c289e96a6dd0be4635e506233d87c12c2a2","subject":"s3api: add x-amz-checksum-type label to checksum metric","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87834c289e96a6dd0be4635e506233d87c12c2a2"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 19:39:55.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-07 00:25:44.000000000","tz":0},"subject":"Fix typos","message":"Fix typos\n\nChange-Id: Ia577a4c2b609af52859235c83290ed2683bdc423\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/866b5766beaa5efc72f45d09740606aa51698584"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/866b5766beaa5efc72f45d09740606aa51698584"}]},"branch":"refs/heads/master"},"bf7df2f549409258891632883d9172de40386f3c":{"kind":"REWORK","_number":19,"created":"2026-09-07 00:49:36.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/24/1004324/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/1004324/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/1004324/19"}}},"commit":{"parents":[{"commit":"87834c289e96a6dd0be4635e506233d87c12c2a2","subject":"s3api: add x-amz-checksum-type label to checksum metric","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87834c289e96a6dd0be4635e506233d87c12c2a2"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 19:39:55.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-07 00:49:36.000000000","tz":0},"subject":"Fix typos","message":"Fix typos\n\nChange-Id: Ia577a4c2b609af52859235c83290ed2683bdc423\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bf7df2f549409258891632883d9172de40386f3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bf7df2f549409258891632883d9172de40386f3c"}]},"branch":"refs/heads/master"},"9f4312dd167fc01eafa6cdb918587a3eee88f8e4":{"kind":"REWORK","_number":20,"created":"2026-09-07 00:51:04.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/24/1004324/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/1004324/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/1004324/20"}}},"commit":{"parents":[{"commit":"87834c289e96a6dd0be4635e506233d87c12c2a2","subject":"s3api: add x-amz-checksum-type label to checksum metric","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87834c289e96a6dd0be4635e506233d87c12c2a2"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 19:39:55.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-07 00:51:04.000000000","tz":0},"subject":"Fix typos","message":"Fix typos\n\nChange-Id: Ia577a4c2b609af52859235c83290ed2683bdc423\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f4312dd167fc01eafa6cdb918587a3eee88f8e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f4312dd167fc01eafa6cdb918587a3eee88f8e4"}]},"branch":"refs/heads/master"},"bb2329bbd4d534ef0673f03f4ff2f72e7fcba2f5":{"kind":"REWORK","_number":21,"created":"2026-09-07 00:52:58.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/24/1004324/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/1004324/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/1004324/21"}}},"commit":{"parents":[{"commit":"87834c289e96a6dd0be4635e506233d87c12c2a2","subject":"s3api: add x-amz-checksum-type label to checksum metric","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87834c289e96a6dd0be4635e506233d87c12c2a2"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 19:39:55.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-07 00:52:54.000000000","tz":0},"subject":"Fix typos","message":"Fix typos\n\nChange-Id: Ia577a4c2b609af52859235c83290ed2683bdc423\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bb2329bbd4d534ef0673f03f4ff2f72e7fcba2f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bb2329bbd4d534ef0673f03f4ff2f72e7fcba2f5"}]},"branch":"refs/heads/master"},"7046f7b72a39b1e6ac207cbe3924402add45a002":{"kind":"REWORK","_number":22,"created":"2026-09-07 00:57:47.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/24/1004324/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/1004324/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/1004324/22"}}},"commit":{"parents":[{"commit":"87834c289e96a6dd0be4635e506233d87c12c2a2","subject":"s3api: add x-amz-checksum-type label to checksum metric","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87834c289e96a6dd0be4635e506233d87c12c2a2"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 19:39:55.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-07 00:57:47.000000000","tz":0},"subject":"Fix typos","message":"Fix typos\n\nChange-Id: Ia577a4c2b609af52859235c83290ed2683bdc423\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7046f7b72a39b1e6ac207cbe3924402add45a002"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7046f7b72a39b1e6ac207cbe3924402add45a002"}]},"branch":"refs/heads/master"},"afd606154a632431c8b7860661a2d65a7e77cf73":{"kind":"REWORK","_number":23,"created":"2026-09-07 00:59:00.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/24/1004324/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/1004324/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/1004324/23"}}},"commit":{"parents":[{"commit":"87834c289e96a6dd0be4635e506233d87c12c2a2","subject":"s3api: add x-amz-checksum-type label to checksum metric","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87834c289e96a6dd0be4635e506233d87c12c2a2"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 19:39:55.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-07 00:58:56.000000000","tz":0},"subject":"Fix typos","message":"Fix typos\n\nChange-Id: Ia577a4c2b609af52859235c83290ed2683bdc423\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/afd606154a632431c8b7860661a2d65a7e77cf73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/afd606154a632431c8b7860661a2d65a7e77cf73"}]},"branch":"refs/heads/master"},"11c980b643830c11be24bbb088a909a3f2f0f599":{"kind":"REWORK","_number":24,"created":"2026-09-07 01:04:31.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/24/1004324/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/1004324/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/1004324/24"}}},"commit":{"parents":[{"commit":"87834c289e96a6dd0be4635e506233d87c12c2a2","subject":"s3api: add x-amz-checksum-type label to checksum metric","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87834c289e96a6dd0be4635e506233d87c12c2a2"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 19:39:55.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-07 01:04:31.000000000","tz":0},"subject":"Fix typos","message":"Fix typos\n\nChange-Id: Ia577a4c2b609af52859235c83290ed2683bdc423\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/11c980b643830c11be24bbb088a909a3f2f0f599"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/11c980b643830c11be24bbb088a909a3f2f0f599"}]},"branch":"refs/heads/master"},"4b4255b2b614d5a855a9b447a09a88abb53ca0a2":{"kind":"REWORK","_number":25,"created":"2026-09-07 01:14:06.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/24/1004324/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/1004324/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/1004324/25"}}},"commit":{"parents":[{"commit":"87834c289e96a6dd0be4635e506233d87c12c2a2","subject":"s3api: add x-amz-checksum-type label to checksum metric","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87834c289e96a6dd0be4635e506233d87c12c2a2"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 19:39:55.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-07 01:14:06.000000000","tz":0},"subject":"Fix typos","message":"Fix typos\n\nChange-Id: Ia577a4c2b609af52859235c83290ed2683bdc423\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4b4255b2b614d5a855a9b447a09a88abb53ca0a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4b4255b2b614d5a855a9b447a09a88abb53ca0a2"}]},"branch":"refs/heads/master"},"c4d70ea369c529d96789eed8baac05b0e74f8078":{"kind":"REWORK","_number":26,"created":"2026-09-07 01:16:01.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/24/1004324/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/1004324/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/1004324/26"}}},"commit":{"parents":[{"commit":"87834c289e96a6dd0be4635e506233d87c12c2a2","subject":"s3api: add x-amz-checksum-type label to checksum metric","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87834c289e96a6dd0be4635e506233d87c12c2a2"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 19:39:55.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-07 01:16:00.000000000","tz":0},"subject":"Fix typos","message":"Fix typos\n\nChange-Id: Ia577a4c2b609af52859235c83290ed2683bdc423\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c4d70ea369c529d96789eed8baac05b0e74f8078"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c4d70ea369c529d96789eed8baac05b0e74f8078"}]},"branch":"refs/heads/master"},"44f8b6b731267590995dbe0ccd1e0ba5bbd33c06":{"kind":"REWORK","_number":27,"created":"2026-09-07 01:17:11.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/24/1004324/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/1004324/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/1004324/27"}}},"commit":{"parents":[{"commit":"87834c289e96a6dd0be4635e506233d87c12c2a2","subject":"s3api: add x-amz-checksum-type label to checksum metric","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87834c289e96a6dd0be4635e506233d87c12c2a2"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 19:39:55.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-07 01:17:10.000000000","tz":0},"subject":"Fix typos","message":"Fix typos\n\nChange-Id: Ia577a4c2b609af52859235c83290ed2683bdc423\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/44f8b6b731267590995dbe0ccd1e0ba5bbd33c06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/44f8b6b731267590995dbe0ccd1e0ba5bbd33c06"}]},"branch":"refs/heads/master"},"7eef125bccaf7e7dffc93495125a248abc2fdbf9":{"kind":"REWORK","_number":28,"created":"2026-09-07 01:18:29.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/24/1004324/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/1004324/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/1004324/28"}}},"commit":{"parents":[{"commit":"87834c289e96a6dd0be4635e506233d87c12c2a2","subject":"s3api: add x-amz-checksum-type label to checksum metric","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87834c289e96a6dd0be4635e506233d87c12c2a2"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 19:39:55.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-07 01:18:29.000000000","tz":0},"subject":"Fix typos","message":"Fix typos\n\nChange-Id: Ia577a4c2b609af52859235c83290ed2683bdc423\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7eef125bccaf7e7dffc93495125a248abc2fdbf9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7eef125bccaf7e7dffc93495125a248abc2fdbf9"}]},"branch":"refs/heads/master"},"a322323091ff42639fa680f9e2f56858f286ea54":{"kind":"REWORK","_number":29,"created":"2026-09-07 01:21:19.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/24/1004324/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/1004324/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/1004324/29"}}},"commit":{"parents":[{"commit":"87834c289e96a6dd0be4635e506233d87c12c2a2","subject":"s3api: add x-amz-checksum-type label to checksum metric","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87834c289e96a6dd0be4635e506233d87c12c2a2"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 19:39:55.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-07 01:20:32.000000000","tz":0},"subject":"Fix typos","message":"Fix typos\n\nChange-Id: Ia577a4c2b609af52859235c83290ed2683bdc423\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a322323091ff42639fa680f9e2f56858f286ea54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a322323091ff42639fa680f9e2f56858f286ea54"}]},"branch":"refs/heads/master"},"8bad7bc9164d9e1501e6a9986871530c06a8f543":{"kind":"REWORK","_number":30,"created":"2026-09-07 01:25:31.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/24/1004324/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/1004324/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/1004324/30"}}},"commit":{"parents":[{"commit":"87834c289e96a6dd0be4635e506233d87c12c2a2","subject":"s3api: add x-amz-checksum-type label to checksum metric","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87834c289e96a6dd0be4635e506233d87c12c2a2"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 19:39:55.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-07 01:25:28.000000000","tz":0},"subject":"Fix typos","message":"Fix typos\n\nChange-Id: Ia577a4c2b609af52859235c83290ed2683bdc423\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8bad7bc9164d9e1501e6a9986871530c06a8f543"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8bad7bc9164d9e1501e6a9986871530c06a8f543"}]},"branch":"refs/heads/master"},"a63809431f3b708759ce51d354834177ec529f0b":{"kind":"REWORK","_number":31,"created":"2026-09-07 01:29:13.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/24/1004324/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/1004324/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/1004324/31"}}},"commit":{"parents":[{"commit":"87834c289e96a6dd0be4635e506233d87c12c2a2","subject":"s3api: add x-amz-checksum-type label to checksum metric","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87834c289e96a6dd0be4635e506233d87c12c2a2"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 19:39:55.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-07 01:29:13.000000000","tz":0},"subject":"Fix typos","message":"Fix typos\n\nChange-Id: Ia577a4c2b609af52859235c83290ed2683bdc423\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a63809431f3b708759ce51d354834177ec529f0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a63809431f3b708759ce51d354834177ec529f0b"}]},"branch":"refs/heads/master"},"59b9c4e1594fcbdd478e518dba014376886ad5c9":{"kind":"REWORK","_number":32,"created":"2026-09-07 01:34:22.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/24/1004324/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/1004324/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/1004324/32"}}},"commit":{"parents":[{"commit":"87834c289e96a6dd0be4635e506233d87c12c2a2","subject":"s3api: add x-amz-checksum-type label to checksum metric","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87834c289e96a6dd0be4635e506233d87c12c2a2"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 19:39:55.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-07 01:33:22.000000000","tz":0},"subject":"Fix typos","message":"Fix typos\n\nChange-Id: Ia577a4c2b609af52859235c83290ed2683bdc423\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/59b9c4e1594fcbdd478e518dba014376886ad5c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/59b9c4e1594fcbdd478e518dba014376886ad5c9"}]},"branch":"refs/heads/master"},"65193d2bfe4f043cf47e224910a3f691205ff9c0":{"kind":"REWORK","_number":33,"created":"2026-09-07 01:48:19.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/24/1004324/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/1004324/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/1004324/33"}}},"commit":{"parents":[{"commit":"87834c289e96a6dd0be4635e506233d87c12c2a2","subject":"s3api: add x-amz-checksum-type label to checksum metric","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87834c289e96a6dd0be4635e506233d87c12c2a2"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 19:39:55.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-07 01:48:19.000000000","tz":0},"subject":"Fix typos","message":"Fix typos\n\nChange-Id: Ia577a4c2b609af52859235c83290ed2683bdc423\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/65193d2bfe4f043cf47e224910a3f691205ff9c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/65193d2bfe4f043cf47e224910a3f691205ff9c0"}]},"branch":"refs/heads/master"},"03215635686c90d28f9201c16033b504f9d3d6bd":{"kind":"REWORK","_number":34,"created":"2026-09-07 01:53:19.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/24/1004324/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/1004324/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/1004324/34"}}},"commit":{"parents":[{"commit":"87834c289e96a6dd0be4635e506233d87c12c2a2","subject":"s3api: add x-amz-checksum-type label to checksum metric","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87834c289e96a6dd0be4635e506233d87c12c2a2"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 19:39:55.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-07 01:53:18.000000000","tz":0},"subject":"Fix typos","message":"Fix typos\n\nChange-Id: Ia577a4c2b609af52859235c83290ed2683bdc423\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/03215635686c90d28f9201c16033b504f9d3d6bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/03215635686c90d28f9201c16033b504f9d3d6bd"}]},"branch":"refs/heads/master"},"0f1ddf14834682f68ccc2a9a0d7a9a3eb8514ba1":{"kind":"REWORK","_number":35,"created":"2026-09-07 01:58:42.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/24/1004324/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/1004324/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/1004324/35"}}},"commit":{"parents":[{"commit":"87834c289e96a6dd0be4635e506233d87c12c2a2","subject":"s3api: add x-amz-checksum-type label to checksum metric","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87834c289e96a6dd0be4635e506233d87c12c2a2"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 19:39:55.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-07 01:54:44.000000000","tz":0},"subject":"Fix typos","message":"Fix typos\n\nChange-Id: Ia577a4c2b609af52859235c83290ed2683bdc423\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0f1ddf14834682f68ccc2a9a0d7a9a3eb8514ba1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0f1ddf14834682f68ccc2a9a0d7a9a3eb8514ba1"}]},"branch":"refs/heads/master"},"a00719737b5b8fd73cbe1ef783a16ca4bef7787f":{"kind":"REWORK","_number":36,"created":"2026-09-07 11:40:53.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/24/1004324/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/1004324/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/1004324/36"}}},"commit":{"parents":[{"commit":"87834c289e96a6dd0be4635e506233d87c12c2a2","subject":"s3api: add x-amz-checksum-type label to checksum metric","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87834c289e96a6dd0be4635e506233d87c12c2a2"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 19:39:55.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-07 11:40:53.000000000","tz":0},"subject":"Fix typos","message":"Fix typos\n\nChange-Id: Ia577a4c2b609af52859235c83290ed2683bdc423\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a00719737b5b8fd73cbe1ef783a16ca4bef7787f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a00719737b5b8fd73cbe1ef783a16ca4bef7787f"}]},"branch":"refs/heads/master"},"f28aaf56f7bf1912702647ca3ab9a75a98c826ca":{"kind":"REWORK","_number":37,"created":"2026-09-07 11:41:23.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/24/1004324/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/1004324/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/1004324/37"}}},"commit":{"parents":[{"commit":"87834c289e96a6dd0be4635e506233d87c12c2a2","subject":"s3api: add x-amz-checksum-type label to checksum metric","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87834c289e96a6dd0be4635e506233d87c12c2a2"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 19:39:55.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-07 11:41:23.000000000","tz":0},"subject":"Fix typos","message":"Fix typos\n\nChange-Id: Ia577a4c2b609af52859235c83290ed2683bdc423\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f28aaf56f7bf1912702647ca3ab9a75a98c826ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f28aaf56f7bf1912702647ca3ab9a75a98c826ca"}]},"branch":"refs/heads/master"},"e1e4f8ea47b0f13219dc440c42cd5efd28887d11":{"kind":"REWORK","_number":38,"created":"2026-09-07 15:50:26.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/24/1004324/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/1004324/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/1004324/38"}}},"commit":{"parents":[{"commit":"87834c289e96a6dd0be4635e506233d87c12c2a2","subject":"s3api: add x-amz-checksum-type label to checksum metric","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87834c289e96a6dd0be4635e506233d87c12c2a2"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 19:39:55.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-07 15:50:26.000000000","tz":0},"subject":"Fix typos","message":"Fix typos\n\nChange-Id: Ia577a4c2b609af52859235c83290ed2683bdc423\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e1e4f8ea47b0f13219dc440c42cd5efd28887d11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e1e4f8ea47b0f13219dc440c42cd5efd28887d11"}]},"branch":"refs/heads/master"},"39ba31f469b2f51679ddf82dafd428944fd3fef0":{"kind":"REWORK","_number":39,"created":"2026-09-18 21:38:14.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/24/1004324/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/1004324/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/1004324/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/1004324/39"}}},"commit":{"parents":[{"commit":"87834c289e96a6dd0be4635e506233d87c12c2a2","subject":"s3api: add x-amz-checksum-type label to checksum metric","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87834c289e96a6dd0be4635e506233d87c12c2a2"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 19:39:55.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-09-18 21:38:05.000000000","tz":-420},"subject":"Fix typos","message":"Fix typos\n\nChange-Id: Ia577a4c2b609af52859235c83290ed2683bdc423\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/39ba31f469b2f51679ddf82dafd428944fd3fef0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/39ba31f469b2f51679ddf82dafd428944fd3fef0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
