)]}'
{"id":"openstack%2Fswift~1004470","triplet_id":"openstack%2Fswift~master~I274312cfed1627753321f2a0ba98469c4101177a","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I274312cfed1627753321f2a0ba98469c4101177a","subject":"crypto: store sensitive sysmeta key_id separately","status":"ABANDONED","created":"2026-09-07 18:22:46.000000000","updated":"2026-09-08 10:44:21.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4aaff29aed3bad38c51d2eb64e53ba5218467c82","_number":1004470,"virtual_id_number":1004470,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-07 19:34:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-07 19:15:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-07 19:34:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fffee43afe219ca15c19af3609ddac6e0735449d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-09-07 18:22:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5c8ef20aaf6b96cda980b3ce8cb4feca92c543ea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 19:15:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/999165e5eeab4d92bf73a1aca3bbb41b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b8e28ae6965f47b58622f54d97262ad1 : SUCCESS in 50m 30s","accounts_in_message":[],"_revision_number":1},{"id":"f46b85259f9cb7ef0f9bb818050bae0d0db071bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 19:34:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd9bf3ba12134ced99bc7968b0c5f7fc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f64b520d96540d088e9a89aa9b29bc9 : SUCCESS in 10m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d8ce915526548aba6deb82fd2ac61e8 : SUCCESS in 8m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/2464100d07674c1db7c5bf2744d7957c : SUCCESS in 1h 02m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e9e6f9e4827544ca8f212107b1839c59 : SUCCESS in 1h 06m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2ed1cd37cc104387a3d3f63b51ee480c : SUCCESS in 55m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d4748714d4a344809b4e2d35871111db : SUCCESS in 48m 04s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e952e1dc2a0e4929b2608a1eeb68beac : SUCCESS in 10m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7ae67a4f56f44559b6c36751cc760366 : SUCCESS in 18m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cfc00737f54343949d6402c4b5b04118 : SUCCESS in 13m 26s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ca040b0ce69543adb52137308f196da3 : SUCCESS in 5m 29s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/58d1492a59fa422494ad26b1d89ba13c : SUCCESS in 10m 33s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d13f59d5170149cc93d1161b38c260cc : SUCCESS in 16m 47s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/96a7a6b88af24a6f92738b8a0b2a1153 : SUCCESS in 13m 35s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/95c7df8e29d14aa486e16f967c9e264c : SUCCESS in 9m 58s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/1b789d0212a346458c00553b74442555 : SUCCESS in 8m 58s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9b18561df81b4d62b04a720ff4837096 : SUCCESS in 12m 58s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/85acf661995149f48fb8d2460f3163ef : SUCCESS in 8m 24s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/079f058ab8fa44de892844ef814e2cb8 : SUCCESS in 17m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fe2201e5ea5b4943a8d81640f391d454 : SUCCESS in 7m 51s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/86028d60145b4e4e9bb985f9c1425f27 : FAILURE in 17m 44s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/8fe71aa6922b4eeb858b606af2a95495 : SUCCESS in 8m 20s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3a3d59b0c5f14a9087bce9f07be4d6e0 : SUCCESS in 57m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2c6593f79e034014bb5f8670f5928019 : SUCCESS in 51m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2732735fda064c3ab9a6552288c6fdd7 : SUCCESS in 26m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/718e1de0f1084fe4ad38d3079dd879a7 : SUCCESS in 30m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de48ee9ff1144b3eaf59760e367d4998 : SUCCESS in 4m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a7cd5d9b74ef4411b9cdeea9143c76c0 : SUCCESS in 20m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/acfbd92a71214726b7b5a3df785aef74 : SUCCESS in 54m 46s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"4aaff29aed3bad38c51d2eb64e53ba5218467c82","tag":"autogenerated:gerrit:abandon","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-09-08 10:44:21.000000000","message":"Abandoned\n\nSquashed into 1003424: crypto: Encrypt object sysmeta declared sensitive | https://review.opendev.org/c/openstack/swift/+/1003424","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1b20193bc0b59dab7b38aa50b55a2674f60f7583","revisions":{"1b20193bc0b59dab7b38aa50b55a2674f60f7583":{"kind":"REWORK","_number":1,"created":"2026-09-07 18:22:46.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/70/1004470/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/70/1004470/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/1004470/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/1004470/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/1004470/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/70/1004470/1"}}},"commit":{"parents":[{"commit":"9941a3b9a4f7019dfeaefdd77d75bb6bda21741d","subject":"crypto: Encrypt object sysmeta declared sensitive","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9941a3b9a4f7019dfeaefdd77d75bb6bda21741d"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-09-07 18:18:08.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-09-07 18:22:30.000000000","tz":60},"subject":"crypto: store sensitive sysmeta key_id separately","message":"crypto: store sensitive sysmeta key_id separately\n\nRe-use the existing pattern for user metadata i.e. the encryption\nkey_id and cipher that are used to encrypt all sensitive sysmeta are\nstored in a single companion header\n\nChange-Id: I274312cfed1627753321f2a0ba98469c4101177a\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1b20193bc0b59dab7b38aa50b55a2674f60f7583"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1b20193bc0b59dab7b38aa50b55a2674f60f7583"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
