)]}'
{"id":"openstack%2Fswift~1004680","triplet_id":"openstack%2Fswift~master~Ifd87ffc033328d38bd12860a8e0fa2500a2ebe15","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"Ifd87ffc033328d38bd12860a8e0fa2500a2ebe15","subject":"sensitive sysmeta: improve docs","status":"ABANDONED","created":"2026-09-08 17:51:28.000000000","updated":"2026-09-09 15:14:16.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7dcab4ffc0110aac2fa6dfd5566197491afe9ca8","_number":1004680,"virtual_id_number":1004680,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-08 18:06:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"70c551dc48dde126156d740684a215dd73c9d165","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-09-08 17:51:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2751f28b370fcd5fb4c80a1c2ff07448732bac22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 18:06:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69d6b8c74d07409689908801a713bd18\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bfbd1d61bdf4ec0b315a5b90eac797d : SUCCESS in 5m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1bea40f890ff4b219679dc743094b7b5 : SUCCESS in 9m 31s","accounts_in_message":[],"_revision_number":1},{"id":"91c9b98f4ffae76a8f1bed24746f2e6fb953f180","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-09-09 14:59:04.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"7dcab4ffc0110aac2fa6dfd5566197491afe9ca8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-09-09 15:14:16.000000000","message":"Abandoned\n\nSquashed into 1003424: crypto: Encrypt object sysmeta declared sensitive | https://review.opendev.org/c/openstack/swift/+/1003424","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a5f3f74b64abd7aea81c7c6b6515a9d4aeca8ae0","revisions":{"daf1a21e5e5b41d27fc8840b8f7c0e14893fbe62":{"kind":"REWORK","_number":1,"created":"2026-09-08 17:51:28.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/80/1004680/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/1004680/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/1004680/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/1004680/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/1004680/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/1004680/1"}}},"commit":{"parents":[{"commit":"33e5aeb2ee1b8e746996e75906ac4772c04e8f08","subject":"crypto: Encrypt object sysmeta declared sensitive","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/33e5aeb2ee1b8e746996e75906ac4772c04e8f08"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-09-08 13:30:49.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-09-08 16:37:16.000000000","tz":60},"subject":"sensitive sysmeta: improve docs","message":"sensitive sysmeta: improve docs\n\nRelated-Change: If9d53f4f1e548dc735813ed9adbc33088037272c\nChange-Id: Ifd87ffc033328d38bd12860a8e0fa2500a2ebe15\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/daf1a21e5e5b41d27fc8840b8f7c0e14893fbe62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/daf1a21e5e5b41d27fc8840b8f7c0e14893fbe62"}]},"branch":"refs/heads/master"},"a5f3f74b64abd7aea81c7c6b6515a9d4aeca8ae0":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-09-09 14:59:04.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/80/1004680/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/1004680/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/1004680/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/1004680/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/1004680/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/1004680/2"}}},"commit":{"parents":[{"commit":"59d118b55119b9dfcb8d70ef00653507514f8c85","subject":"crypto: Encrypt object sysmeta declared sensitive","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/59d118b55119b9dfcb8d70ef00653507514f8c85"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-09-08 13:30:49.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-09-09 14:35:11.000000000","tz":60},"subject":"sensitive sysmeta: improve docs","message":"sensitive sysmeta: improve docs\n\nRelated-Change: If9d53f4f1e548dc735813ed9adbc33088037272c\nChange-Id: Ifd87ffc033328d38bd12860a8e0fa2500a2ebe15\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a5f3f74b64abd7aea81c7c6b6515a9d4aeca8ae0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a5f3f74b64abd7aea81c7c6b6515a9d4aeca8ae0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
