)]}'
{"id":"openstack%2Fswift~1004682","triplet_id":"openstack%2Fswift~master~I163779e414bb18af7e31b3c8af1fa3d989b13845","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-09-11 15:40:34.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-09-11 18:15:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I163779e414bb18af7e31b3c8af1fa3d989b13845","subject":"decrypter: add missing docstring param","status":"MERGED","created":"2026-09-08 17:51:28.000000000","updated":"2026-09-11 18:20:46.000000000","submitted":"2026-09-11 18:15:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004682","meta_rev_id":"aeca64818f87daa9a6412ca1c89745e941f9691d","_number":1004682,"virtual_id_number":1004682,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-11 18:15:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2026-09-11 15:40:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-11 15:39:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2026-09-11 15:40:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-08 18:48:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-08 18:56:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-10 19:17:07.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"349543971a29abd07a9df5403b570cda14251394","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-09-08 17:51:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a1c20c956da728c66a82cb48fc1bbda939b224b3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 18:48:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05c404f4c9864d008322e70bdcea673c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7659909f12e345c4bd807b9708e78b4f : SUCCESS in 54m 03s","accounts_in_message":[],"_revision_number":1},{"id":"133694de90d4f139a4c84ef65508c678f72ee865","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 18:56:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9b076fbb05a84a91903aa826898dfa55\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1dc021c9b6546c49e800d33defcaa83 : SUCCESS in 5m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e034a13d9874525bb456e0310daef61 : SUCCESS in 7m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/13861ed95f564f7f8371149ab5f6c67d : SUCCESS in 37m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/df5d9c295c1d4af5986f3cb1c5c03478 : FAILURE in 18m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/accb15b7911c49539210f8e5a38a1b1b : SUCCESS in 56m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/23303e3d763c4a71a64cc34e128c18f2 : SUCCESS in 34m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f895a8fe48f34dc1baa5665973708577 : SUCCESS in 21m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f57f5f255c00440996d5c6c5dbb827cf : SUCCESS in 20m 24s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cb64a57abb604764bc2c3021a3e98417 : SUCCESS in 10m 58s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e86d27816f9b4bda8af2e105b658af67 : SUCCESS in 4m 28s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c55b834712c04a4987721dab5f42759b : SUCCESS in 27m 20s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/476d47ffcb1d4b268175eb6fdf928883 : SUCCESS in 12m 25s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/bb55bb2cb3024a7588b290c804e2dd7b : SUCCESS in 23m 30s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/076a1525eeb44b93b0fe2859e32134dd : SUCCESS in 21m 44s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/dd3f0e9fcf4a41be9d8b06b12dd62ee9 : SUCCESS in 10m 10s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b7b812826a974dec9b282bcb439b714e : SUCCESS in 15m 51s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f6b1c34c5aa64f22a8fd9f74ea5eaf90 : SUCCESS in 15m 33s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/3edd3f6cfe824c49aa4e3592773c0775 : SUCCESS in 13m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f10d983f63c34b12812f13596cedcf15 : SUCCESS in 7m 53s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/94c88db51f4148f3a39715bbf5fcdcb9 : FAILURE in 32m 12s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/c644bf5682fa40b6b7fc431b4d6069e7 : SUCCESS in 8m 23s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/97b50133295443b1b34f0f2aa9c31d7e : SUCCESS in 58m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1b270d8bf9994142a53a3a10882d8cb3 : SUCCESS in 53m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/16a530cc29e54f8289eefae7361586f9 : SUCCESS in 35m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/632a6cb174ad49e2b26baba2045a9935 : SUCCESS in 27m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb76e444cdb2459d83a84c78001f3af5 : SUCCESS in 4m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/850bd8689f194be3bff7842a6cedd72b : SUCCESS in 16m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6c6cda60024b40f59b2d59853692eb17 : SUCCESS in 57m 13s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"5d84f2ed7e26fea80874158cf192670db800a513","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-09-09 14:59:04.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"000d49d3653ee226de80599d97e7589e1fd2f71f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"real_author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-09-09 15:15:04.000000000","message":"Patch Set 3: Patch Set 2 was rebased on behalf of \u003cGERRIT_ACCOUNT_7847\u003e","accounts_in_message":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}],"_revision_number":3},{"id":"89cbd76fb883ef41130911d46c291e973331a517","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"real_author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-09-09 15:16:20.000000000","message":"Patch Set 4: Patch Set 3 was rebased on behalf of \u003cGERRIT_ACCOUNT_7847\u003e","accounts_in_message":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}],"_revision_number":4},{"id":"b628b26575b0bc1774234d357cef60fbd81e8729","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"real_author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-09-09 15:26:16.000000000","message":"Patch Set 5: Patch Set 4 was rebased on behalf of \u003cGERRIT_ACCOUNT_7847\u003e","accounts_in_message":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}],"_revision_number":5},{"id":"4c8bd49b3f34f799ab4fd48eaaa64389a976b4c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 16:59:43.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/50eb196f422c4d458903bb9c257ec75e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ad7dde71b414cc28397907bd72cc868 : SUCCESS in 10m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2474ca1ee0524b56a97204b9c2648552 : SUCCESS in 10m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/a02c973ab9b841e9a6c4a6749993ca85 : SUCCESS in 1h 04m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b26547b52201402ead5dc1e5b7a92147 : SUCCESS in 1h 13m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9bd2391813184caba7276ede2e5e464b : SUCCESS in 40m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/370aec3b71b34505abfa51ffa4a19fcb : SUCCESS in 49m 41s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c6d1d01caabb4678a2c2ffa2c9d9b753 : SUCCESS in 18m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/95e450c79d09483aa41b9b06c7372b1f : SUCCESS in 19m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ba05aa155c4648af84f0fc089b74198f : SUCCESS in 21m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/33e9519993d84cc4a91831103733a16a : SUCCESS in 8m 27s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/114ef9ed9fa041c4a22852aafd57a4f2 : SUCCESS in 28m 16s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/32342e9ff14b4bd184597ff18046899f : SUCCESS in 22m 15s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/75724d91127541fe838cdfd6731f1858 : SUCCESS in 12m 00s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/846d671e326d491092344493973b3eb5 : SUCCESS in 20m 52s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d965790f4f1b4ff084fac220c3e5d13c : SUCCESS in 14m 10s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/254b3871c7a440f8b223559c09f2a490 : SUCCESS in 16m 43s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a9338466a8434821b9410d927c28c757 : SUCCESS in 16m 27s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/7959065d8791460dac5e372954d2aa9a : SUCCESS in 16m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/86d9b57560714526ab2d28ccb1508571 : SUCCESS in 7m 03s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/28b9f34619ba426391e15b906ebdab90 : FAILURE in 31m 50s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/faed4a56e402484d83f7be123e21085f : SUCCESS in 10m 05s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/54ba3a4181ac4d2d99c91f7a5850bba1 : SUCCESS in 55m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d4dff218b86244e28f9aae161aee3b55 : SUCCESS in 47m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/390f299de88a40fd940dda2da094cbce : SUCCESS in 46m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aaacd96a485242e9af5b4aa3fdc90e64 : SUCCESS in 30m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76e08d74ab0f43ad8f4ed48ee5d55c7c : FAILURE in 5m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9667c5743a23498c888fef8e7cc81d67 : SUCCESS in 24m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a206702f9b68442ca36cbebe6e8f1cdb : SUCCESS in 42m 59s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  Comments left for invalid file swift/common/middleware/versioned_writes/object_versioning.py","accounts_in_message":[],"_revision_number":5},{"id":"d3e6735f74192418ab1278cfce39dfb7fd9b1d82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"real_author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-09-09 17:38:24.000000000","message":"Patch Set 6: Patch Set 5 was rebased on behalf of \u003cGERRIT_ACCOUNT_7847\u003e\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}],"_revision_number":6},{"id":"a26b5e07c969bcfacc65b72111e6b1102ebce5de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 18:50:20.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25644103beb44d6bb4bc2b083616e744\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc6e7b5dacdf4952b65c8de57042eb42 : SUCCESS in 5m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/92751893c65543c1aaacc61a88f032c0 : SUCCESS in 10m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/0473c92ec32b45fb984314634102cb59 : SUCCESS in 1h 04m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f2b2711e9fe5481ba46f643410a68969 : SUCCESS in 1h 06m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fa5afb73b22b4ce2b3f83d6ca9abc322 : SUCCESS in 28m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/92a7d6a2231c45fda7283eda779e8da9 : SUCCESS in 1h 02m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5b1d66282cad40dcb099112d97d3f3e6 : SUCCESS in 19m 52s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f13dfe108a594883820eabcd3a74c9c4 : SUCCESS in 21m 20s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0d378d55e44c4075b7ca80050abdf91e : SUCCESS in 23m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9446f2c2267b474da1898be318a7b114 : SUCCESS in 7m 27s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/21778cb6dd2c46f4a46b67dc82efd57a : SUCCESS in 27m 20s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/2d22a8c82e904d35b53c0891bbc6a489 : SUCCESS in 23m 36s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d8dd7e94b8254814a5b570471052c893 : SUCCESS in 13m 24s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/2ada6fe9d1a0454195dcd5e6b5ef3009 : SUCCESS in 9m 51s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/95612aa718234a808e4d643de9f485f0 : SUCCESS in 16m 08s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c98ce64c20d64835a9eda643c0237efe : SUCCESS in 15m 20s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c02615c26f484ebe9a74b78b85507621 : SUCCESS in 15m 17s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/56dc849bf11f40a8ac1929cc95c1e75a : SUCCESS in 9m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/58510e6eda4b4a5982fdaf74dfaa3e86 : SUCCESS in 6m 09s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a64a1f879f5f4e27b4259fcb58f6cf04 : FAILURE in 32m 24s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/3a235edf8f4d48df8a4dcc96fd281a2b : SUCCESS in 10m 13s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/29f2dba7aef14e9ca7ea516b6aef77c4 : SUCCESS in 38m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/afd951a5e98e4647a16ec2af80b6224e : SUCCESS in 59m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/865e9ff322574ed7811e8decc17e06e1 : SUCCESS in 56m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ac98d903a08a4eae86161054d02f62a1 : SUCCESS in 28m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08ceffab916f43c6816ea977436fc50b : SUCCESS in 4m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a0577ba3f64c48138676e533f29d2c79 : SUCCESS in 18m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a3e1e50d34564a58995d781935ae5459 : SUCCESS in 29m 01s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":6},{"id":"4e5345799123c32be844ead82931b55cbaa283ae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 19:10:58.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0b06560790d49f18a7b09898350b451\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9659f0e8feca42ba9324de3b1c88c2cc : SUCCESS in 1h 05m 39s","accounts_in_message":[],"_revision_number":6},{"id":"2daeeb8f6a93774fa8726ed49627e14c4589a9e9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-10 19:17:07.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"774b7ee25a7a899e8ff71839bd1cc1ee7f602f12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-09-11 15:39:36.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"7859e4420721178713dd646aa3756980835e3d31","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-09-11 15:40:34.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"6f81a4ba1955cd8ca3903e3d1a147a215e89f071","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-09-11 15:40:53.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"973b296656d7fcf46ff493f08a50b34ad94b5090","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 16:34:26.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b8f8274964b4333b1b23309659d1a20\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/575ac2fc69214fcca61431dcc5973e85 : SUCCESS in 52m 42s","accounts_in_message":[],"_revision_number":7},{"id":"a0a40e23176e95f8081510d7836db895288f1a89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 16:51:48.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad3b3f314ffb4c778a41daa93167bec4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/556fcdf23d3c4773b4a6fc0e6e69746d : SUCCESS in 5m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/863650a3c10f40ffaa28f0fef2e21bd5 : SUCCESS in 10m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/2528d43755234c1b917c38093ae46be4 : SUCCESS in 30m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/70cbd6fdcb954ffaaa5dc532a788dba6 : SUCCESS in 1h 06m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e574a7f6f97d401096348d11fcd7d422 : SUCCESS in 52m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e838c389252648eda16cea59979bea6d : SUCCESS in 1h 03m 38s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/693220a6943b466e9bf27d1e728eb8fe : SUCCESS in 19m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/98a206bdc06347e4a990f9d326587de1 : SUCCESS in 23m 55s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/146aca055e2f402482a39db0843ef099 : SUCCESS in 20m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/81d504acc63d498ca71d9b61f0b8a3fd : SUCCESS in 6m 36s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7bf399a1ef2546dc8750db46f3f9f54b : SUCCESS in 11m 25s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/5f788bd96df44145a51f2d93a5c97f4d : SUCCESS in 23m 24s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/b2be7be94ca74474927f54f443f990d1 : SUCCESS in 27m 08s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/f097008b8189408ba8bb24e58911edb4 : SUCCESS in 10m 48s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7cf61510505a4f0d92d08f37047aa112 : SUCCESS in 14m 50s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/cf9e349fbed5452fb47c5989836af1e7 : SUCCESS in 15m 21s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/12e281fc8d6c49e09f05a4ed1eca8643 : SUCCESS in 15m 05s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/3f715b0b4fd442b485059be532630ba2 : SUCCESS in 17m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a4f80b7df9fe4e1aa970e81008e2c951 : SUCCESS in 4m 09s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/aba131d7557f489498073c59f03ad43d : FAILURE in 30m 30s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/a6a79e03dfa24ff9b6bafb9d3c8c53bd : SUCCESS in 5m 49s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/017f1fe2804a4818acd0bd631213638c : SUCCESS in 1h 01m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4e58f7aa97c44c49a51f7cd48372479a : SUCCESS in 52m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ccddc57036e14b27bb13038893f93f47 : SUCCESS in 53m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6114609be2654269bc285ef2f0fb88cf : SUCCESS in 28m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e1ac63fce0343179f67730b46eed306 : SUCCESS in 4m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/af6c3c7032f847e1bdb67ab284c832e9 : SUCCESS in 27m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6c6cd53eff984ba7b406c5011ef9cfbf : SUCCESS in 53m 50s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":7},{"id":"1d0d00c03a8bed9033a233b4d8e9b350e4156049","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 16:52:19.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"d1acc340d73bace6b4baeca156bf5b3a01709529","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 18:15:48.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/998f1bd1d1c34d5f9097be4ea1119b87\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/232bc82cec204e4497b093ccc0c628af : SUCCESS in 6m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/09c12fa3115a408499966865f8813a12 : SUCCESS in 9m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/4c3e086df9314da89006d560d6c9bfb3 : SUCCESS in 1h 03m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f04a563023394b299cd5997f9f71bfb0 : SUCCESS in 47m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ad6a9eb480404ed9aa028bb8b2bb8acb : SUCCESS in 28m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d51e05818b3c4509a10095f8c095d440 : SUCCESS in 58m 05s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/ca19bd8650c84270b9ca9854e4aa447b : SUCCESS in 7m 16s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/181471b7288d4db4a0b53015842bf17a : SUCCESS in 25m 21s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/f4958e85fa8444c18e43ba08a4ccaaa6 : SUCCESS in 23m 29s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/034845bde5784d66b9be7765635c1c9e : SUCCESS in 34m 52s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/e5613835aa72477187fa291daf4158e1 : SUCCESS in 9m 38s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/1763809fa86b4df7aeeb6971d87103aa : SUCCESS in 8m 50s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/669fc183a6324b798704e3f79ae7cd1d : SUCCESS in 9m 53s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5591cfff3e304d8b93b7a563fffd9f9d : SUCCESS in 17m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e766dac6fa79443eb7ad3200c2a0dfd9 : SUCCESS in 8m 32s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/56bab3681c6d4507827e2f2be28eda0c : SUCCESS in 9m 08s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/78e47083a4704002a9afab58b29b1c2e : SUCCESS in 1h 06m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7d4ff927066d443da724617214c2d9fa : SUCCESS in 37m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ec7c335b90a147ffadd82e996600b4d1 : SUCCESS in 26m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b52134079aa247b5b33ca54a150ee624 : SUCCESS in 25m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99826dc3c6e24b91b77a6f1f621e4910 : SUCCESS in 4m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cc0bd816c5a942d9a385e66f79151136 : SUCCESS in 18m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bab2b683c8774af59fcee7c2366991ac : SUCCESS in 57m 35s","accounts_in_message":[],"_revision_number":7},{"id":"b9efb1b622c41e25d28cea3eb0378914dc828de4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 18:15:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"aeca64818f87daa9a6412ca1c89745e941f9691d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 18:20:46.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7954449e41847dab1aaeb740342f421\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9958e39bedd044ccb1ffeca2b73ccfa5 : SUCCESS in 1m 17s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/406c6d4871464b6e91f35c4435165476 : SUCCESS in 48s (non-voting)","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"5c30238d615bc7cc14bc07933015b7972aea8ce8","revisions":{"4558579e6e82f8b1a2de602f531b37223aab31de":{"kind":"REWORK","_number":1,"created":"2026-09-08 17:51:28.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/82/1004682/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/82/1004682/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/1004682/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/1004682/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/1004682/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/82/1004682/1"}}},"commit":{"parents":[{"commit":"21504f4e437077236e59bb523f9a58d6aa5d22ff","subject":"sensitive sysmeta: add probe test assertions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/21504f4e437077236e59bb523f9a58d6aa5d22ff"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-09-08 13:42:45.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-09-08 16:37:16.000000000","tz":60},"subject":"decrypter: add missing docstring param","message":"decrypter: add missing docstring param\n\nChange-Id: I163779e414bb18af7e31b3c8af1fa3d989b13845\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4558579e6e82f8b1a2de602f531b37223aab31de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4558579e6e82f8b1a2de602f531b37223aab31de"}]},"branch":"refs/heads/master"},"533aa9b6b1dcd20da16fe7b6095dd3bfe917f847":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-09-09 14:59:04.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/82/1004682/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/82/1004682/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/1004682/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/1004682/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/1004682/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/82/1004682/2"}}},"commit":{"parents":[{"commit":"2b2b99ffbd913f45c78282ba23748af13320010a","subject":"sensitive sysmeta: add probe test assertions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2b2b99ffbd913f45c78282ba23748af13320010a"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-09-08 13:42:45.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-09-09 14:35:11.000000000","tz":60},"subject":"decrypter: add missing docstring param","message":"decrypter: add missing docstring param\n\nChange-Id: I163779e414bb18af7e31b3c8af1fa3d989b13845\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/533aa9b6b1dcd20da16fe7b6095dd3bfe917f847"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/533aa9b6b1dcd20da16fe7b6095dd3bfe917f847"}]},"branch":"refs/heads/master"},"6505650eb9ca51974dec4f10723108277be56cfe":{"kind":"NO_CHANGE","_number":3,"created":"2026-09-09 15:15:04.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"real_uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/82/1004682/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/82/1004682/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/1004682/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/1004682/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/1004682/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/82/1004682/3"}}},"commit":{"parents":[{"commit":"946ecbbc4f80a4aaf80f3beb8e5af0b77b26f469","subject":"crypto: Encrypt object sysmeta declared sensitive","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/946ecbbc4f80a4aaf80f3beb8e5af0b77b26f469"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-09-08 13:42:45.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-09-09 15:15:04.000000000","tz":0},"subject":"decrypter: add missing docstring param","message":"decrypter: add missing docstring param\n\nChange-Id: I163779e414bb18af7e31b3c8af1fa3d989b13845\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6505650eb9ca51974dec4f10723108277be56cfe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6505650eb9ca51974dec4f10723108277be56cfe"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"2b2b99ffbd913f45c78282ba23748af13320010a","ours":"533aa9b6b1dcd20da16fe7b6095dd3bfe917f847","theirs":"946ecbbc4f80a4aaf80f3beb8e5af0b77b26f469","merge_strategy":"recursive","contains_conflicts":false}},"16bd098b9e03556faef08ed06924fbc8791673a9":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-09-09 15:16:20.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"real_uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/82/1004682/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/82/1004682/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/1004682/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/1004682/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/1004682/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/82/1004682/4"}}},"commit":{"parents":[{"commit":"718e1e4b00e910902e3f8d8b31f5c3f8752b1f71","subject":"crypto: Encrypt object sysmeta declared sensitive","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/718e1e4b00e910902e3f8d8b31f5c3f8752b1f71"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-09-08 13:42:45.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-09-09 15:16:20.000000000","tz":0},"subject":"decrypter: add missing docstring param","message":"decrypter: add missing docstring param\n\nChange-Id: I163779e414bb18af7e31b3c8af1fa3d989b13845\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/16bd098b9e03556faef08ed06924fbc8791673a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/16bd098b9e03556faef08ed06924fbc8791673a9"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"946ecbbc4f80a4aaf80f3beb8e5af0b77b26f469","ours":"6505650eb9ca51974dec4f10723108277be56cfe","theirs":"718e1e4b00e910902e3f8d8b31f5c3f8752b1f71","merge_strategy":"recursive","contains_conflicts":false}},"cc5f23071864344932d3e85b37019025042877a9":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-09-09 15:26:16.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"real_uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/82/1004682/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/82/1004682/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/1004682/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/1004682/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/1004682/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/82/1004682/5"}}},"commit":{"parents":[{"commit":"a0aac4c34ea337b3707a1625fde96b75af531969","subject":"crypto: Encrypt object sysmeta declared sensitive","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a0aac4c34ea337b3707a1625fde96b75af531969"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-09-08 13:42:45.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-09-09 15:26:16.000000000","tz":0},"subject":"decrypter: add missing docstring param","message":"decrypter: add missing docstring param\n\nChange-Id: I163779e414bb18af7e31b3c8af1fa3d989b13845\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cc5f23071864344932d3e85b37019025042877a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cc5f23071864344932d3e85b37019025042877a9"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"718e1e4b00e910902e3f8d8b31f5c3f8752b1f71","ours":"16bd098b9e03556faef08ed06924fbc8791673a9","theirs":"a0aac4c34ea337b3707a1625fde96b75af531969","merge_strategy":"recursive","contains_conflicts":false}},"61f3dfbd068b91a15ef4d32d9fa1310ba981862f":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-09-09 17:38:24.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"real_uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/82/1004682/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/82/1004682/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/1004682/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/1004682/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/1004682/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/82/1004682/6"}}},"commit":{"parents":[{"commit":"d73d0b3fb00013404fd436c70c746693b6d52271","subject":"crypto: Encrypt object sysmeta declared sensitive","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d73d0b3fb00013404fd436c70c746693b6d52271"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-09-08 13:42:45.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-09-09 17:38:24.000000000","tz":0},"subject":"decrypter: add missing docstring param","message":"decrypter: add missing docstring param\n\nChange-Id: I163779e414bb18af7e31b3c8af1fa3d989b13845\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/61f3dfbd068b91a15ef4d32d9fa1310ba981862f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/61f3dfbd068b91a15ef4d32d9fa1310ba981862f"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"a0aac4c34ea337b3707a1625fde96b75af531969","ours":"cc5f23071864344932d3e85b37019025042877a9","theirs":"d73d0b3fb00013404fd436c70c746693b6d52271","merge_strategy":"recursive","contains_conflicts":false}},"5c30238d615bc7cc14bc07933015b7972aea8ce8":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-09-11 15:39:36.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/82/1004682/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/82/1004682/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/1004682/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/1004682/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/1004682/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/82/1004682/7"}}},"commit":{"parents":[{"commit":"9318950da879bd33b3b5216d0d0cfed4dea45d9c","subject":"Merge \"replicator: Quarantine suffixes we cannot delete\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9318950da879bd33b3b5216d0d0cfed4dea45d9c"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-09-08 13:42:45.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-09-11 15:35:55.000000000","tz":60},"subject":"decrypter: add missing docstring param","message":"decrypter: add missing docstring param\n\nChange-Id: I163779e414bb18af7e31b3c8af1fa3d989b13845\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5c30238d615bc7cc14bc07933015b7972aea8ce8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5c30238d615bc7cc14bc07933015b7972aea8ce8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
