)]}'
{"id":"openstack%2Fswift~1004814","triplet_id":"openstack%2Fswift~master~I5c911c3bf2b63b788eef3c98e7292ec8c584bc30","project":"openstack/swift","branch":"master","attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-09-11 15:21:20.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-09-11 15:21:20.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"hashtags":[],"change_id":"I5c911c3bf2b63b788eef3c98e7292ec8c584bc30","subject":"sysmeta: recommend unreserved namespace for out of tree use","status":"NEW","created":"2026-09-09 14:06:57.000000000","updated":"2026-09-11 16:02:28.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"4de2883de91c0c5123886c4ad6a4c0959b1e31d2","_number":1004814,"virtual_id_number":1004814,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-11 16:02:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-09 14:47:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-10 19:19:54.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"0ecb0763f1eb0e4070aa2890fc155746d3e0082b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-09-09 14:06:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"197ef414613489af55ace5996d9c60ac669f95c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 14:47:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/900b1d25d5a2456a90f76c344aaa41dc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3ce0620247b464e927548c08cfb7ee1 : SUCCESS in 11m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/acf5ed940ee44fdc88b6a665850afef6 : SUCCESS in 10m 01s","accounts_in_message":[],"_revision_number":1},{"id":"7b385ba6519bf1ac715f0955ec6c5234d5e405ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-09-09 14:59:04.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"16171e1098fba046598a740e4f8598baad4dc6a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"real_author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-09-09 15:15:13.000000000","message":"Patch Set 3: Patch Set 2 was rebased on behalf of \u003cGERRIT_ACCOUNT_7847\u003e","accounts_in_message":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}],"_revision_number":3},{"id":"a62a8f4d7359b9531796a84649986f3229ff1ce1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"real_author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-09-09 15:16:26.000000000","message":"Patch Set 4: Patch Set 3 was rebased on behalf of \u003cGERRIT_ACCOUNT_7847\u003e","accounts_in_message":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}],"_revision_number":4},{"id":"21f4551708e61162231d992bf000f206a508b203","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"real_author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-09-09 15:26:24.000000000","message":"Patch Set 5: Patch Set 4 was rebased on behalf of \u003cGERRIT_ACCOUNT_7847\u003e","accounts_in_message":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}],"_revision_number":5},{"id":"6f789e1e3120a7c97213b5ecb8da1b464150ebc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 15:53:24.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d747a0ef6f7d4b858452e897edce0aa3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c60f1e095a9430bbb6efd2e8f6f0011 : SUCCESS in 10m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6cee4a899a4b494eacc2aecb6daad36e : SUCCESS in 9m 12s","accounts_in_message":[],"_revision_number":5},{"id":"7c1a33cfeca214f9d3cb13480083c7133947b39f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"real_author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-09-09 17:38:47.000000000","message":"Patch Set 6: Patch Set 5 was rebased on behalf of \u003cGERRIT_ACCOUNT_7847\u003e\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}],"_revision_number":6},{"id":"0541ff3839d57fe11460e8b6cadbc635aafcc166","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 17:57:04.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61518080b3594ff2bcfc6cc73203042f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e50e6f5e311447839ce3a517f81d2a02 : SUCCESS in 12m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d759c6d91a64153859b0ae1d5530b3f : SUCCESS in 10m 36s","accounts_in_message":[],"_revision_number":6},{"id":"a29d37a6a68883882bceba1d1b4cb50ce368d778","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-10 19:19:54.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d34a4a78a04459fca64c71ada0dfcb9c6d94f4fc","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-09-11 15:21:20.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"589a4f0628f2c2e16b123b91a5fbee62c8499878","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-09-11 15:51:40.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"4de2883de91c0c5123886c4ad6a4c0959b1e31d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 16:02:28.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/177c5b8b87294030a5d06aa2e8ec4a68\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c7cab3dc42443469dc3f3a3ade6a5ce : SUCCESS in 8m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a2ea4ee82e3408fb69aad082b211764 : SUCCESS in 6m 23s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"882953e8bcf7ce5c1cb529451df9ef44d1af2725","revisions":{"77040dbda049c882cc80546c062fe8ae12b5ecb1":{"kind":"REWORK","_number":1,"created":"2026-09-09 14:06:57.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/14/1004814/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/1004814/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/1004814/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/1004814/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/1004814/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/1004814/1"}}},"commit":{"parents":[{"commit":"4558579e6e82f8b1a2de602f531b37223aab31de","subject":"decrypter: add missing docstring param","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4558579e6e82f8b1a2de602f531b37223aab31de"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-09-09 13:49:06.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-09-09 14:05:21.000000000","tz":60},"subject":"sysmeta: recommend unreserved namespace for out of tree use","message":"sysmeta: recommend unreserved namespace for out of tree use\n\nThere is no reserved sysmeta namespace for upstream Swift\nmiddlewares. New upstream sysmeta use cases have adopted a\nsub-namespace of \u0027x-\u003ctype\u003e-sysmeta-\u0027. For example, the Related-Change\nhas adopted the \u0027x-object-sysmeta-sensitive-\u0027 namespace.\n\nThis patch belatedly defines unreserved sysmeta namespaces that\nthird-party middleware developers may use without risk of clashing\nwith current upstream sysmeta namespaces.\n\nRelated-Change: If9d53f4f1e548dc735813ed9adbc33088037272c\nChange-Id: I5c911c3bf2b63b788eef3c98e7292ec8c584bc30\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/77040dbda049c882cc80546c062fe8ae12b5ecb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/77040dbda049c882cc80546c062fe8ae12b5ecb1"}]},"branch":"refs/heads/master"},"5980edb87ad2d72921ad2e176acd30885601ffed":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-09-09 14:59:04.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/14/1004814/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/1004814/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/1004814/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/1004814/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/1004814/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/1004814/2"}}},"commit":{"parents":[{"commit":"533aa9b6b1dcd20da16fe7b6095dd3bfe917f847","subject":"decrypter: add missing docstring param","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/533aa9b6b1dcd20da16fe7b6095dd3bfe917f847"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-09-09 13:49:06.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-09-09 14:35:11.000000000","tz":60},"subject":"sysmeta: recommend unreserved namespace for out of tree use","message":"sysmeta: recommend unreserved namespace for out of tree use\n\nThere is no reserved sysmeta namespace for upstream Swift\nmiddlewares. New upstream sysmeta use cases have adopted a\nsub-namespace of \u0027x-\u003ctype\u003e-sysmeta-\u0027. For example, the Related-Change\nhas adopted the \u0027x-object-sysmeta-sensitive-\u0027 namespace.\n\nThis patch belatedly defines unreserved sysmeta namespaces that\nthird-party middleware developers may use without risk of clashing\nwith current upstream sysmeta namespaces.\n\nRelated-Change: If9d53f4f1e548dc735813ed9adbc33088037272c\nChange-Id: I5c911c3bf2b63b788eef3c98e7292ec8c584bc30\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5980edb87ad2d72921ad2e176acd30885601ffed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5980edb87ad2d72921ad2e176acd30885601ffed"}]},"branch":"refs/heads/master"},"637783ffbbf139daeec2561f80c1378f47e3a71a":{"kind":"NO_CHANGE","_number":3,"created":"2026-09-09 15:15:13.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"real_uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/14/1004814/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/1004814/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/1004814/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/1004814/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/1004814/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/1004814/3"}}},"commit":{"parents":[{"commit":"6505650eb9ca51974dec4f10723108277be56cfe","subject":"decrypter: add missing docstring param","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6505650eb9ca51974dec4f10723108277be56cfe"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-09-09 13:49:06.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-09-09 15:15:13.000000000","tz":0},"subject":"sysmeta: recommend unreserved namespace for out of tree use","message":"sysmeta: recommend unreserved namespace for out of tree use\n\nThere is no reserved sysmeta namespace for upstream Swift\nmiddlewares. New upstream sysmeta use cases have adopted a\nsub-namespace of \u0027x-\u003ctype\u003e-sysmeta-\u0027. For example, the Related-Change\nhas adopted the \u0027x-object-sysmeta-sensitive-\u0027 namespace.\n\nThis patch belatedly defines unreserved sysmeta namespaces that\nthird-party middleware developers may use without risk of clashing\nwith current upstream sysmeta namespaces.\n\nRelated-Change: If9d53f4f1e548dc735813ed9adbc33088037272c\nChange-Id: I5c911c3bf2b63b788eef3c98e7292ec8c584bc30\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/637783ffbbf139daeec2561f80c1378f47e3a71a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/637783ffbbf139daeec2561f80c1378f47e3a71a"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"533aa9b6b1dcd20da16fe7b6095dd3bfe917f847","ours":"5980edb87ad2d72921ad2e176acd30885601ffed","theirs":"6505650eb9ca51974dec4f10723108277be56cfe","merge_strategy":"recursive","contains_conflicts":false}},"96837716cb9804109f78e2528a148b68740f92c4":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-09-09 15:16:26.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"real_uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/14/1004814/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/1004814/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/1004814/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/1004814/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/1004814/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/1004814/4"}}},"commit":{"parents":[{"commit":"16bd098b9e03556faef08ed06924fbc8791673a9","subject":"decrypter: add missing docstring param","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/16bd098b9e03556faef08ed06924fbc8791673a9"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-09-09 13:49:06.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-09-09 15:16:26.000000000","tz":0},"subject":"sysmeta: recommend unreserved namespace for out of tree use","message":"sysmeta: recommend unreserved namespace for out of tree use\n\nThere is no reserved sysmeta namespace for upstream Swift\nmiddlewares. New upstream sysmeta use cases have adopted a\nsub-namespace of \u0027x-\u003ctype\u003e-sysmeta-\u0027. For example, the Related-Change\nhas adopted the \u0027x-object-sysmeta-sensitive-\u0027 namespace.\n\nThis patch belatedly defines unreserved sysmeta namespaces that\nthird-party middleware developers may use without risk of clashing\nwith current upstream sysmeta namespaces.\n\nRelated-Change: If9d53f4f1e548dc735813ed9adbc33088037272c\nChange-Id: I5c911c3bf2b63b788eef3c98e7292ec8c584bc30\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/96837716cb9804109f78e2528a148b68740f92c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/96837716cb9804109f78e2528a148b68740f92c4"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"6505650eb9ca51974dec4f10723108277be56cfe","ours":"637783ffbbf139daeec2561f80c1378f47e3a71a","theirs":"16bd098b9e03556faef08ed06924fbc8791673a9","merge_strategy":"recursive","contains_conflicts":false}},"b659cefcb7cc0a0b11e5a88fd2f864c9f84044fd":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-09-09 15:26:24.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"real_uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/14/1004814/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/1004814/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/1004814/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/1004814/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/1004814/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/1004814/5"}}},"commit":{"parents":[{"commit":"cc5f23071864344932d3e85b37019025042877a9","subject":"decrypter: add missing docstring param","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cc5f23071864344932d3e85b37019025042877a9"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-09-09 13:49:06.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-09-09 15:26:24.000000000","tz":0},"subject":"sysmeta: recommend unreserved namespace for out of tree use","message":"sysmeta: recommend unreserved namespace for out of tree use\n\nThere is no reserved sysmeta namespace for upstream Swift\nmiddlewares. New upstream sysmeta use cases have adopted a\nsub-namespace of \u0027x-\u003ctype\u003e-sysmeta-\u0027. For example, the Related-Change\nhas adopted the \u0027x-object-sysmeta-sensitive-\u0027 namespace.\n\nThis patch belatedly defines unreserved sysmeta namespaces that\nthird-party middleware developers may use without risk of clashing\nwith current upstream sysmeta namespaces.\n\nRelated-Change: If9d53f4f1e548dc735813ed9adbc33088037272c\nChange-Id: I5c911c3bf2b63b788eef3c98e7292ec8c584bc30\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b659cefcb7cc0a0b11e5a88fd2f864c9f84044fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b659cefcb7cc0a0b11e5a88fd2f864c9f84044fd"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"16bd098b9e03556faef08ed06924fbc8791673a9","ours":"96837716cb9804109f78e2528a148b68740f92c4","theirs":"cc5f23071864344932d3e85b37019025042877a9","merge_strategy":"recursive","contains_conflicts":false}},"6a30bac5c51675214ccbafb7135f4f92b07da42a":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-09-09 17:38:47.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"real_uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/14/1004814/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/1004814/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/1004814/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/1004814/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/1004814/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/1004814/6"}}},"commit":{"parents":[{"commit":"61f3dfbd068b91a15ef4d32d9fa1310ba981862f","subject":"decrypter: add missing docstring param","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/61f3dfbd068b91a15ef4d32d9fa1310ba981862f"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-09-09 13:49:06.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-09-09 17:38:47.000000000","tz":0},"subject":"sysmeta: recommend unreserved namespace for out of tree use","message":"sysmeta: recommend unreserved namespace for out of tree use\n\nThere is no reserved sysmeta namespace for upstream Swift\nmiddlewares. New upstream sysmeta use cases have adopted a\nsub-namespace of \u0027x-\u003ctype\u003e-sysmeta-\u0027. For example, the Related-Change\nhas adopted the \u0027x-object-sysmeta-sensitive-\u0027 namespace.\n\nThis patch belatedly defines unreserved sysmeta namespaces that\nthird-party middleware developers may use without risk of clashing\nwith current upstream sysmeta namespaces.\n\nRelated-Change: If9d53f4f1e548dc735813ed9adbc33088037272c\nChange-Id: I5c911c3bf2b63b788eef3c98e7292ec8c584bc30\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6a30bac5c51675214ccbafb7135f4f92b07da42a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6a30bac5c51675214ccbafb7135f4f92b07da42a"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"cc5f23071864344932d3e85b37019025042877a9","ours":"b659cefcb7cc0a0b11e5a88fd2f864c9f84044fd","theirs":"61f3dfbd068b91a15ef4d32d9fa1310ba981862f","merge_strategy":"recursive","contains_conflicts":false}},"882953e8bcf7ce5c1cb529451df9ef44d1af2725":{"kind":"REWORK","_number":7,"created":"2026-09-11 15:51:40.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/14/1004814/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/1004814/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/1004814/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/1004814/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/1004814/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/1004814/7"}}},"commit":{"parents":[{"commit":"4d36f9e79587fcc807a511b933d097fdfac2eedd","subject":"crypto: Encrypt object sysmeta declared sensitive","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4d36f9e79587fcc807a511b933d097fdfac2eedd"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-09-09 13:49:06.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-09-11 15:48:29.000000000","tz":60},"subject":"sysmeta: recommend unreserved namespace for out of tree use","message":"sysmeta: recommend unreserved namespace for out of tree use\n\nThere is no reserved sysmeta namespace for upstream Swift\nmiddlewares. New upstream sysmeta use cases have adopted a\nsub-namespace of \u0027x-\u003ctype\u003e-sysmeta-\u0027. For example, the Related-Change\nhas adopted the \u0027x-object-sysmeta-sensitive-\u0027 namespace.\n\nThis patch belatedly defines unreserved sysmeta namespaces that\nthird-party middleware developers may use without risk of clashing\nwith current upstream sysmeta namespaces.\n\nRelated-Change: If9d53f4f1e548dc735813ed9adbc33088037272c\nChange-Id: I5c911c3bf2b63b788eef3c98e7292ec8c584bc30\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/882953e8bcf7ce5c1cb529451df9ef44d1af2725"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/882953e8bcf7ce5c1cb529451df9ef44d1af2725"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
