)]}'
{"id":"openstack%2Fswift~1004912","triplet_id":"openstack%2Fswift~master~I6caefecca0af023ec3ada8260531826e5e1a0064","project":"openstack/swift","branch":"master","topic":"s3-checksum-persistence","hashtags":[],"change_id":"I6caefecca0af023ec3ada8260531826e5e1a0064","subject":"WIP: sq: move checksum value to sensitive sysmeta namespace","status":"ABANDONED","created":"2026-09-09 17:57:14.000000000","updated":"2026-09-10 13:53:28.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ea8a8cc404c7aa6a3b6971efafb6a4dc0688b258","_number":1004912,"virtual_id_number":1004912,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-09 19:03:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-09 19:03:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"08cf48d96bc8e5d6824536fbf22864b7bedc4264","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-09-09 17:57:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9e6ad7171836c133d050734165e4f29c7409c8ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 19:03:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d26f08fb20141a9bf19775086d4c902\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f12dc647feb418fb08dd0468c42cbf6 : SUCCESS in 6m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b0343f911c1486eb5811ca1b3ebf853 : SUCCESS in 10m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/2d74845219c04df698817a7b89ab6b0a : SUCCESS in 1h 02m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c5e582afb4084e68a43db33b1334e709 : SUCCESS in 1h 03m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c5fdd38bd9a74e77a22e854b63888cd8 : SUCCESS in 50m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8c1c8b7ae83046719a4390e3e1e510f6 : SUCCESS in 34m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5336ac34a65e4f84bea62c12e8384a6d : SUCCESS in 19m 59s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/87a629879e9f4c178c70c181a5ff6388 : SUCCESS in 20m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4c05fc20ea1e4cd1924cd7621b92358f : SUCCESS in 12m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4a915082846c4d759eab56ca4c9cbf6b : SUCCESS in 8m 53s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/df1bf53068354e2cad169307903c1f8a : SUCCESS in 14m 36s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/dc04eb81e1014cd48c705b28024d217c : SUCCESS in 10m 53s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c82baf2e59734bb3830338da427754f2 : SUCCESS in 12m 57s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/f8217d44227f4d8b96caa7be5ed10f50 : SUCCESS in 11m 31s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/eadf20cb16ae4f41904fef513b58bebd : SUCCESS in 14m 24s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/62163046292648318c254f0344237653 : SUCCESS in 15m 51s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3625b8fc185446e9b3d42d22dbd46e31 : SUCCESS in 10m 11s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/d04f0d9c5af041fcbc035e0cf6930dbb : SUCCESS in 11m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2adc6b70b7c24783b6674e15cf204efc : SUCCESS in 15m 05s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fde6279067e04463afbb047c12e4da27 : FAILURE in 32m 56s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/58f2d2f66e224cd8841f073c783abeeb : SUCCESS in 11m 03s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5c08895c629f4ff5ab7f260347e03e81 : SUCCESS in 37m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/486eb979c5cb46d383b2a148ac7ea0ea : SUCCESS in 52m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8fa1671bd81a4b5491ab90eef0a8c5ee : SUCCESS in 59m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b720ee49d83d4f68ab8ebb4eb9d99fe0 : SUCCESS in 12m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01391d26b1324ec887ce0bd361307ec9 : SUCCESS in 4m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9a4d036edf9b4dd28154b918b8bc8345 : SUCCESS in 9m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e9e65923380545bc95e3eea3287bbf3d : SUCCESS in 27m 53s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"792a3ba5e8d467dc16b9c4143b178ba06581550b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 19:05:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0aaf706f9654e5da41afbc71d4d7e53\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c6b8cd2a74c742dda6189d18e7629c6a : SUCCESS in 58m 48s","accounts_in_message":[],"_revision_number":1},{"id":"ea8a8cc404c7aa6a3b6971efafb6a4dc0688b258","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-09-10 13:53:28.000000000","message":"Abandoned\n\nsquashed to parent after modifications/fixes","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fe0f318868cf99237daf43edd3349d2d7fb5335d","revisions":{"fe0f318868cf99237daf43edd3349d2d7fb5335d":{"kind":"REWORK","_number":1,"created":"2026-09-09 17:57:14.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/12/1004912/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/1004912/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/1004912/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/1004912/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/1004912/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/1004912/1"}}},"commit":{"parents":[{"commit":"f6f50ed0336c4218058fc28c70f88eb572b3f690","subject":"s3api: Checksum Persistence for Full Object Upload PUTs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f6f50ed0336c4218058fc28c70f88eb572b3f690"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-09-09 17:49:57.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-09-09 17:53:03.000000000","tz":60},"subject":"WIP: sq: move checksum value to sensitive sysmeta namespace","message":"WIP: sq: move checksum value to sensitive sysmeta namespace\n\ngetting this off my laptop. I want to check over before\nsquashing into the parent.\n\nChange-Id: I6caefecca0af023ec3ada8260531826e5e1a0064\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe0f318868cf99237daf43edd3349d2d7fb5335d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe0f318868cf99237daf43edd3349d2d7fb5335d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
