)]}'
{"id":"openstack%2Fswift~1005145","triplet_id":"openstack%2Fswift~master~Idb10b7307bcf605a6290cc1c1446ab7ffd62cd15","project":"openstack/swift","branch":"master","topic":"replicator-quarantine","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2026-09-15 22:33:52.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},"38767":{"account":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"last_update":"2026-09-16 00:00:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idb10b7307bcf605a6290cc1c1446ab7ffd62cd15","subject":"replicator: Continue handoff cleanup after quarantine failures","status":"MERGED","created":"2026-09-10 23:33:15.000000000","updated":"2026-09-16 00:02:22.000000000","submitted":"2026-09-16 00:00:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":36,"unresolved_comment_count":6,"has_review_started":true,"submission_id":"1005145-replicator-quarantine","meta_rev_id":"fc8942a39856bc391edd91d8d84de9ca21a9c639","_number":1005145,"virtual_id_number":1005145,"owner":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-16 00:00:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2026-09-15 22:33:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2026-09-15 22:33:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-11 00:57:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-11 14:37:18.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"f1523576379030872c6f04dad42a43d5f8986d9f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-09-10 23:33:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b1fbbef478bc909e005c09a47fbb88a9cee5db9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 00:57:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/52471b110dea467e8bed14a6ec780260\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/907e46d71c8c4d518f69720b796d34ba : SUCCESS in 11m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b7d10a094bfe4387a897b6407a2beda7 : SUCCESS in 7m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/20444926d6f6415bada7cb275dde6e18 : SUCCESS in 1h 18m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8e754b4f045c4171ab14d3e80c7cf0af : SUCCESS in 33m 40s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a4dfa007ab4c4b07983b885f275bc5b5 : SUCCESS in 33m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f7560202ace5471abad3e91b101c694e : SUCCESS in 58m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ab53b5ef3d8b4ba68cb30c02aa57823d : SUCCESS in 17m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/34adc170b73e4e27b1f59419dd3253f8 : SUCCESS in 12m 39s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7e125b60c2f64cda98d91d8e26fb5228 : SUCCESS in 12m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/48dac8aad385464e827121c03efbe64a : SUCCESS in 5m 56s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/cc6f13305845429cb5440b663d7a85fa : FAILURE in 40m 45s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/809d847a78584b028cd0863c48f2de8f : FAILURE in 34m 28s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c211e52113f940aeb852668ba5096d20 : FAILURE in 30m 14s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/4ff9c00922054d29bd6d4bffa6176f1a : FAILURE in 24m 02s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/28b1f84bc46b4191be0f9a4ef5271052 : SUCCESS in 14m 42s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/35542cf6114c4487adc685fdb6024d21 : SUCCESS in 14m 49s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9fc1d632d92f41099c8ab6d95afe89c2 : SUCCESS in 15m 59s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/a23b58b8aa2942eab56f60f8d1a91544 : SUCCESS in 16m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b9e276868de64c20b044f4ad95b72803 : SUCCESS in 6m 02s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4384925b532b4af782e4f17b0beb0845 : FAILURE in 32m 22s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/0abaebfde5994abf8dfbd71b127553e6 : SUCCESS in 8m 03s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8d5fcab8b0be4598b28d3ca7a1cc9c5a : SUCCESS in 59m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/75c93916702d4e238bd96c62cb794f1b : SUCCESS in 52m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3d20d424443d48e480e4f891faeef43c : SUCCESS in 51m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/19919a74a7b2472b8f0295790b153b37 : TIMED_OUT in 41m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30cf3ce3f6724253899acb6a5bafc827 : SUCCESS in 2m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dfd467a4e612491ca519a306ade25b49 : SUCCESS in 19m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/94228733e08149e0b920a5c490bedf04 : SUCCESS in 42m 50s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"c270aa305386159ceacb1fe9c1428dab08a1baf4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 01:14:46.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3635e0c0548d4fbf85d2c6d8983bb879\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/743fe4b3b0d647c1bb405f57681fce2d : FAILURE in 1h 39m 01s","accounts_in_message":[],"_revision_number":1},{"id":"2a78d03b541a7ee9d858850c240151aa83e73d30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-09-11 14:36:51.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"c9512079ed191fe86c20ad61575f96946fa3833e","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-09-11 14:37:18.000000000","message":"Patch Set 2: Code-Review+1\n\n(12 comments)","accounts_in_message":[],"_revision_number":2},{"id":"61d899993b5360db9a057b4651e6cd817c4e0d7e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 15:36:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e6ffc6a61914ef1805c68d816efa013\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/22913e2886fa466bb77b45a3ce1bac68 : SUCCESS in 57m 45s","accounts_in_message":[],"_revision_number":2},{"id":"c1eebb6d0d86d5a1de4439ff0492de176f5b6951","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 15:50:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9513a742361e43cbb835f2ecf4200143\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80e37fa05aa3420eae90a1387774504b : SUCCESS in 11m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c930eeb5a124f0da823a917aaffb20a : SUCCESS in 9m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/bf6493c482e840b0aedf36c12cbf8d3c : SUCCESS in 1h 07m 07s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/64e0ddf2b77e47408ad521dbbb525d66 : SUCCESS in 56m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e0a4ce80f47249279dcf979ce45a3135 : SUCCESS in 30m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0db4c3e21bea41cc8a26f944cc944900 : SUCCESS in 59m 19s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4ce38cd1722a49ed870b613350f5d06f : SUCCESS in 18m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4184f96735a141b9b2090631bb8d02f6 : SUCCESS in 12m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b51b7eaa3f76474488c7ed75c8d270f5 : SUCCESS in 22m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3dd6c549334f4d61804f654fae1c7092 : SUCCESS in 4m 52s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/034fc657030a4604b3feb6d8ed0e96bc : FAILURE in 32m 09s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/14691d4a466b414bb3a58deef959b023 : FAILURE in 25m 23s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c97cdfbe0ff2409497f8ef9a8c6fc256 : FAILURE in 19m 02s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/66c8cf3c9b334156ad4c4d411fddad3b : FAILURE in 31m 28s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b2c5ecd23ae24be7a9b5e67f5dbde543 : SUCCESS in 15m 15s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ecec72e8847045559f114ca575b2d1b9 : SUCCESS in 14m 39s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e9b1c7bd1fe64dadba8168d445e702fd : SUCCESS in 10m 03s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/ec6fe6f7592c4db0b95707800f96a0ee : SUCCESS in 18m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/78ca2072d52f4d86b73ae5e909696539 : SUCCESS in 3m 39s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a4f59b353d184761a6942aa8f971409a : FAILURE in 30m 37s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/93c8b78b5e094cf5a6df8fa87b04297a : SUCCESS in 5m 37s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a0c28e2d84944ccb816e361b02100425 : SUCCESS in 41m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7acaa8183c954e48b2d071dd5351330f : SUCCESS in 27m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/049d11d8cd0d4cf8a841b68d2e270399 : SUCCESS in 1h 01m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2b641c287b224cca834ef71899d7fe3b : TIMED_OUT in 41m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39cee74e7d3945c78f81ad13c87d6b48 : SUCCESS in 3m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ea63ec7ac6604667aa1edeb1330305c2 : SUCCESS in 19m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4be4dd34eac743f9883f9eeae8e61133 : SUCCESS in 1h 01m 36s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"8bc94c68e9cf9da9de0e68b0d6a150039b80469a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-09-15 02:29:37.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"20d29ec4497a51783f35c62f42f97009e4163da8","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-09-15 02:33:28.000000000","message":"Patch Set 3:\n\n(12 comments)","accounts_in_message":[],"_revision_number":3},{"id":"7d1a3182e26f2a57ea76160cc819d49965bbc71f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 04:02:57.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d58648d3e9c240db83e66ec21b7f6770\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b3d132149764db49ac116124d162493 : SUCCESS in 5m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d62ce93d874849d5b2b5af204d4b356e : SUCCESS in 10m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/8ece8db837814f259219f886bcd6c545 : SUCCESS in 42m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/eba7f16417de44a2ac5455048d441c88 : SUCCESS in 1h 02m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/43b31e6d347a4adaabc3b4890b0293ab : SUCCESS in 54m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d654b68ae24a4ccdabf57b76b504d60c : POST_FAILURE in 7m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5f3edbe40e1e43c6a80db95fa62e0f95 : SUCCESS in 19m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/92c9798025d443f4bdd7f38703b3e61f : SUCCESS in 12m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/75a6f9190f1f4622bef6f9fc45b72943 : SUCCESS in 21m 32s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1dd245c1115c4ea8b233356ac4a5ddb6 : SUCCESS in 8m 22s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/77bf4c1664ef44ba9cac8f8c3e291cb5 : FAILURE in 43m 00s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/4c99a892db114645b7e6bf7c6f1492e8 : FAILURE in 33m 22s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/9ec18afa178b46b98f34aa1118c9e951 : FAILURE in 38m 49s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/4e220da088ed4d2c9eeeb44206dc884e : FAILURE in 29m 29s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/89e6a529419c40918bfbc1a6c18e13dd : SUCCESS in 17m 18s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b6d95107a6714ff28ba36d1cfbfad328 : SUCCESS in 15m 36s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/26db5d2481fb44809531372e3ae198da : SUCCESS in 15m 35s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/492f49b8c9894ddd822da9c1a9086d33 : SUCCESS in 18m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e3a0534994094c918c6694b4a4fa930d : SUCCESS in 7m 50s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/eb3a33c8a5bc41c3a8fe8d71ef32fb36 : SUCCESS in 10m 18s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d7a17eda53974a1984063a0a1005d202 : SUCCESS in 46m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cd2eca9e3bd547ffb4985ed42a82eacc : SUCCESS in 39m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4965382a8d5d49dabb3b675aaa8bd73d : SUCCESS in 33m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/32eb6ee821924b5ca84ec6f3e0d42a9a : FAILURE in 16m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c0a0ab96a794fe19d13e5f131fda80f : SUCCESS in 4m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dd137c325c454de1a70d998a8714757b : SUCCESS in 19m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cc8e0c6ab04046879d93fb49eff63fa5 : SUCCESS in 28m 19s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":3},{"id":"b9d28315055b8e342e69a86ee2a1d89fc09cb2e3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 04:05:41.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d84bf0bb4c8642abae79ad138085a371\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/491cec47c95946339e18874cb0c21d67 : FAILURE in 53m 49s","accounts_in_message":[],"_revision_number":3},{"id":"90aebd83de0dfce0395e2be317f8d2508d138d85","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-09-15 15:20:22.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"29bb136994d0099add5ed42099aee33981736465","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 15:59:01.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/929a8fd435a143219343994f505344c2\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2dd74deb4a6b4e298a18cd3773898fed : SUCCESS in 35m 41s","accounts_in_message":[],"_revision_number":4},{"id":"15ed541a46d80d622f338590db8e63e4762ffb3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 16:42:24.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19cf0ae0b4964c9b817c11d9dfb6527a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8fddee3d33641f89153f13a9d300425 : SUCCESS in 6m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c1f8b085f41e4df086c8a664a0f2e845 : SUCCESS in 10m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/382f6ad19c85460f86e7e5aed56b317d : SUCCESS in 42m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b684dd7c28504e7d8e86200166851e24 : SUCCESS in 1h 07m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fed21a4c021b4f3b8279a251bb592704 : SUCCESS in 52m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1291b6ffc6a14c2295670fcb3fe39e08 : SUCCESS in 50m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b13b7f5c8e054b72be2f80b715f6212c : SUCCESS in 17m 41s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/324765f62a13478dbf9a163abffb8f3d : SUCCESS in 16m 24s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/15a7ab52ffdb4f7392cfcf2e6257ec38 : SUCCESS in 11m 08s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/331e36e3fbb249c49903e7a33d94c6a0 : SUCCESS in 7m 16s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/59ac800c90c34e8abc422ba0a0a0f1e6 : SUCCESS in 20m 53s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/6688695a43bd47b19776635ccc4d6fb0 : SUCCESS in 13m 23s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/ab81243319354696974e5809ac6528b9 : SUCCESS in 11m 41s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/5649b399960a4a58ba22c70eed0dce1d : SUCCESS in 9m 06s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f7eb5e2de1d946fb82c11ff95a628d61 : SUCCESS in 9m 51s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/eb8f3ff3cc194b3aa2aecfdc035ba310 : SUCCESS in 8m 29s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7c5d24af8adb4362be7c4f2525373775 : SUCCESS in 15m 43s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/3f82b9809bc94819bd864da3e83f829e : SUCCESS in 9m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/86c985f7af854665bb8f7e0c41791060 : SUCCESS in 3m 40s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/126a08243f6b4dfc98388548ad2f8b07 : SUCCESS in 9m 31s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7710a6904fa848709dc6137902ed9b96 : SUCCESS in 1h 10m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/22a0b52cde87494ebf1846f04e8db17d : SUCCESS in 30m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f53152b15d6a446cad2d4fd63898cf3c : SUCCESS in 55m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f465ee2f44ec4c259022766d2dff7c8d : SUCCESS in 11m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5ef45e5d7b042059c8a76c57a8ca1a3 : SUCCESS in 4m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1fdafbd5be3f4ed1ba60685c3c5054b5 : SUCCESS in 20m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/de0d1d8ecdb643249163add0436a9b0d : SUCCESS in 1h 05m 24s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":4},{"id":"d3a3cb2c9fce944b92da4c39b0a3f4a9fd933f33","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-09-15 22:33:52.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(12 comments)","accounts_in_message":[],"_revision_number":4},{"id":"7a1b0b032cafeec5d86149d65602932fbaf8a78d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 22:34:53.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"af16104f2b2a1aafc48010fbf7ff17f6d3d76946","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 00:00:34.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3bdd060108e24d3ca59b8484a61ff173\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/487f734611924653a2404291832b29c6 : SUCCESS in 6m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/57d7074f136142e2a3ed0ca46c880b89 : SUCCESS in 5m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/bb2ffd62660341a5a16e4826f7be922b : SUCCESS in 49m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2c8cab4310734659ab648ee17e415787 : SUCCESS in 49m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/43399f03e9364a6a8d73c6643766f508 : SUCCESS in 48m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9f5cebee899b4a9badd750cdefebc148 : SUCCESS in 32m 52s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/d9e07cdcf0074101bcfb5c33b1c06c60 : SUCCESS in 8m 51s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3d084fc836a3407b931aa95542e3c85b : SUCCESS in 24m 57s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/f7d4b1e2cae54fc49587786cff802dbb : SUCCESS in 22m 28s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/29ec1cc8a27248c4bfc484c6f8aeb50e : SUCCESS in 25m 30s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/3bd64456a1b14e2f9977fafa045fe721 : SUCCESS in 21m 56s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/931e6ccee55a456bb872538edb523b2a : SUCCESS in 15m 28s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/40c9ccb60a96483984ae052335b51fb9 : SUCCESS in 9m 42s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c813cabb4e4746efb50ec88392fc375a : SUCCESS in 15m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b2628718dddc4261865cf3ceeaa9accf : SUCCESS in 6m 11s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/c8e8a0948f114b61a0e3bdec294a78c6 : SUCCESS in 8m 12s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ea4580bc69654eaea0efbdca22c1a7e5 : SUCCESS in 51m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/25a006989f514f22af0b253f3df5897e : SUCCESS in 49m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/96a20637701c43eb9ed0f95df89d2de4 : SUCCESS in 55m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/156edfb091ee42f4aba6f0ca043f1972 : SUCCESS in 29m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f59065891e04b7aaab18d4530150188 : SUCCESS in 4m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7775b777e6224ea3bbd70cebfb10004e : SUCCESS in 9m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/067fe7ee6c8a4979837f63fbd9207e4c : SUCCESS in 49m 47s","accounts_in_message":[],"_revision_number":4},{"id":"327dc76cec8696bc962004b73deb4c64d26eb8be","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 00:00:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"fc8942a39856bc391edd91d8d84de9ca21a9c639","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 00:02:22.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c47366c35c249dfa366edb8b3aab0f9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/365a4fed52b64ce1b3a1daa940922028 : SUCCESS in 47s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/9d2d5b26189a4a92b644dd3ce51aea88 : SUCCESS in 51s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"bd664103928dde998f7b79ec73c572153c8b1b52","revisions":{"9a2d198e0130860bc8403f83e96bdefe68858fa9":{"kind":"REWORK","_number":1,"created":"2026-09-10 23:33:15.000000000","uploader":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"ref":"refs/changes/45/1005145/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/45/1005145/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/45/1005145/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/45/1005145/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/45/1005145/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/45/1005145/1"}}},"commit":{"parents":[{"commit":"be191cf7d452c852995b21b1e6584727bc729388","subject":"replicator: Quarantine suffixes we cannot delete","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/be191cf7d452c852995b21b1e6584727bc729388"}]}],"author":{"name":"Wael Halbawi","email":"whalbawi@nvidia.com","date":"2026-09-10 23:19:01.000000000","tz":-420},"committer":{"name":"Wael Halbawi","email":"whalbawi@nvidia.com","date":"2026-09-10 23:27:29.000000000","tz":-420},"subject":"sq? replicator: Continue handoff cleanup after quarantine failures","message":"sq? replicator: Continue handoff cleanup after quarantine failures\n\nCatch quarantine failures per suffix so that already-replicated objects\nin other suffixes can still be removed during the same pass.\n\nChange-Id: Idb10b7307bcf605a6290cc1c1446ab7ffd62cd15\nSigned-off-by: Wael Halbawi \u003cwhalbawi@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9a2d198e0130860bc8403f83e96bdefe68858fa9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9a2d198e0130860bc8403f83e96bdefe68858fa9"}]},"branch":"refs/heads/master"},"904829a948e57c40f7dc82a7dec08145dc8c7058":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-09-11 14:36:51.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/45/1005145/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/45/1005145/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/45/1005145/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/45/1005145/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/45/1005145/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/45/1005145/2"}}},"commit":{"parents":[{"commit":"be191cf7d452c852995b21b1e6584727bc729388","subject":"replicator: Quarantine suffixes we cannot delete","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/be191cf7d452c852995b21b1e6584727bc729388"}]}],"author":{"name":"Wael Halbawi","email":"whalbawi@nvidia.com","date":"2026-09-10 23:19:01.000000000","tz":-420},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-09-11 14:36:51.000000000","tz":0},"subject":"replicator: Continue handoff cleanup after quarantine failures","message":"replicator: Continue handoff cleanup after quarantine failures\n\nCatch quarantine failures per suffix so that already-replicated objects\nin other suffixes can still be removed during the same pass.\n\nChange-Id: Idb10b7307bcf605a6290cc1c1446ab7ffd62cd15\nSigned-off-by: Wael Halbawi \u003cwhalbawi@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/904829a948e57c40f7dc82a7dec08145dc8c7058"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/904829a948e57c40f7dc82a7dec08145dc8c7058"}]},"branch":"refs/heads/master","description":"Edit commit message"},"4932bf1e50357fd7d5418b4244a10c88ea7dda4f":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-09-15 02:29:37.000000000","uploader":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"ref":"refs/changes/45/1005145/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/45/1005145/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/45/1005145/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/45/1005145/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/45/1005145/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/45/1005145/3"}}},"commit":{"parents":[{"commit":"ef169bada576dbe7a6b87da4bf94c7431b448b16","subject":"Merge \"memcached: do not grow the pool after a failed connect\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ef169bada576dbe7a6b87da4bf94c7431b448b16"}]}],"author":{"name":"Wael Halbawi","email":"whalbawi@nvidia.com","date":"2026-09-10 23:19:01.000000000","tz":-420},"committer":{"name":"Wael Halbawi","email":"whalbawi@nvidia.com","date":"2026-09-15 02:27:56.000000000","tz":-420},"subject":"replicator: Continue handoff cleanup after quarantine failures","message":"replicator: Continue handoff cleanup after quarantine failures\n\nCatch quarantine failures per suffix so that already-replicated objects\nin other suffixes can still be removed during the same pass.\n\nChange-Id: Idb10b7307bcf605a6290cc1c1446ab7ffd62cd15\nSigned-off-by: Wael Halbawi \u003cwhalbawi@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4932bf1e50357fd7d5418b4244a10c88ea7dda4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4932bf1e50357fd7d5418b4244a10c88ea7dda4f"}]},"branch":"refs/heads/master"},"bd664103928dde998f7b79ec73c572153c8b1b52":{"kind":"REWORK","_number":4,"created":"2026-09-15 15:20:22.000000000","uploader":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"ref":"refs/changes/45/1005145/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/45/1005145/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/45/1005145/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/45/1005145/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/45/1005145/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/45/1005145/4"}}},"commit":{"parents":[{"commit":"ef169bada576dbe7a6b87da4bf94c7431b448b16","subject":"Merge \"memcached: do not grow the pool after a failed connect\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ef169bada576dbe7a6b87da4bf94c7431b448b16"}]}],"author":{"name":"Wael Halbawi","email":"whalbawi@nvidia.com","date":"2026-09-10 23:19:01.000000000","tz":-420},"committer":{"name":"Wael Halbawi","email":"whalbawi@nvidia.com","date":"2026-09-15 02:28:18.000000000","tz":-420},"subject":"replicator: Continue handoff cleanup after quarantine failures","message":"replicator: Continue handoff cleanup after quarantine failures\n\nCatch quarantine failures per suffix so that already-replicated objects\nin other suffixes can still be removed during the same pass.\n\nChange-Id: Idb10b7307bcf605a6290cc1c1446ab7ffd62cd15\nSigned-off-by: Wael Halbawi \u003cwhalbawi@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bd664103928dde998f7b79ec73c572153c8b1b52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bd664103928dde998f7b79ec73c572153c8b1b52"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
