)]}'
{"id":"openstack%2Fswift~1005767","triplet_id":"openstack%2Fswift~master~I58f85818126ce921ca139089662b4c740033bce5","project":"openstack/swift","branch":"master","topic":"p-s3api-content-length-missing","attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-09-15 18:19:27.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I58f85818126ce921ca139089662b4c740033bce5","subject":"WIP s3api: check for missing content-length","status":"NEW","created":"2026-09-15 17:10:23.000000000","updated":"2026-09-15 18:19:27.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"19fa93cf774918ad9e3438ffe05c6ed2d562cacf","_number":1005767,"virtual_id_number":1005767,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-15 18:19:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":-1,"date":"2026-09-15 17:19:28.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-15 17:51:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-15 18:19:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3d41ebd93390a8cdcd354dc58d52102891acde1e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-09-15 17:10:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"72a7f08c213434e2d8c3341ba60e8d4ef0e79f6e","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-09-15 17:19:28.000000000","message":"Patch Set 1: Workflow-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"919121a86a41d7adc65cd0f0d2af5a4bd73afe9c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 17:51:51.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9474c9890bf74254a8e400903e4b6640\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/068071f5e5404d189d0e765e519a9754 : FAILURE in 38m 55s","accounts_in_message":[],"_revision_number":1},{"id":"19fa93cf774918ad9e3438ffe05c6ed2d562cacf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 18:19:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d86d300773004f3c83a58233d32894f6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bcb27f428724b61a5854648379bfaea : SUCCESS in 9m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5381f28a11aa4126a6067355eb127f12 : SUCCESS in 10m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/d8a84cd5914e494884c9459806ba6c2e : SUCCESS in 1h 05m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/17b6024516814563b4a8dde59a92c99a : SUCCESS in 37m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f4b7726cd6fd461dba61c09f5b8f7d8f : SUCCESS in 32m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c8bac81fc6bf496c8ec0398224751217 : SUCCESS in 48m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9026e57ac98d47a0b3c3a7c71254e3e8 : SUCCESS in 10m 50s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5fb880419ad44a72b0de1aaec5220c73 : SUCCESS in 11m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e8ace572153d42cd8caa20c3137c68dc : SUCCESS in 20m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3662dcf3b9854af6b7155c3c4ea54f75 : SUCCESS in 4m 42s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ab747fddd0ea4caebc1c19ca6aee204a : FAILURE in 22m 34s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/711e366e8367452b9b8e312875d1a830 : FAILURE in 14m 43s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/e6e1287fce3b472db61536fcf44c9c91 : FAILURE in 33m 19s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/ebdb0dc6e52943d7ab1877eb543598fd : FAILURE in 11m 56s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0c6d0563cb034986bda437ebaf8372bd : SUCCESS in 9m 56s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/834f1fc946a346a2bfc1d96ff88d0f95 : SUCCESS in 10m 41s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/2ef09b6cb93c409c8db5c059fba783f8 : SUCCESS in 8m 26s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/fec2b0f72d8a45b9bb9aec0336a31949 : SUCCESS in 9m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2fe8af580c9c45269e9ebc20ae719b59 : SUCCESS in 6m 38s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e64fa79786094e3fb3606b8773631d09 : FAILURE in 8m 15s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/31b3a0abe0f941e58caf325decab9c50 : SUCCESS in 41m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c736543e2b9a4b8fb0012c6e4c2a8baf : SUCCESS in 53m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ca94b49de6684da88a4c1fa614be3a79 : SUCCESS in 59m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f3617e626fc453d8ad259dfbef67c12 : FAILURE in 11m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0577f13e9559442689f42fb1141bdd9d : SUCCESS in 3m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ff790c9fcf7a49e6a0b9e7e23ec97755 : SUCCESS in 27m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9c43851117c046bdb20591b82d3218a7 : SUCCESS in 53m 44s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4e3ad03058d420bf17c0edc8a829fd6517194f09","revisions":{"4e3ad03058d420bf17c0edc8a829fd6517194f09":{"kind":"REWORK","_number":1,"created":"2026-09-15 17:10:23.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/67/1005767/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/1005767/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/1005767/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/1005767/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/1005767/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/1005767/1"}}},"commit":{"parents":[{"commit":"ef169bada576dbe7a6b87da4bf94c7431b448b16","subject":"Merge \"memcached: do not grow the pool after a failed connect\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ef169bada576dbe7a6b87da4bf94c7431b448b16"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-09-15 15:01:47.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-09-15 17:08:35.000000000","tz":60},"subject":"WIP s3api: check for missing content-length","message":"WIP s3api: check for missing content-length\n\nCheck for missing content-length *before* installing input hashers\nthat assume it is available.\n\nAlso, raise MissingContentLength instead of MalformedTrailerError when\na request has both errors.\n\nChange-Id: I58f85818126ce921ca139089662b4c740033bce5\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4e3ad03058d420bf17c0edc8a829fd6517194f09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4e3ad03058d420bf17c0edc8a829fd6517194f09"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
