)]}'
{"id":"openstack%2Fswift~1005968","triplet_id":"openstack%2Fswift~master~Id6bd419ca7134d5e4550fb98ec0820bf89a8be15","project":"openstack/swift","branch":"master","topic":"s3_checksum_algo_base_labels","attention_set":{},"removed_from_attention_set":{"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2026-09-17 17:27:55.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"Id6bd419ca7134d5e4550fb98ec0820bf89a8be15","subject":"s3api: Use base labels for checksum metrics","status":"NEW","created":"2026-09-16 22:28:03.000000000","updated":"2026-09-17 17:27:55.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":6,"unresolved_comment_count":5,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"cbde4143368e4f6daed20514ede2760822312161","_number":1005968,"virtual_id_number":1005968,"owner":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-16 23:39:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":-1,"date":"2026-09-17 12:12:56.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-16 23:39:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-17 12:11:10.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2026-09-17 12:12:56.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"f5f1f55192ccc9c9e8dbc24bc5309f445b0fcfba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-09-16 22:28:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"841f7d3cf0ce04a73bb03df9800f51c2d1010423","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-09-16 22:28:18.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"ee6d0757e85c87f0d68f1c3642a7cca549ecf17d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 23:39:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d025bed8f8941f99f0c5051c5af3f5d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfeb5375d15841fdba8dccd721360c3e : SUCCESS in 10m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce0d03dc13b340998fa544e7883a8002 : SUCCESS in 3m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/dd1b6dd98b34452db87e51aabc31696b : SUCCESS in 1h 01m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7e9b78a4570c4301af0b874e55672ac6 : SUCCESS in 1h 05m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/74f3bf1945964cfebcbb6dca464f6145 : SUCCESS in 54m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1a95f081292640c89921b724ddcbff66 : SUCCESS in 41m 12s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c7bc6d22473a4188908c66298975b487 : SUCCESS in 18m 15s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8585f10048734858868824519164d9f6 : SUCCESS in 16m 56s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/06ac423da5cc4035b5a29de29f7f2784 : SUCCESS in 11m 52s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a807e7ec3df54bab9810b781aaaa21ac : SUCCESS in 6m 34s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/eabc34b0d4d64d21926300deef825d87 : SUCCESS in 26m 19s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/8c4db58b8d0a40408b5941b95051ddc4 : SUCCESS in 14m 02s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c53b0ccbffc84cccbafdfc2d936bfdd8 : SUCCESS in 20m 00s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/f26cd5a45f6f4763b7391a404dbb8541 : SUCCESS in 22m 11s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a4d4acc6b1884649b1b6c01e4f235e83 : SUCCESS in 16m 14s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9036291610af4e10823b25e6aeb58181 : SUCCESS in 15m 38s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/02dda1ced03f4cd9adddf0abe8b44c9e : SUCCESS in 16m 15s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/0757c94d03594829a938175626737f9d : SUCCESS in 17m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/19e05a8e18da4197a6d1592bfe68b093 : SUCCESS in 8m 16s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/0fc454403d3b437c8a7593d27d09675c : SUCCESS in 11m 22s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6c7a25d775564736b099964d6b3bfc04 : SUCCESS in 38m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8b8595321fe54e978d75adf42edbd61e : SUCCESS in 50m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1ec1a91840f0425fbecd5fb32ab57e33 : SUCCESS in 52m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fca55b02bcd14335b11b5298d7242dde : SUCCESS in 12m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6eca60fcd7b44d3b98c06c791bff4aa : SUCCESS in 3m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d805ea927d0f40f9be5c56d2b7ce1b66 : SUCCESS in 14m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/19e0f52a530f47819709c1dfff3ef912 : SUCCESS in 51m 10s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"47950a51250e905beb59b6391503aeb749d4237f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 00:06:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/befe3fa696ef48639236389a561906f0\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/689053e80ff642d4861084ab6381b7a5 : SUCCESS in 58m 14s","accounts_in_message":[],"_revision_number":1},{"id":"d3518b6b436837b06015ebb25f4b3dea23d90591","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-09-17 12:11:10.000000000","message":"Patch Set 1:\n\n(6 comments)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"c6a61bd8422d91013fe9c28bae1978cc0cf244a0","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-09-17 12:12:56.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"cbde4143368e4f6daed20514ede2760822312161","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-09-17 17:27:55.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"875364b80483d79c11beb37f87d67aeed65dd4df","revisions":{"875364b80483d79c11beb37f87d67aeed65dd4df":{"kind":"REWORK","_number":1,"created":"2026-09-16 22:28:03.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/68/1005968/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/68/1005968/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/1005968/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/1005968/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/1005968/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/68/1005968/1"}}},"commit":{"parents":[{"commit":"ef169bada576dbe7a6b87da4bf94c7431b448b16","subject":"Merge \"memcached: do not grow the pool after a failed connect\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ef169bada576dbe7a6b87da4bf94c7431b448b16"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2026-09-10 23:16:37.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2026-09-16 22:27:16.000000000","tz":-420},"subject":"s3api: Use base labels for checksum metrics","message":"s3api: Use base labels for checksum metrics\n\nUse the shared base labels for swift_s3_checksum_algo_request.\nKeep the existing label setup as a fallback for pipelines that\ndo not use base labels. Rename the type label to resource for\nconsistency.\n\nChange-Id: Id6bd419ca7134d5e4550fb98ec0820bf89a8be15\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/875364b80483d79c11beb37f87d67aeed65dd4df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/875364b80483d79c11beb37f87d67aeed65dd4df"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
