)]}'
{"id":"openstack%2Fswift~1006009","triplet_id":"openstack%2Fswift~feature%2Fmpu~Ib0e2fa83132464a5374cfd46867f2ba753b16ab6","project":"openstack/swift","branch":"feature/mpu","topic":"merge-master","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-09-18 22:48:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib0e2fa83132464a5374cfd46867f2ba753b16ab6","subject":"Merge remotes/origin/master into feature/mpu","status":"MERGED","created":"2026-09-17 10:08:27.000000000","updated":"2026-09-18 22:50:20.000000000","submitted":"2026-09-18 22:48:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"1006009-merge-master","meta_rev_id":"455e0c07af6e80ce082951146eea1643570039c8","_number":1006009,"virtual_id_number":1006009,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-18 22:48:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-18 21:59:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-09-18 21:59:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-17 11:20:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-18 21:59:48.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"34cde527d6e4cde2ea52a344d6e794e3fb1595b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-09-17 10:08:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bca91449e746ff92b78f2f97bbbd13a504624fb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 11:20:48.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7c20153895f74c62a29bc94b0bdb3f8d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58f22b50fe6a485a9efa6ebf80ab7845 : SUCCESS in 10m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c6d557afb4f14dce936cfcd46399ec6b : SUCCESS in 10m 12s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/22e48b7c0afe41d7bd949c1b4a80b288 : SUCCESS in 21m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b861d3ed36914609afbeb7e00b960867 : SUCCESS in 22m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2b6a05a6cde847f598b612b6c5aebfd3 : SUCCESS in 23m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/70e2c0a5713a426f82c4becbfdd32716 : SUCCESS in 6m 41s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a2f976165de046d8a58c7a125e348597 : SUCCESS in 30m 01s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/7d02368855884210886552d7792ff9cc : SUCCESS in 23m 33s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/09184368695f4aae8e4bd473f31f620e : SUCCESS in 27m 43s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/dc78f97e21dd4a29b5e671dd97fe5190 : SUCCESS in 21m 44s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a58f0b91310e447384028ce21e934aee : SUCCESS in 14m 54s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c1f5d87ebac74e51a52e5a163151298d : SUCCESS in 16m 16s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/76011500c61f482e9ff81429454847c2 : SUCCESS in 16m 35s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/802c283867b74a9da6ec323531f391d4 : SUCCESS in 18m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/13bf799b3c484ee38e2b70534f68e673 : SUCCESS in 6m 03s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/6633d35e7a2f4219aec7a11c2b4d8add : SUCCESS in 8m 54s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5660ac6aa150461b9bb21c40b82f8661 : FAILURE in 1h 05m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9e9dfaf638334a4abe43bba52c8c75a2 : SUCCESS in 29m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4c92943c60947239b43a54e3c31a36e : FAILURE in 4m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1290d114de3b4d91b16b566a5fc350c9 : SUCCESS in 20m 09s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"77add4529494dd70f0a564898b541917b1f6da88","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 11:23:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e35b751a68d44a869a821a2f2bc90ad4\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ee4501a034f848ff8b41f108db02bc80 : SUCCESS in 1h 07m 13s","accounts_in_message":[],"_revision_number":1},{"id":"2affcd8bc6336f788a329be5f194a7ae9c483b72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-09-17 15:57:08.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"7bdcff465d0a9f2fa3a06253e48f636c8c627d1f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 16:33:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de283a7d2340423ea7b056c5bfbcf9f4\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e542e29752014ec98c862d71ab734dcb : SUCCESS in 34m 03s","accounts_in_message":[],"_revision_number":2},{"id":"2089131041b8305dd1937edd5d76ea7bfc814e52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 17:08:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03ea21c2ee1943e886c37cc444b6a8aa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edf2a8deab23435ba8d55820ef667ff4 : SUCCESS in 5m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2944652cc4d64b8ca0c4af3d316f8c61 : SUCCESS in 9m 29s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e039534a427f4624acad4dfa32a36fda : SUCCESS in 19m 48s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c4862a785b134c739696a37c79209c2b : SUCCESS in 19m 08s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e2995f02838042f7967da350bcdbe2af : SUCCESS in 22m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0034e1ccbd2e4938996ef632eaa26c3a : SUCCESS in 6m 05s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/fcbe6e35eef5400e8780389b3ed455f7 : SUCCESS in 26m 21s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/dd6b0d241a624290bbd73238bc29aa46 : SUCCESS in 22m 28s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/06c58157e4ef40d2bb4bdb7a664b85a7 : SUCCESS in 25m 13s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/59f1683241a44762bfb9a136120fa563 : SUCCESS in 10m 01s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/8d87ab40e26f4c09980601e8b14256d0 : SUCCESS in 15m 52s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3b292d565d4a4243b5df44083ba52c80 : SUCCESS in 15m 29s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/02be5933cc0f4cb6a999da87e818bb80 : SUCCESS in 15m 44s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/c1cc6b63785248a0835fe40dbf16147f : SUCCESS in 18m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e2743efae6e2492db1010adba3114c79 : SUCCESS in 6m 30s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/79e9ca3c354d44f0b40be18e3e7fd636 : SUCCESS in 9m 05s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/21223e82faa346b7a8903946560462a8 : SUCCESS in 1h 05m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/93042f46211e414586f9e1cc41e2d7ce : SUCCESS in 22m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38be3304514c42d4b1bf9174edab3947 : SUCCESS in 5m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4d084a284201469593e57006fea1615b : SUCCESS in 16m 16s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"2ef07d38b0c3316d8019a16f782481fa995a0975","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-09-18 16:21:22.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"26dbf26f69bf24abef1203e9014314ddec6e329b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 17:02:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee7bb0093c3d4288ba7774586ba04d0c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23e5d67b845644ed9837b67900205ca3 : SUCCESS in 10m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e77388e0db8447ed8f2a7a6dec876dcc : SUCCESS in 9m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/65fc496c5bac464fa1453156425b41aa : SUCCESS in 11m 06s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9ca2c2b32b33443daa7bbc0a5a4d3169 : SUCCESS in 21m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d6a105c44a274bf1b6f8fa3a43f318cc : SUCCESS in 19m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/51dd8aadffe14236b627297294b93d12 : SUCCESS in 5m 25s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a8cee06f4ea64164a1cccdb17ee6c0e1 : SUCCESS in 18m 07s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/89528df23f8b4cb88e165ee179b03747 : SUCCESS in 24m 03s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/39e89b99c3bf45e7a0a8d3f4d260df96 : SUCCESS in 12m 46s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/3f9cb8db86cb42c4a5f6b50576d6fce6 : SUCCESS in 20m 00s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/50f186872d3a4569a950b6ce0c131be1 : SUCCESS in 15m 50s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ea548ebc183c49abb6ffab24bc6bd20f : SUCCESS in 16m 32s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d6252407d8e641dabfa84acd5ae7564e : SUCCESS in 17m 01s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/89e9d0cdd091401182748e7ae0080710 : SUCCESS in 11m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6abed32f3d43462e886c9837fab9693d : SUCCESS in 5m 47s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/336330d732c7454d865a71fa154565e2 : SUCCESS in 5m 42s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/47678e1f48d74202957a119e61fc2038 : SUCCESS in 39m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fba3a7d0b93547f6a16b8711f6539bf1 : SUCCESS in 15m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7deee83af41422098963499006ff5ba : SUCCESS in 4m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f56b23099ce6439a92b4f509408af9e2 : SUCCESS in 10m 01s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":3},{"id":"e7f0b3967d60a25010d85378e3448b54f484bc3a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 17:17:26.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a1b1aa8ffef499f9061622f24f23ac9\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1353c57609ad4a03abc98959fe4457dd : SUCCESS in 51m 17s","accounts_in_message":[],"_revision_number":3},{"id":"0b0af20dbfe309cc4d314e576c483d3ea81a0f85","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-18 21:59:48.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"def935790972438615f00fb3cc0afcaf6da051c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 22:00:19.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"75eca9c15bb268b6153065e670ca9c28ad3c0277","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 22:48:19.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7695544b7a3942ceb2b0faa5e29d622f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2659341390e0439c85e89bdd3a068192 : SUCCESS in 6m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b4254db639c4dc18547de1713cc352f : SUCCESS in 3m 48s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/9b6cddbc99ad46019a9bcb3f57557f13 : SUCCESS in 4m 51s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/37de4149f35e49a1beb3b22f778f61b3 : SUCCESS in 14m 13s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/6ddc5bc881d54de5a0cbeda625846620 : SUCCESS in 11m 40s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/6e92803ef98d427ba3f848c7df465ff1 : SUCCESS in 33m 41s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/35d76750fa1f4e488634d9d3d086ebca : SUCCESS in 20m 50s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/43e8f81f35444df090485aea6ada1fee : SUCCESS in 9m 49s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8052231f95224058a360125a01a05dab : SUCCESS in 9m 01s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4c3ffccfb5134f548ab05054d22e3ec1 : SUCCESS in 9m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b930fabebb1b40218f67e93bfb2dc776 : SUCCESS in 3m 38s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/72320100d927499c8c0ed79a28323911 : SUCCESS in 5m 34s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4adcde842b0f41008d173d1d314169fe : SUCCESS in 47m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a99da1c830f040ce95632efd27a4d657 : SUCCESS in 31m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ba5a6ab1d45441d82fc490f42dd940b : SUCCESS in 3m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/09f2f6d2b6a64d66b06465e116fba9f2 : SUCCESS in 9m 50s","accounts_in_message":[],"_revision_number":3},{"id":"e954ccb144f5d0f49f39e412c9a2163ecdca19ce","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 22:48:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"455e0c07af6e80ce082951146eea1643570039c8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 22:50:20.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75cb613df67e42ccb3d9476ae6f179da\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5d713617e954479b58d1d8d70579524 : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5fa7212965484d93b6f51ea91677a2d4 : SUCCESS in 1m 14s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/aaeda81eea6f4b00a3f864c390a49cfb : SUCCESS in 54s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c6c7138a0fbba2bba24f04d8002dacbbbb417a48","revisions":{"78a521da5fe4bf9085c65de3445a326e250ed9b6":{"kind":"REWORK","_number":1,"created":"2026-09-17 10:08:27.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/09/1006009/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/09/1006009/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/1006009/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/1006009/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/1006009/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/09/1006009/1"}}},"commit":{"parents":[{"commit":"d6c596b0b6e47a091788109b9ddfc1478e613366","subject":"Merge \u0027remotes/origin/master\u0027 into feature/mpu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d6c596b0b6e47a091788109b9ddfc1478e613366"}]},{"commit":"bd664103928dde998f7b79ec73c572153c8b1b52","subject":"replicator: Continue handoff cleanup after quarantine failures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bd664103928dde998f7b79ec73c572153c8b1b52"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-09-17 09:48:51.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-09-17 10:02:53.000000000","tz":60},"subject":"Merge remotes/origin/master into feature/mpu","message":"Merge remotes/origin/master into feature/mpu\n\nChange-Id: Ib0e2fa83132464a5374cfd46867f2ba753b16ab6\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/78a521da5fe4bf9085c65de3445a326e250ed9b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/78a521da5fe4bf9085c65de3445a326e250ed9b6"}]},"branch":"refs/heads/feature/mpu"},"11312617a4225d166edad5acd7d96a5439e7b91b":{"kind":"REWORK","_number":2,"created":"2026-09-17 15:57:08.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/09/1006009/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/09/1006009/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/1006009/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/1006009/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/1006009/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/09/1006009/2"}}},"commit":{"parents":[{"commit":"d6c596b0b6e47a091788109b9ddfc1478e613366","subject":"Merge \u0027remotes/origin/master\u0027 into feature/mpu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d6c596b0b6e47a091788109b9ddfc1478e613366"}]},{"commit":"bd664103928dde998f7b79ec73c572153c8b1b52","subject":"replicator: Continue handoff cleanup after quarantine failures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bd664103928dde998f7b79ec73c572153c8b1b52"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-09-17 09:48:51.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-09-17 15:55:58.000000000","tz":60},"subject":"Merge remotes/origin/master into feature/mpu","message":"Merge remotes/origin/master into feature/mpu\n\nChange-Id: Ib0e2fa83132464a5374cfd46867f2ba753b16ab6\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/11312617a4225d166edad5acd7d96a5439e7b91b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/11312617a4225d166edad5acd7d96a5439e7b91b"}]},"branch":"refs/heads/feature/mpu"},"c6c7138a0fbba2bba24f04d8002dacbbbb417a48":{"kind":"REWORK","_number":3,"created":"2026-09-18 16:21:22.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/09/1006009/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/09/1006009/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/1006009/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/1006009/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/1006009/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/09/1006009/3"}}},"commit":{"parents":[{"commit":"d6c596b0b6e47a091788109b9ddfc1478e613366","subject":"Merge \u0027remotes/origin/master\u0027 into feature/mpu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d6c596b0b6e47a091788109b9ddfc1478e613366"}]},{"commit":"bd664103928dde998f7b79ec73c572153c8b1b52","subject":"replicator: Continue handoff cleanup after quarantine failures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bd664103928dde998f7b79ec73c572153c8b1b52"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-09-17 09:48:51.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-09-18 16:21:12.000000000","tz":60},"subject":"Merge remotes/origin/master into feature/mpu","message":"Merge remotes/origin/master into feature/mpu\n\nChange-Id: Ib0e2fa83132464a5374cfd46867f2ba753b16ab6\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c6c7138a0fbba2bba24f04d8002dacbbbb417a48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c6c7138a0fbba2bba24f04d8002dacbbbb417a48"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
