)]}'
{"id":"openstack%2Fswift~1006018","triplet_id":"openstack%2Fswift~master~I9d1ea362a0b0dc4207e5f3fe95fb531339ba5b18","project":"openstack/swift","branch":"master","topic":"s3_checksum_algo_base_labels","attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-09-17 13:19:18.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I9d1ea362a0b0dc4207e5f3fe95fb531339ba5b18","subject":"sq? wip: re-use base_labels helpers in s3api","status":"NEW","created":"2026-09-17 12:08:22.000000000","updated":"2026-09-17 13:54:29.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"59a2e2da349150cd13ecc3dbbea9a23696530b59","_number":1006018,"virtual_id_number":1006018,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-17 13:19:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-17 13:19:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d27b3acd66572ca4b5d5fc50aeba61a95fb88008","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-09-17 12:08:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f4246830cd745319efaca55956103d373c7ca312","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-09-17 12:12:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f03e8cc32fb99cbad51bac66d44bbbe4ac8c33d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 13:19:18.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c752374c343c4c5db5f65f77dd208588\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9034827f03bd402ca794a0db58abd057 : SUCCESS in 10m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a0acc1eef74e4ddd9a28eb141a52fb3e : SUCCESS in 9m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/ea8b4c192000425abc8cf194d45a43ab : SUCCESS in 1h 03m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/01b09172e31f489689bddad9e5f78e27 : SUCCESS in 1h 05m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/da09e255a9804618ba2b5cee8e453aa2 : SUCCESS in 51m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9bac5e9f1b2e4397ac0a9b172c54d44e : SUCCESS in 49m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/259032253e0b4cff94b908fca8c21ab2 : SUCCESS in 20m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/aca7f7feabf3436bbbbfb94524c33c94 : SUCCESS in 19m 58s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/77c36a9fc7ef435fb973c49650b55cdb : SUCCESS in 21m 25s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/172032391bc2416e948b1d85aa884111 : SUCCESS in 6m 44s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/918ba724525a431ea93f19a2a85c05cd : FAILURE in 28m 56s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ddc12dc9b2044ffab5b3e7edc264783b : FAILURE in 21m 57s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/15f64ab132fa4f54b96f4f59783201f3 : FAILURE in 31m 49s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/e89c2bb0b60a4b73b6313181509bf18d : FAILURE in 20m 21s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5c13592b587540098a5ed078260dc7cf : SUCCESS in 15m 33s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/fee54bbe8a274f028b1a7971d1290b96 : SUCCESS in 15m 30s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/46baa40dd53d468496200b91cc7f482b : SUCCESS in 13m 41s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/c663398d849749b8bbbdf06ff8630ad3 : SUCCESS in 18m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4aa5d0f15de348a3addeb0cb4a2d3a6f : SUCCESS in 6m 36s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/16872ebb256e417e9092f36f51cef50b : SUCCESS in 8m 54s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c5ff115fef4245e5bc305b185ae8e76a : SUCCESS in 1h 01m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/81c22641e18f4a548a42de53f612d227 : SUCCESS in 55m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/244089ce257a40f28593e8667687a0dd : SUCCESS in 53m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3a522c334cdb4747ad6786a89800fd5c : FAILURE in 30m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a62b168d191490c88e065a07d29082a : SUCCESS in 4m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d752c5a8f6cb430a9fb513a15a75993c : SUCCESS in 20m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f5207898a31f48ca99f1cd78345621a5 : SUCCESS in 57m 34s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"59a2e2da349150cd13ecc3dbbea9a23696530b59","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 13:54:29.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a25224c61ce644078fec22e7802ff108\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8531b4ec54eb456d8c2f91bdda4e9779 : FAILURE in 1h 00m 19s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0ac23c21b38c457521f95908a0193a7e11ecb910","revisions":{"8546542375dfba142c6beed19e8f74f7ad7e8fbf":{"kind":"REWORK","_number":1,"created":"2026-09-17 12:08:22.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/18/1006018/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/1006018/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/1006018/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/1006018/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/1006018/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/1006018/1"}}},"commit":{"parents":[{"commit":"875364b80483d79c11beb37f87d67aeed65dd4df","subject":"s3api: Use base labels for checksum metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/875364b80483d79c11beb37f87d67aeed65dd4df"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-09-17 12:04:15.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-09-17 12:04:15.000000000","tz":60},"subject":"sq? wip: re-use base_labels helpers in s3api","message":"sq? wip: re-use base_labels helpers in s3api\n\nWIP:\n\n* some tests fail because this patch results in more labels on\n  e.g. 400 responses.\n\n* I\u0027m not sure if we want the base_labels helpers in\n  request_helpers.py, or in statsd.py, or in their own module.\n\nIf proxy-logging has not installed base_labels in the request environ,\nthen s3api needs to do it, so refactor to make the base_labels helper\nfunctions available to s3api.\n\nChange-Id: I9d1ea362a0b0dc4207e5f3fe95fb531339ba5b18\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8546542375dfba142c6beed19e8f74f7ad7e8fbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8546542375dfba142c6beed19e8f74f7ad7e8fbf"}]},"branch":"refs/heads/master"},"0ac23c21b38c457521f95908a0193a7e11ecb910":{"kind":"REWORK","_number":2,"created":"2026-09-17 12:12:34.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/18/1006018/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/1006018/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/1006018/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/1006018/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/1006018/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/1006018/2"}}},"commit":{"parents":[{"commit":"875364b80483d79c11beb37f87d67aeed65dd4df","subject":"s3api: Use base labels for checksum metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/875364b80483d79c11beb37f87d67aeed65dd4df"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-09-17 12:04:15.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-09-17 12:12:22.000000000","tz":60},"subject":"sq? wip: re-use base_labels helpers in s3api","message":"sq? wip: re-use base_labels helpers in s3api\n\nWIP:\n\n* some tests fail because this patch results in more labels on\n  e.g. 400 responses.\n\n* I\u0027m not sure if we want the base_labels helpers in\n  request_helpers.py, or in statsd.py, or in their own module.\n\nIf proxy-logging has not installed base_labels in the request environ,\nthen s3api needs to do it, so refactor to make the base_labels helper\nfunctions available to s3api.\n\nChange-Id: I9d1ea362a0b0dc4207e5f3fe95fb531339ba5b18\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0ac23c21b38c457521f95908a0193a7e11ecb910"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0ac23c21b38c457521f95908a0193a7e11ecb910"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
